)]}'
{"id":"openstack%2Frequirements~851796","triplet_id":"openstack%2Frequirements~master~I48c26cc432a0322c5482bc1645537f59c1f6717d","project":"openstack/requirements","branch":"master","topic":"uncap-importlib-metadata","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-08-10 17:52:55.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-08-08 23:50:32.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"I48c26cc432a0322c5482bc1645537f59c1f6717d","subject":"Update code to use a std. read_requirements_file()","status":"MERGED","created":"2022-08-02 05:58:17.000000000","updated":"2022-08-10 17:53:59.000000000","submitted":"2022-08-10 17:52:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851796-uncap-importlib-metadata","meta_rev_id":"6d8a664170988b925ae6f8edf53c187313c1dfde","_number":851796,"virtual_id_number":851796,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 17:52:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-09 01:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-09 03:24:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-10 15:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 07:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 01:15:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-08-09 03:24:34.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"05d701f3e429785cb2ae228c91ff25f0fec0ad21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 05:58:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8eb443c86ba569144cbc828dce2e706fc7871af","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 05:59:03.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"845f1ab7abcfc6bee88866990d353cae166cc75f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 07:50:07.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af8e483c84254975abf96869912414b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f8c91324fa04344b44c51b8e60e3572 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42af2dea5eb34430914ebdf4161fb165 : FAILURE in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2546333fb9940568bd1689f91b67c7b : FAILURE in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aa74d09d4814accbe7f8c4aad31fd5f : FAILURE in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a8f0696d33447c785bdf5e0b026e5da : FAILURE in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cfc344131d34d0f87ee3b2af55ea0fa : SUCCESS in 3m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16d15b0db47d4895868c5b9f228180a8 : SUCCESS in 53m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6c43926039d447f9dafa9ad8638df58 : SUCCESS in 1h 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e271cba832d2cfcf48d8f983475b1953f6f5b14b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 09:19:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25e2a90932d16de5c1bab625779345db75ff7f7e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 09:23:57.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"c792f5e2673ec10b2292abc0b6953afe552a4df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 10:50:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432492d2aee249b1b12ce211a3096a08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2de1e8dbea415da7c9eb4ce0ce7077 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f277c41f2ba24f33b8113d8fa15b111b : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44fe6d83b9e4465d9f670f6926a26b7a : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/744816fc0038443b90aba02bde3a2e02 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6eac980247b4fab95343aa7b4c87444 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9318a4f34c1c40cead7dd86374dede6a : SUCCESS in 4m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/807683189b3d43adba21cad14c4d3b22 : SUCCESS in 49m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be2d2fb89a9045718a1b88f84ff641bb : SUCCESS in 1h 29m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9d2f5b8b607d83a42328c0db9bfdbcd8b912e1df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-08-03 10:14:04.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5dfbad2021496f9851d5314718f36c5acfbe50c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 11:47:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/845281729bde4c75ae6e37ea104d46b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2aa2ee1f15b41b28e32ef0afe6b375c : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df01e1160ea843fda5fbeecef406bdef : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e3ba125de0f4c79a53ce02bc8113398 : FAILURE in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7697a44b6102416a8183a33dc66a8048 : FAILURE in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f11a990d90e4446d8e78570e34b06e41 : FAILURE in 6m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d30ef96d9064a0a90e4bfe2a92c9a24 : SUCCESS in 5m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcaa5c5ae8f5451d91aae4aa98040e6c : SUCCESS in 58m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/58bbc0c6da91401d98da3b9112587110 : SUCCESS in 1h 31m 50s","accounts_in_message":[],"_revision_number":3},{"id":"0df4abb1d270412ac2638f81bb68f9819332960d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-08 22:34:26.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"3ca3ed12da9d98b8891bbb8f04e5b3ee12f6e99e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 23:50:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a5ade88ea494715a479644da85432c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adbbe394cb54fc9b56a35962816403f : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a65cf20bcdc84f34a82dcbdac25c664a : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04b8bffef71a4f3089bb2e917fdc9444 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/948b3208db614a21a63a5a1f01ed6a3a : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49300c8f3f4347c598085529d6d518bf : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d888fb842bb4e48a469eb1742e4b00f : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/434d35b9c1d04676ba9983864143d000 : SUCCESS in 39m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35a15109fa2d40dbbed1b0bcecea915d : SUCCESS in 1h 14m 52s","accounts_in_message":[],"_revision_number":4},{"id":"3843e1ccdbf6906d66d4222457110fc98bb42ad5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-09 01:15:06.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"082fbef4cadfe0fae7e1d202328fdd663bdc0c62","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-08-09 03:24:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"951624de7d33ed12b2b36122ae8d6d4519131507","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-10 15:59:58.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"92e36322d28e744c7aab540cb66c8f2d0efb2526","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 16:00:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fe19b39ecd30afae24695127f105b9cb3a9d33f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 17:52:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbc0b93220164b19ad31019126c54e33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169d66af12ce432f8441e33ecfb16c80 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/090ddb84604942758fd6a11e1bb6f38e : SUCCESS in 18m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4fe5ddc947f4d6dabec77d6153ef5aa : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d5af518d4f24bffb747ba86f186f0a7 : SUCCESS in 6m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c19576e8f044515ac3c6e155cf2f5e0 : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afacee33c1a141e3a33c398bff06b8b5 : SUCCESS in 52m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a789bc98c3a54c52978e4f6aca4b2f7c : SUCCESS in 1h 50m 40s","accounts_in_message":[],"_revision_number":4},{"id":"c5301d73a02872c36420672af9f04be37d8d4a1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 17:52:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6d8a664170988b925ae6f8edf53c187313c1dfde","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 17:53:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989ce5bca8794a5196d01aa2e42c3b85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff6ee512d764ade87aca6367fbb32ad : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b12966c487c055f1830ae257ffd3489b691cc830","revisions":{"554b6059dbe4e620cfc7caa2f8ec7a4e2fd1b5e0":{"kind":"REWORK","_number":1,"created":"2022-08-02 05:58:17.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/96/851796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/851796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/851796/1"}}},"commit":{"parents":[{"commit":"41adecc457bb788063417b993b5e92031ece8436","subject":"Allow backports to have arbitrary python_version markers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41adecc457bb788063417b993b5e92031ece8436"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 05:54:32.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 05:56:20.000000000","tz":600},"subject":"Update code to use a std. read_requirements_file()","message":"Update code to use a std. read_requirements_file()\n\nAlso re-work the BACKPORTS location to be an external file\n\nChange-Id: I48c26cc432a0322c5482bc1645537f59c1f6717d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/554b6059dbe4e620cfc7caa2f8ec7a4e2fd1b5e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/554b6059dbe4e620cfc7caa2f8ec7a4e2fd1b5e0"}]},"branch":"refs/heads/master"},"44399da98d274f874bc89c1e6ed8f8566a69b35e":{"kind":"REWORK","_number":2,"created":"2022-08-02 09:19:56.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/96/851796/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/851796/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/851796/2"}}},"commit":{"parents":[{"commit":"e1d936c4e137505c6af6d46bf766068acd7f69f3","subject":"Move the def. of BACKPORTS out of code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1d936c4e137505c6af6d46bf766068acd7f69f3"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 05:54:32.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 09:19:32.000000000","tz":600},"subject":"Update code to use a std. read_requirements_file()","message":"Update code to use a std. read_requirements_file()\n\nChange-Id: I48c26cc432a0322c5482bc1645537f59c1f6717d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44399da98d274f874bc89c1e6ed8f8566a69b35e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44399da98d274f874bc89c1e6ed8f8566a69b35e"}]},"branch":"refs/heads/master"},"f1321c3185943cd8d5dec5e96b0d1b992e0e808b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-03 10:14:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/96/851796/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/851796/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/851796/3"}}},"commit":{"parents":[{"commit":"1d068d6d2efa0b67bbcf2a8b966f4129d2aac325","subject":"Move the definition of BACKPORTS out of code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d068d6d2efa0b67bbcf2a8b966f4129d2aac325"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 05:54:32.000000000","tz":600},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-08-03 10:13:51.000000000","tz":60},"subject":"Update code to use a std. read_requirements_file()","message":"Update code to use a std. read_requirements_file()\n\nChange-Id: I48c26cc432a0322c5482bc1645537f59c1f6717d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1321c3185943cd8d5dec5e96b0d1b992e0e808b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f1321c3185943cd8d5dec5e96b0d1b992e0e808b"}]},"branch":"refs/heads/master"},"b12966c487c055f1830ae257ffd3489b691cc830":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-08-08 22:34:26.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/96/851796/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/96/851796/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/96/851796/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/96/851796/4"}}},"commit":{"parents":[{"commit":"5d31764c152b5f200e608f3da44f947ad504998c","subject":"Move the definition of BACKPORTS out of code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d31764c152b5f200e608f3da44f947ad504998c"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 05:54:32.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-08 22:30:16.000000000","tz":-300},"subject":"Update code to use a std. read_requirements_file()","message":"Update code to use a std. read_requirements_file()\n\nChange-Id: I48c26cc432a0322c5482bc1645537f59c1f6717d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12966c487c055f1830ae257ffd3489b691cc830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12966c487c055f1830ae257ffd3489b691cc830"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
