)]}'
{"id":"openstack%2Frequirements~851831","triplet_id":"openstack%2Frequirements~master~I414f7c386b838248f45fbce003a64d0c83a9919c","project":"openstack/requirements","branch":"master","topic":"uncap-importlib-metadata","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-08-10 17:45:27.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-08-10 17:45:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I414f7c386b838248f45fbce003a64d0c83a9919c","subject":"Move the definition of BACKPORTS out of code","status":"MERGED","created":"2022-08-02 09:19:56.000000000","updated":"2022-08-10 17:46:26.000000000","submitted":"2022-08-10 17:45:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851831-uncap-importlib-metadata","meta_rev_id":"3fd82ae5ec277c9df73fee01ea0722770ac62ae4","_number":851831,"virtual_id_number":851831,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 17:45:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-09 01:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-10 15:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 10:51:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-08-02 11:00:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 01:15:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"479e8b2e1123fe8eca36150f875d67dfdcce4d35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 09:19:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86d8ddececbb16fe209ce4bf3712c1dbb6f1b9b7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-08-02 10:51:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5a8961883dbcc74996e5b03a8a28ae4637537e7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:00:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3058550855e04229a6bcf4ede2a6d609\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4b87738215144eea7d76b1ad0add420 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb8e0730b692499790f27ce80e9225a3 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/627d61c7489f4bf7bbb402d83ed759bc : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44a64deba722443bad067e9d7da5170f : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3ae98892f9e47f4918645199ddf1758 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f9ae47d760243109d47eef2643e19ab : SUCCESS in 3m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f81960c7fa645e7bd2fc0242fc37297 : SUCCESS in 57m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93122d3192ec46ea90d9aab43175a187 : SUCCESS in 1h 35m 24s","accounts_in_message":[],"_revision_number":1},{"id":"636137ea07e96c39083478f25a72d9887b1c1a82","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 22:44:56.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"78363f901d2fdbfe1894158cba49d7cc2d9ad551","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-08-03 10:14:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"38b0b27d9d492c9a733077d4b5e9ce5155749e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 11:59:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f903ae3d7ee432080c90a158e7166e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2f59f1cddf74b328f555cb92da40c0d : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2764ce2e18c24bfaa4e0d717f101223b : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce6886a5d2644eeca029584985e64ea2 : FAILURE in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e90993bc78e742dbbb57c7b8577fb36c : FAILURE in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b875f70dbf1c43dc88dd34ce9c08e52a : FAILURE in 3m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca8a753d7dc04ad1bcb345eec0937a10 : SUCCESS in 2m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dac91aa2a727485da5972e6d001be6d7 : SUCCESS in 49m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ced3477b7bc3479091014d551a908715 : SUCCESS in 1h 39m 58s","accounts_in_message":[],"_revision_number":2},{"id":"979b2ecfa150f0838e1818bc2b88fcf0137b4e6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-08 22:34:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"40b0253fd35b9c8a9620a9eee77dc3b33525998b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 00:13:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef54745e92cb4f9bae533cb0ab557eeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08ce149947f04694bed203ea73024828 : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a86bf80d4414ca79ab5bf07f7a90758 : SUCCESS in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f08a0eafa7e34729be3eb3d39fa01295 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30b3970600994f3aa1675a6195e35758 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5da880ef920246118aceed20ee81c312 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/51e8370ca2704179a20c50630df7ada4 : SUCCESS in 3m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b5270da7b164277a06ef197b2b3b8c2 : SUCCESS in 51m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b094cf6970b042c09cb79aa66bfe1dae : SUCCESS in 1h 34m 33s","accounts_in_message":[],"_revision_number":3},{"id":"69fd0c478ba426557fe051b236183c18835698e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-09 01:15:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"be721f50a32077bbb03f71f58c512a8ffc9f5d2f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-10 15:59:58.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bf940bb5fca693c71014b7ec626f26a871f1e074","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 16:00:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c1870c4c7e1e083019194a054d9b98b407d7fafa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 17:45:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1403fb1a7764704aae32d52ff48f664\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d675b7afa14cc89b31569d16a611f5 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/932fdd571ffc452ab9469f63393a695a : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0096b3767f954e9f997f9dcc515d397b : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4bb0f9fd4c141cda556ba315e683dd6 : SUCCESS in 19m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/55f0f668aff54d80892ce404065acdc5 : SUCCESS in 5m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fb7bcef207d49e3b6dd00f0ef0f10a2 : SUCCESS in 55m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/525ef4de1a7247908f8d258e161ec2d3 : SUCCESS in 1h 43m 37s","accounts_in_message":[],"_revision_number":3},{"id":"b3245fcb20c3b3c2c591cb3b10fcd4a8f64985da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 17:45:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3fd82ae5ec277c9df73fee01ea0722770ac62ae4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 17:46:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b74debbcbdb4fd699a282c092d67c7e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e0ec9de4b5489381366c27d4409ebc : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5d31764c152b5f200e608f3da44f947ad504998c","revisions":{"e1d936c4e137505c6af6d46bf766068acd7f69f3":{"kind":"REWORK","_number":1,"created":"2022-08-02 09:19:56.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/31/851831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/851831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/851831/1"}}},"commit":{"parents":[{"commit":"a968919ebe3cd8d25df293d5d5f1eefadd79d542","subject":"Allow backports to have arbitrary python_version markers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a968919ebe3cd8d25df293d5d5f1eefadd79d542"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 09:04:52.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 09:19:24.000000000","tz":600},"subject":"Move the def. of BACKPORTS out of code","message":"Move the def. of BACKPORTS out of code\n\nStore the library names that we are flexible with ... backports from the\nstd.lib ... from a hard coded list in BACKPORTS into a txt file\n\nChange-Id: I414f7c386b838248f45fbce003a64d0c83a9919c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1d936c4e137505c6af6d46bf766068acd7f69f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e1d936c4e137505c6af6d46bf766068acd7f69f3"}]},"branch":"refs/heads/master"},"1d068d6d2efa0b67bbcf2a8b966f4129d2aac325":{"kind":"REWORK","_number":2,"created":"2022-08-03 10:14:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/851831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/851831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/851831/2"}}},"commit":{"parents":[{"commit":"9dcc4636b1db4da27a74c631333e4fae23c9f877","subject":"Merge \"Allow backports to have arbitrary python_version markers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcc4636b1db4da27a74c631333e4fae23c9f877"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 09:04:52.000000000","tz":600},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-08-03 10:13:11.000000000","tz":60},"subject":"Move the definition of BACKPORTS out of code","message":"Move the definition of BACKPORTS out of code\n\nMove the library names that we are flexible with - backports from the\nstd.lib - from a hard coded list in code into a txt file\n\nChange-Id: I414f7c386b838248f45fbce003a64d0c83a9919c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d068d6d2efa0b67bbcf2a8b966f4129d2aac325"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1d068d6d2efa0b67bbcf2a8b966f4129d2aac325"}]},"branch":"refs/heads/master"},"5d31764c152b5f200e608f3da44f947ad504998c":{"kind":"REWORK","_number":3,"created":"2022-08-08 22:34:26.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/31/851831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/851831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/851831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/851831/3"}}},"commit":{"parents":[{"commit":"d79553627daede71690782a98d6418102e4ad93e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d79553627daede71690782a98d6418102e4ad93e"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 09:04:52.000000000","tz":600},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-08 22:29:55.000000000","tz":-300},"subject":"Move the definition of BACKPORTS out of code","message":"Move the definition of BACKPORTS out of code\n\nMove the library names that we are flexible with - backports from the\nstd.lib - from a hard coded list in code into a txt file\n\nChange-Id: I414f7c386b838248f45fbce003a64d0c83a9919c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d31764c152b5f200e608f3da44f947ad504998c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5d31764c152b5f200e608f3da44f947ad504998c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
