)]}'
{"id":"openstack%2Frequirements~851835","triplet_id":"openstack%2Frequirements~master~I5575d79d2a17a9c699f678a8e7fbe15e9a261d56","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5575d79d2a17a9c699f678a8e7fbe15e9a261d56","subject":"update constraint for openstackdocstheme to new release 3.0.0","status":"MERGED","created":"2022-08-02 10:22:37.000000000","updated":"2022-08-02 17:33:34.000000000","submitted":"2022-08-02 17:31:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851835-new-release","meta_rev_id":"342fb63f5c5575048a19cd3b18c009a732ac377f","_number":851835,"virtual_id_number":851835,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-02 17:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-02 14:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-08-02 11:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-02 14:31:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 10:42:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-02 11:59:27.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-08-02 12:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 14:31:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c211e283a876e30fbab53c25f8b67b8b8667f3b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-02 10:22:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"599dfb5f61ef73e43216ca9478396f2ba7fc969e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 10:42:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/164eda64fb3f472d82d0a16fc88dc625\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/092294a3c3cf4605bd2a84cb9b72349c : FAILURE in 14m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f809ed6d8ab75cebe234db20b8b24006caede542","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-08-02 11:59:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4d9e4389a5d978f43ce2209572d64328b1a03a7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 12:21:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ca8126ec5a4c049c48657098a2c943\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6be1a40d0c408eab9360e82e00cf3d : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54dd0adf1dff4a69beed63f93b879847 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a12bb60d8fb54263a88f59a3efc5ae6b : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/408847381dbd45fe92aa49b857d92a04 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7189b9692e3d4cd185b8bc2b4d6ce83c : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce72719391a44167ae22753378a4fa2b : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b8ad782d34004076a55f764ed2c17979 : SUCCESS in 4m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e15d1dc19274604ba9b75ca204711f1 : SUCCESS in 11m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c5218bcdda094088988e26798c02a3bc : SUCCESS in 12m 22s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/be01bc2abecc4eee9bfad5c4a92e9ea1 : SUCCESS in 9m 43s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e50ee84e2fd74db2ba7920cb6e0448ae : SUCCESS in 5m 33s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c71f279acb91460ebaefe20fa38b2605 : SUCCESS in 8m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0ee75bd09845442c8fd2fc85dac5bc06 : SUCCESS in 12m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9c591bab09374a499c5e60ce0a146615 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b0e1c460f64c44f5a56f3f124892c780 : SUCCESS in 8m 40s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/772bbe00e84b4c4aa65fb56c280501c1 : SUCCESS in 11m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/afd5485794564db2b4e3b477c65c89c6 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ca206b4c5d34cbab26022e0b19a6787 : SUCCESS in 10m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/70aa0da8c6394d9ba9e33024ca81a52b : SUCCESS in 10m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7211a6dfaf014099bf5aa44025cfc07b : SUCCESS in 10m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7f36946fbfb34834a9a06b1e4e7df62e : SUCCESS in 6m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1fe593ddbfee42909c43afd43514aa15 : SUCCESS in 10m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f427afefc1db4b72bfef56cd9cd2f262 : SUCCESS in 7m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/59490fb49fa3444b9899b024ffc468b2 : SUCCESS in 26m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3bfcc52ab32f41bf97570a75d0465be1 : SUCCESS in 14m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/06eb74eb90d94a96b3678e5b8beab92a : SUCCESS in 8m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59d33909d3c04d88870dd3a1b879b209 : SUCCESS in 6m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80e293cdbaf84fa586f6944f4137ed89 : SUCCESS in 22m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/424d8094fd8e460ab7c6385707bf9973 : SUCCESS in 6m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3fd005665d694881a600ff6b346fbeef : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad955fb20de8411d98100cd13c15e3bc : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7aa938e2101a445bbac4a62da305b3d5 : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70e987387a6346a989af5ae9ff235473 : SUCCESS in 55m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33bca034094545768e4e5af4f28b8adc : SUCCESS in 1h 53m 26s","accounts_in_message":[],"_revision_number":1},{"id":"723311f3f31efd1bad7b635383c5b1b3156e3e8d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-02 14:31:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e16db8da0a977fd0293f897756f70801f9f17de0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 14:32:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"672fadf29ff674f94055b6a2cf0dc5bdb42c6c1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 17:31:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e1bcaa811e148a0a210ac6d87302ebc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1994e816661497c98bc20c16b5ac972 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d64b34386b48c385aaad5e41c43fb7 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/395d15d85164412e86ba28781a6dd200 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d46773e68434d23aaee2fe1ba625543 : SUCCESS in 4m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/54138ca74165458ebff44cde8e2704e4 : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/be15daed3f5c43e58598c9caac9ebd71 : SUCCESS in 5m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0ffe7281a2cb4568a9d14e96e4d80c22 : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/31247f2a82cd4117bb926d750c1495e7 : SUCCESS in 13m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/87e733c6812a4429b507d56474854258 : SUCCESS in 13m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/07261c34e14746688b42fb99a7d0d3ad : SUCCESS in 5m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3a193b7dd8524f799f13edd879e94c3a : SUCCESS in 6m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/68b87dbf58844b9982b8ac4e1d1f0998 : SUCCESS in 8m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5396a6c7afde48e4930ccfd9589c8516 : SUCCESS in 9m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f9110fe16bdf4ad988b66ae06bc74bc9 : SUCCESS in 9m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fec66e4fba0d4f948a3a584d153b2fcd : SUCCESS in 7m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/246f602da6444417b7cd4931bea0a211 : SUCCESS in 10m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c5c497a770a84d5db7f8200913a60643 : SUCCESS in 4m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/849e786f14b04e4384d9348605ad47fc : SUCCESS in 8m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bf19adb912cd48419b1c5e647812092b : SUCCESS in 6m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/05704b407cf84ec6869b115e9f9608d2 : SUCCESS in 24m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ff5bf8f7a39c485a9febfd3f356389e8 : SUCCESS in 11m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ef0823af68fd41309103f6faed9e2f76 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a267c9fa9b041f8856ae708639589db : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d4b6c4a684f4343beef43ac15871827 : SUCCESS in 23m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ae7a4df0e23943ecb966fb3bca943dfd : SUCCESS in 9m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b2d1c5ea25c54117a3f9bcd119a237a4 : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/162d4e5cbb104c12b9570e221bad5335 : SUCCESS in 7m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3ab750d128e8426785d85dc3f243dfdb : SUCCESS in 13m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99b9c64215414412bc7e275073f5a69c : SUCCESS in 48m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0963a0aef0ca4b9cacedd3a1c07ca9b0 : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":1},{"id":"02c0806fa87801c9f3f4c03ec927361ffe50b266","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 17:31:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"342fb63f5c5575048a19cd3b18c009a732ac377f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 17:33:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecf1120d4b994d3dbdd5cf8e72beb33d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca580e6e869747229ee8d4ec7f1fa6f4 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbbc8728879f8ef63df446bb54e1137805d0f588","revisions":{"bbbc8728879f8ef63df446bb54e1137805d0f588":{"kind":"REWORK","_number":1,"created":"2022-08-02 10:22:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/851835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/851835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/851835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/851835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/851835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/851835/1"}}},"commit":{"parents":[{"commit":"bed46f4b7ab58badcddd45e0666481eb89aedcaf","subject":"Exclude Werkzeug 2.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bed46f4b7ab58badcddd45e0666481eb89aedcaf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-02 10:22:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-02 10:22:35.000000000","tz":0},"subject":"update constraint for openstackdocstheme to new release 3.0.0","message":"update constraint for openstackdocstheme to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Change-Id: I37aa44c9cc8a9df90ebc2d98c6313b1edc0c7f0c\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I5575d79d2a17a9c699f678a8e7fbe15e9a261d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bbbc8728879f8ef63df446bb54e1137805d0f588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bbbc8728879f8ef63df446bb54e1137805d0f588"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
