)]}'
{"id":"openstack%2Frequirements~852142","triplet_id":"openstack%2Frequirements~master~I7ee1cc1a6d062f10568d6362be2aa9b86685c137","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7ee1cc1a6d062f10568d6362be2aa9b86685c137","subject":"update constraint for oslo.concurrency to new release 5.0.0","status":"MERGED","created":"2022-08-04 10:20:44.000000000","updated":"2022-08-04 16:10:46.000000000","submitted":"2022-08-04 16:09:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852142-new-release","meta_rev_id":"fce56ea7db6dc9a38c4f594e6bfe3483bfcda93a","_number":852142,"virtual_id_number":852142,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 16:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-04 14:32:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-08-04 11:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-04 14:32:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 10:41:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-04 11:06:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-08-04 12:25:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 14:32:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fd8eda526d4094328321ccd28eff23286ea6fa1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-04 10:20:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0b26e9675e01a1836614598509500439046e40b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:41:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baa6eb17554549e594b8a3cf29826389\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/80105a6eff5d467ea67d9287c9246613 : FAILURE in 15m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2dadd9a0dbbd7c0bb8eb21190d761a8e9a02ab85","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-08-04 11:06:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d314f89868e8a6cddc044ddeaf61028136d86ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 12:25:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ac613315834f84ac6f5bcef20a63e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892712929aa24bb780c6c97e944e788f : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9b8864cecdc417db9c50ed2f8bbb37f : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9214d9d645f64410aa927504aa8c57b4 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26e845edf0ee43d4af1cbff74541c7e2 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6165c66e6c19436e9b707c8fbf34dbb2 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/25ebd5a6a4134b6cb691692243c8c582 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c6b14637cad459f9b8a0d1add803534 : SUCCESS in 4m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1dce00a875d5427db6cb7c554df714b0 : SUCCESS in 9m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f4bad6ff1bb04da4b33374f9970fbf35 : SUCCESS in 13m 31s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/4cc483b3a41046f1ac062dafd47e20ab : SUCCESS in 7m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/264f1419c80a4dbb9b37c78199f6c313 : SUCCESS in 5m 47s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e9a1e152ae61403e917b8263394b3cbd : SUCCESS in 7m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d010ffa805fa4376adaccc53f8e14264 : SUCCESS in 11m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3e9c25d49e164907904240cef9ea34f6 : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ad146cf9f48d4becbb9dd8445ab90fbc : SUCCESS in 6m 48s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ee1d297082bd4dfbb94f65e48a3bde58 : SUCCESS in 8m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b4276c8cf8a145e68c710c9ac968751a : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e1aeba323b14ec281659d65a753940c : SUCCESS in 8m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/90d86dd590a74291b2d83cd9d7b8d6cb : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c35ce19ba7874a6383646ab5f363257b : SUCCESS in 11m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0c20d81926fa4b96972ee2f51148f733 : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ad41080c6e4e4d4388bfe4e8cb92a581 : SUCCESS in 10m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/edef0dde85b74559bdb7f1627694076f : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/259159745cf04914824a8e27609bec55 : SUCCESS in 26m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0f9060dc093c461699d1a93eef38cf9e : SUCCESS in 13m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f4a3497128ba40c3b91e0779e2e87d6d : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/437dfe6d1d09497aa7b6ceef839a61bd : SUCCESS in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eca3e06b71e740cb81404cef906be04b : SUCCESS in 20m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f042a74da60f43c6a2b70f19ec90e59e : SUCCESS in 7m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d7b91d74a0d64bd0965d422881c0548a : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1788ecab240403483e1b138293943bb : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ad39ea93dc9c437c9463d0dea6e761e2 : SUCCESS in 11m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64b6f1f9907145c48cc6895907972c88 : SUCCESS in 1h 07m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f968f81531a04ef9b01cb06215f5b030 : SUCCESS in 1h 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"20d5278666df2902b5f352efca187b60656f1f4e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-04 14:32:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9ba9873f4ce83106480383eb5ae11d2c9740b9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 14:33:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7012c23ee7d6a363ae1f43c6d3ed68374a79a57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:09:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1febf83868ad4f919ff890f19d76c27f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7508d919dd9947e1a420d6ab5f32e0bc : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37f70b8a9d6d4beb8a51727ad139a8fc : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd4cf66045b1420292d740d1fb17e06b : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/636caed2cca24de0879e5c3ee712b069 : SUCCESS in 4m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a637c62f10a4bb6aa569c3829a2009b : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ff23e0fa6a04ee598cf5b5369154706 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/af3c0daf364044d69e4f99b4a9b3707c : SUCCESS in 9m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/893c36d78ac44d5f8a517e20e3f517ab : SUCCESS in 14m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b0b7aa804e744bdd85ce0b1f562e82b4 : SUCCESS in 14m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/85d6fc3ae12b4509908c7799ae9b2802 : SUCCESS in 4m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3aa4b4f09c9c4792a3e0cee12fd85c69 : SUCCESS in 6m 21s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6efcc023a0bc45c0a1e0a83dd0369a23 : SUCCESS in 9m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4def413bbe0b4f7490efaca33621dd6f : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a8b518f43254414ac56d477f685914c : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/65b2332644d941559c572ea4c54beca0 : SUCCESS in 9m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4fb45747bf93418c88ecb54e9946ae4b : SUCCESS in 11m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/19e9b7814867462794cbbadbcd34b833 : SUCCESS in 5m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6487002f37734b4db79daee26f318cae : SUCCESS in 9m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/add21f007d0e4d7a9695f3531368740e : SUCCESS in 8m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f3a5e4fd8c654a10a97a2cb2ffc63385 : SUCCESS in 26m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/08a0494b8c01424a9d079d3043e375d3 : SUCCESS in 14m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f21da1394b0b4516acdea4253093258b : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e18ec827b7d14b9b865847772d7f797c : SUCCESS in 6m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/276160585b554a86a4c9206b488593c2 : SUCCESS in 20m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5857c897ad11439096a8d148f25958f1 : SUCCESS in 8m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/23a801225a1f4bb2a6040398740ed6d7 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0db747d9769e4507a21660964f8d1b6e : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d79127d86c34424da4332aa2715b4be6 : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63bb7d66003148519ace8cd32fe25e7c : SUCCESS in 1h 04m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/189df42a89864c4fa2abfb1ac94e28fb : SUCCESS in 1h 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c2c0df05295e35a1ad1949723004a14f59ec574d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:09:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fce56ea7db6dc9a38c4f594e6bfe3483bfcda93a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:10:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a81bce8e85624cf1883cdeeb9d04353d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f6e8588a62a49c38e16761b65e4bfb6 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7f734b786c6609882bf16b8da79ff626e78bb45","revisions":{"f7f734b786c6609882bf16b8da79ff626e78bb45":{"kind":"REWORK","_number":1,"created":"2022-08-04 10:20:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/42/852142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/42/852142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/852142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/852142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/42/852142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/42/852142/1"}}},"commit":{"parents":[{"commit":"8efe11a74a87f4a1522099ab5734c6af628c540c","subject":"Merge \"update constraint for openstacksdk to new release 0.100.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8efe11a74a87f4a1522099ab5734c6af628c540c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 10:20:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 10:20:43.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 5.0.0","message":"update constraint for oslo.concurrency to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iff79935d876bf7996419c13adb39a2d7c3afe5fd\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7ee1cc1a6d062f10568d6362be2aa9b86685c137\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7f734b786c6609882bf16b8da79ff626e78bb45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7f734b786c6609882bf16b8da79ff626e78bb45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
