)]}'
{"id":"openstack%2Frequirements~852151","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I017c9139cbf592f1a19e627efa918a2e6cc0787a","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I017c9139cbf592f1a19e627efa918a2e6cc0787a","subject":"update constraint for tooz to new release 2.8.3","status":"MERGED","created":"2022-08-04 10:29:51.000000000","updated":"2022-08-04 16:00:57.000000000","submitted":"2022-08-04 16:00:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852151-new-release","meta_rev_id":"1020d40ddddc9c611fe56f07af10a2597ebdf156","_number":852151,"virtual_id_number":852151,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 16:00:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-04 14:09:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-08-04 11:07:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-04 14:09:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 10:42:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-04 11:07:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-08-04 12:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 14:09:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6242012814343d89277199358e68b27bfac14168","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-04 10:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ee9c6aa08e85bf84b1ff9a1480144242ec91131","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:42:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b829cff8388e40efbdea37b074eba4a1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/85f7cd5d25a44b10befadb7af20b04be : FAILURE in 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"07a57c83e528d1ed214da4a026aa8bfc3ee1be0c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-08-04 11:07:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6cea7fd25fe0e0597e86748acd459e835560865c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 12:24:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d9e14cdaa9e40478a50363c34d72211\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae479b25e2e34edb8c113c3509ffef7b : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a28be556ab414af9bbcbdc7f553d895b : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a05d3ee871b443bae3404117a9068d0 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67398da3e68a459db2b0f4a9e0f117b1 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ab4ba24808748abb8c3daed32f32c1f : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ca0b1bf378a4113a1523f4403cfee69 : SUCCESS in 4m 28s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/33b24ab520ce46f79fbc27c5a8c49d95 : SUCCESS in 9m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9aa79f541ec2420bb12b1e442e73bdd6 : SUCCESS in 9m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1ca8340b757246ec894bce7f7953fba6 : SUCCESS in 10m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a7813a769393482580800aa93baf7b46 : SUCCESS in 5m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7a83fccb7c6d4e56b3e7d83ee9fef2c9 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/84e9b05dc684437c81d03bfd1086061b : SUCCESS in 7m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b389169e0bae488c80ba7cf2a7ab878a : SUCCESS in 8m 58s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/f56b4c96243f4e278e6c078f54ee36d0 : SUCCESS in 14m 18s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/ef12f0f7110142498a41fb50c2f6cd0d : SUCCESS in 5m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/dcd3b99d86fd4dd2ada53fd765df77a7 : SUCCESS in 30m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b6cbd295641e470ba1ed6219f36157bc : SUCCESS in 13m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/16d72e5fab654ee38d4fa9966c7e477c : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/41a8c51ec21d439b8564872877ba1eb9 : SUCCESS in 19m 06s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/70378677e7414e0aa4bc079b72a3c200 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8a006b1eddcb437aa7d8987e34cab1ea : SUCCESS in 10m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bf4c68a57dfb422c9b5f021c60b7606c : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0be373d36a144b5a9381b152ab049cd : SUCCESS in 1h 00m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2fffbe915414d98a6259021347e2b88 : SUCCESS in 1h 47m 20s","accounts_in_message":[],"_revision_number":1},{"id":"37d0c40cfbaecdf0b2458f182553cecd24105bed","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-04 14:09:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d015206d2a5cba08ac7f2d1bbe60258543cd6fdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 14:10:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b72d2da8471fd5a0b60933e895c169efb557c231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:00:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0885b39d74ee4c50b28c1c11c807a08d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2894843cc5f9420098c9b8d4103b8128 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a33b928fa7c44d2a9ce06a1f5d61ab2 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eae2e08724244a6aa2134d808fa5b865 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7e81e7baf53433dabd8ed98520ca57f : SUCCESS in 3m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6555c92aeee14267a24ec2596afcef39 : SUCCESS in 3m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0fcee5ac015b4a519f581bb95f415a0f : SUCCESS in 10m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1c03ebf9b6f442cdb0b8ee72b680e72e : SUCCESS in 11m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3d253bdd0fe242fcaea142bbfa4519ba : SUCCESS in 13m 08s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6b9bbf73f2ad45ba8285858f4e1cdb88 : SUCCESS in 7m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/247cd45565514df49b1b6c7990038d02 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/edad702360cd43e18c6003cefc1d511d : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/56cac0a117b246c783647ff9b99c5058 : SUCCESS in 9m 20s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aae3fb97a95443fb93994b5b9ee04ddf : SUCCESS in 17m 49s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/004e3f0eec17463881c2d85467c173ca : SUCCESS in 6m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/71cf6d5c7ab2460db2084deb567b75ea : SUCCESS in 32m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b7d7e4998f39484b9f78cccfe79ec990 : SUCCESS in 12m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8a74b1cfbbc84967aae25968270096b2 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de527e328d51494b8475fd0ae0a21d30 : SUCCESS in 21m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c75abb0855f44333bf407c896d83b828 : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/55b624057e24410883ac0abebf6c0675 : SUCCESS in 7m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fccc5c1da3e344b58a6c91568802e9ab : SUCCESS in 9m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1207c3868a004db28745d0373a3341d2 : SUCCESS in 50m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa6d0102bc314aa88ae166358facf6e3 : SUCCESS in 1h 33m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1020d40ddddc9c611fe56f07af10a2597ebdf156","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:00:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ffbaedffa79472d2793dcd5aa5770f5c1bebc94","revisions":{"9ffbaedffa79472d2793dcd5aa5770f5c1bebc94":{"kind":"REWORK","_number":1,"created":"2022-08-04 10:29:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/852151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/852151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/852151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/852151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/852151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/852151/1"}}},"commit":{"parents":[{"commit":"92e98bf195f2e24b0b6df27d3c5be5a61c5fe0fc","subject":"update constraint for ironic-lib to new release 4.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92e98bf195f2e24b0b6df27d3c5be5a61c5fe0fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 10:29:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 10:29:50.000000000","tz":0},"subject":"update constraint for tooz to new release 2.8.3","message":"update constraint for tooz to new release 2.8.3\n\nmeta: version: 2.8.3\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I94b81c95184dd2564ebf61aff7b8940d8a0c4776\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I017c9139cbf592f1a19e627efa918a2e6cc0787a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ffbaedffa79472d2793dcd5aa5770f5c1bebc94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ffbaedffa79472d2793dcd5aa5770f5c1bebc94"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
