)]}'
{"id":"openstack%2Frequirements~852152","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I121aa9f786a19cfb5ad4589c8ba0ca001693220b","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I121aa9f786a19cfb5ad4589c8ba0ca001693220b","subject":"update constraint for oslo.vmware to new release 3.8.2","status":"MERGED","created":"2022-08-04 10:32:17.000000000","updated":"2022-08-04 16:08:42.000000000","submitted":"2022-08-04 16:08:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852152-new-release","meta_rev_id":"125d409a87d1fb931e8a99fa30cc84addfb5da2e","_number":852152,"virtual_id_number":852152,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 16:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-04 14:09:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-08-04 11:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-04 14:09:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 10:44:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-04 11:07:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-08-04 12:07:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 14:09:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5731665cb78dbedc284ba973dd8893ae505996ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-04 10:32:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc677c7d091378b9600f38ef3cd5526d0a88c05d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:44:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4acd3451624a4fe1927865c9b2f899a1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6f1d2eae853647a4a716c998fa1d383f : FAILURE in 10m 35s","accounts_in_message":[],"_revision_number":1},{"id":"94a0eb49d8be77180d1474e588b2f9187dafdb62","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-08-04 11:07:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"74c3c5c2e88ceaef8d14e47df576790a00909aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 12:07:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/570029a0b39b434f9bc2354cea5bc00f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae3329c22d1f4901827d494753f9e579 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1090da0f8824d74bfbba6e2d6c23d69 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f934a06ebe7d40ed88c72851ec32775b : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b71231070abe4c959cedd345c45776b1 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac45f323cb4e4b97b1ec14b4613edaf4 : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77ae7c4e048044698a52c60729a7aa2f : SUCCESS in 6m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8e91df0156224d189291628623cf4316 : SUCCESS in 9m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6bb6edec6cde4cd5910fa178f236e90f : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/edd6f52a93424309976fea580d764bc1 : SUCCESS in 10m 00s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dbd4d62b85844b328af7827c931440fe : SUCCESS in 6m 44s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/11b8f1e0c9f74671aa97ee226b3afe89 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb2ff93e932742dbbb69cdbf9411da22 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/10ecb6d999eb4f2799b08dc2b975b1db : SUCCESS in 10m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dfc656a98b204d33aeb21bd9756c87f9 : SUCCESS in 19m 24s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a4609d70522d43e5806554fafc2a014f : SUCCESS in 6m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8b1ac053ad5347b0943ed4d089df1797 : SUCCESS in 25m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0a01a95771e74aa89be0cfdf0caa3a50 : SUCCESS in 12m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0b26bc63e2db4bb6a0aaaff85c767917 : SUCCESS in 4m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b1142e8b776f4b3fab99c9a03b067fa5 : SUCCESS in 17m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/99bdf93378184a458851f69d784bd1ab : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3275892c39314773a8c692bf4200621d : SUCCESS in 8m 59s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d8eaeda348c34225a283fea849624812 : SUCCESS in 15m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/492e313ea1c4418d801c1cd36afd711a : SUCCESS in 51m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/583824b22166423c9db57689d7627c16 : SUCCESS in 1h 32m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e3c85c495138fadbcbcbc67667c4391ee587ef49","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-04 14:09:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1ffc50838318a16e5a851bdf1da49bd99eb47ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 14:10:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f846d8a7054a71a46886eb378f2a8ba9d117eef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:08:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5b2545acc7f4135a827a0e79e1fdcdd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/165b8bb0aa83476eb0799dac826dc505 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ccdd95622bf4178a9bc27e5721466c7 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/576dccde9648418dad19f0a037d1ccd4 : SUCCESS in 5m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1eb1afd3f01343f68018e56e3c7f3ec9 : SUCCESS in 5m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b24f4bc141c04444a101e9250885f634 : SUCCESS in 5m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce09b9b9bbbf4ccdae78744fe8ec014b : SUCCESS in 8m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a9ed3830ff746718be7417790cb934b : SUCCESS in 7m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d7fe3e5eb66468d9be4b61b64802c5b : SUCCESS in 11m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/99c7be4d196943e08329b510eefc8a8e : SUCCESS in 7m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0fcf74f224d7421fa59cce5cb561c387 : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/02fe2771414144c9b4ce497083f012bc : SUCCESS in 9m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d1d03f32a96f44eebe9221c9878b2a21 : SUCCESS in 9m 07s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8e88b5c9cc574662a68baeed481fce2d : SUCCESS in 14m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/1996e286103c48cea06fcd6fbcf68f13 : SUCCESS in 6m 35s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf7dd52047b345f8b9bfdecd069fa4f6 : SUCCESS in 34m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aec029e30d064f9089dd95492b4e09b8 : SUCCESS in 13m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d4e5dee912554f8496885278ea520ded : SUCCESS in 5m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a5197787c06d4472a66b382e52518cfa : SUCCESS in 21m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/131a73cb37ae40e5ba3045e42e868861 : SUCCESS in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c0d36fc8f50a46debc9c00a2ea2b02e2 : SUCCESS in 8m 33s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d76b84d2675c481895bddfd7501c6976 : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b37ad1b413984a3aab7acc9e26d071f7 : SUCCESS in 1h 01m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18d613bf9baf45fc9ac97de5288dc3f2 : SUCCESS in 1h 35m 03s","accounts_in_message":[],"_revision_number":1},{"id":"125d409a87d1fb931e8a99fa30cc84addfb5da2e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:08:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b170bdf27d65627c93c016b7ffc54247b3c9d2ae","revisions":{"b170bdf27d65627c93c016b7ffc54247b3c9d2ae":{"kind":"REWORK","_number":1,"created":"2022-08-04 10:32:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/852152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/852152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/852152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/852152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/852152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/852152/1"}}},"commit":{"parents":[{"commit":"92e98bf195f2e24b0b6df27d3c5be5a61c5fe0fc","subject":"update constraint for ironic-lib to new release 4.6.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92e98bf195f2e24b0b6df27d3c5be5a61c5fe0fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 10:32:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 10:32:16.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 3.8.2","message":"update constraint for oslo.vmware to new release 3.8.2\n\nmeta: version: 3.8.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I94b81c95184dd2564ebf61aff7b8940d8a0c4776\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I121aa9f786a19cfb5ad4589c8ba0ca001693220b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b170bdf27d65627c93c016b7ffc54247b3c9d2ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b170bdf27d65627c93c016b7ffc54247b3c9d2ae"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
