)]}'
{"id":"openstack%2Frequirements~852370","triplet_id":"openstack%2Frequirements~master~I81c77f1775691e492ae410989713341aaeadcf5d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-08-14 15:24:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I81c77f1775691e492ae410989713341aaeadcf5d","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-08-08 02:31:44.000000000","updated":"2022-08-14 15:25:10.000000000","submitted":"2022-08-14 15:24:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852370-openstack/requirements/constraints/noclob","meta_rev_id":"c79ced86194176436bf186e1c97ddc66119ec4c2","_number":852370,"virtual_id_number":852370,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-14 15:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-08-14 13:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-08-14 13:43:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-08 02:47:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-08 04:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-14 02:37:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-08-14 07:51:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-08-14 13:43:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"8b3eb267b4dbe7555bdc8520f20bdd0ffc97eb7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-08 02:31:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79618dd8cb5b73339f408954873344e31d3addf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 02:47:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe359a13f4e941eabcbf90a9aab3d64d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af938b43d57e431484a865f0bca0992e : FAILURE in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"2bebbf37d1310c748410fce6222b80f1a5bddbdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 04:08:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9854230304ad476b80b2dfcfca4aeb71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b7f27c074642f7a9e9c968c1eaecb0 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89eb91cc2d7544ce9320f6f7f1363635 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f783dc2f90b4cc7988d2b4df4d835b5 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21cc10c6b20c48f9a0625a0f75d29bf5 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfe3379325494a0f9874ef3aa67ce2c1 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f97e246677f94a3982efeb296cf7d95e : FAILURE in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5a17f9be60c44bca827c8964eb1ecaf7 : SUCCESS in 3m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ea0765e1d3742dfb5fb5dd0789d7994 : FAILURE in 4m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fb31ec49fe714712bfaf3e5874c3150e : FAILURE in 4m 33s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/c74f4cfa86bf40008517289e7fea9f63 : SUCCESS in 7m 45s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3fcc9612f0ff4e188ca534e2b9d8781a : FAILURE in 8m 15s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/44cd797b5625474b8f6014bac4a30c93 : SUCCESS in 7m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ab75a9c0ef91471387da0cbb534968dd : SUCCESS in 11m 10s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c8697d00fcd9484581795bb57d697156 : SUCCESS in 5m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fa2f1e0f6cae4c39bdf37d97c76956eb : SUCCESS in 6m 17s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/146d0ee4ca3e4002b0f7636ea72c2aa9 : SUCCESS in 7m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d2861038e23043d79ad580bc8041e55e : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4b0d85a87924a6c926c20a891117011 : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c25b227b34cd4d4cb023627afddb0e50 : FAILURE in 8m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a7eef73445614d90842101f46294a35a : SUCCESS in 12m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2751cf0b6a194d4bae2e661d7af80252 : SUCCESS in 4m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6ba5ea788bb04477af09b425c91ba90a : SUCCESS in 9m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9243487d26d5460ebc95d2181b0de4be : SUCCESS in 4m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3097402108f242eb853ac09723678043 : SUCCESS in 31m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6d9da50e21f743f1aef0440e2c444de4 : SUCCESS in 17m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5dfe9ddd7174406599632314875f7370 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f32b549d738427da658c49bf852c40c : SUCCESS in 6m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a6d728fcb6144b6acae9c5d9237df67 : SUCCESS in 21m 19s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/452a3bdaf48c454296b9b0ba5c001ce9 : SUCCESS in 7m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/34eacf55b84d4ce5a816c1d7fc9497cc : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/490da09905d6401a83805f0e678fe0fc : FAILURE in 6m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8221537f24cb4f569c4b18697b149406 : SUCCESS in 9m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a63ab758b94a49daaf49f6ab7b996192 : SUCCESS in 46m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22afd1c253f640e2a7fe2a1b704f0cb5 : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"284d7160e13ad503ae46ac6a167d8dad9118a857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-09 02:41:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"91ba97ad5d8d1eb52c72d279e2197251bba2d4d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 02:56:32.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b32877a41cb04756b8f51edfe36f9894\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5788d1694f4c43d6ab20bee5a93d100d : FAILURE in 6m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d4df8907e1f1167800c610fcade3704ec10693fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 04:24:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b47ed31fdc61427b84bcc45ddc700ff0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23b0d35bbad4f51a22b9ed8de453af3 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a12475626203440193511632989acff8 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4421bc1a66644c078448ae1f249da8c9 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d7d04686e9e454f83f171c96eda5710 : SUCCESS in 7m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/735516f1fe8d47e4bd7aaea28ef662b1 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/829dac2e3b924d7ea669a282afc8dd3b : FAILURE in 6m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f278be8ef964dca8cc3945342086f75 : SUCCESS in 7m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e04145f791e8446786ea34db447a3460 : FAILURE in 4m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6695f85a9dbb4dae853f76f546dc4d8f : FAILURE in 4m 59s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a05baf5120324ed9b6f23e9f6b2b12ed : SUCCESS in 9m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ecfe9ee1afca4a9794f06368b557e3dd : FAILURE in 5m 48s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2fd9aced867044c195f8cf4956e89f49 : SUCCESS in 10m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/10ca5adb37ca458f88aeb88fca77673b : SUCCESS in 13m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c7592267f71944b19548342c81d5e2e6 : SUCCESS in 7m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/03498a44ae954d41be6998a4e880d3e2 : SUCCESS in 5m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e10add01164f42099b0978c12902973d : SUCCESS in 8m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8144e9a0eea34ca3b42bfbdc18746869 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/809177bd267e4a26b04729b686f74641 : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9531f68efd8041d9b321b46e8b9e72a8 : FAILURE in 8m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7dc96f596a8847d5bfb3b7ba9adf55a3 : SUCCESS in 15m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a61b0db2aef44b7bad7090eedcb487ea : SUCCESS in 5m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/146fcbdd155c4f14935c32726f559462 : SUCCESS in 12m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8f7f716d057e4a8cbc7f29b12f9a6694 : SUCCESS in 5m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/73f868501985470e8bf35615ae9b4810 : SUCCESS in 27m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d09e4b7167334cd0bf13c1481b1331dd : SUCCESS in 12m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/76145fe44540411a9986e2de58900920 : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa59c30566104c128e318e7236daaa1d : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fc5d55c89f04675b143325f9245007b : SUCCESS in 19m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b80d58c6b7634c0c892b4ffbfdca7b1a : SUCCESS in 7m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/94c9d95daac640b2b37a72f8c0d825be : SUCCESS in 6m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/52a4a9fc1eef4d029483347aad85942b : FAILURE in 4m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/67e6eb9f932449c092c1827a13a6879a : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bc14e2a5672049ec822f1f3058f1d5d7 : SUCCESS in 52m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e76149f489f44f958b67f5ef3ae0f9f9 : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":2},{"id":"3fa761c3048fa1b1aea3d1498992f0462bff4037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-10 02:37:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"88d2320287077cd07d4e6e49085c45f4310f05ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 02:54:59.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e694535a7924e1fbb8885595491f801\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b7954d1c57b6414989c8b153a12ac570 : FAILURE in 11m 46s","accounts_in_message":[],"_revision_number":3},{"id":"5d2d498f211f9f900bfb99045632d85fc0bf6169","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 04:18:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e43f9dcb3d54fe9aab6970ca8292fd0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac1dc95e0a141b588e406778e9e1404 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e71e3bd0be426dafcf762487362206 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aed7eb04e42d4049be19ed02deccfab9 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c30d555b7cd348c7a9af6d0f08e05c8c : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f17e681e8837446989f418ea9af8d6dc : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4de7889775a24288a67b2c24695a7677 : FAILURE in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aa03e64aade14781b69491d6bd3c4225 : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/18bf99bbfd6240e0ae3c3066dfea45bf : FAILURE in 4m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0df4440893084313afb5170c6b493f1f : FAILURE in 4m 27s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/628df2efadf4428087add32fa5e57378 : SUCCESS in 9m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ce754fdace5c422da3e5f334f96afacd : FAILURE in 6m 19s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/30c8d186d4964602988d18c8046ac10e : SUCCESS in 8m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3008c9db2945412fabe099554c6fb5f3 : SUCCESS in 10m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/745edb8a185a4f04911872661f290068 : SUCCESS in 5m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/87b0e0c0808449b68eda19b18d4526ad : SUCCESS in 5m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7bb9e6c904144f44997eda86dd613e72 : SUCCESS in 8m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/946dc0d3bcd145109a80cde10aa4a07b : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d04a66573b624860a396fe6b753f77aa : SUCCESS in 14m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/017da7f3d2f84ba4b0f885cf67ad48dd : FAILURE in 10m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a62eee1c0e0e418589fec41d4b0f5666 : SUCCESS in 13m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cd2111ec228848e78de4736b385b4fab : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ec169f73963c457baf898438794351a6 : SUCCESS in 8m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ebd79dc06e7b4a329706dfaa45763008 : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8903140daf0549aba25c6ff7c4b9c4a4 : SUCCESS in 23m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6c7025e185c14c86acc5c0d07b07a2fa : SUCCESS in 11m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cbfd5a712a80433fa0f1c6dbb0ed490b : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/966594314c654b0b9531ede571413f50 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e62ba4954c93404fa7dc381cca9df21c : SUCCESS in 19m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/de945fb8f93042b3be00220c103c43c7 : SUCCESS in 5m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d18faa419a7640959a43fac25b8ca978 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6feb11016cd14c369cc9fa887d8d1408 : FAILURE in 5m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6e0f21cc6f424191995b988e5df1056a : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2d9a90a4b6fc433d9ad948eca93956c3 : SUCCESS in 48m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c4b5436aaa54f978b1b65f4f468d010 : SUCCESS in 1h 26m 16s","accounts_in_message":[],"_revision_number":3},{"id":"b5a55e2e451311ffea3d8f2a00e29bd43a4a1e42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-11 02:33:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4bd301d6b58510b8e29e487d53cc8ec699314f94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 02:53:38.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc44499f8a5a45308ced370e7a772b6f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d06ecb380144e1b8a9cd5f71ac9c239 : FAILURE in 7m 22s","accounts_in_message":[],"_revision_number":4},{"id":"ec368b900826cd0148f1d8d6dbfedb41c23af95c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 04:14:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a67e84ea4284575831e2b39965a8223\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab92433dbb746059385601b28a17745 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9b80514ab5e4998b1cbdece27baea00 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46a356e53edb48d8a46794609db875b1 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/722472f4e35341cb8a6107431054459e : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cc8d3ff64bf41b6822b450c39c17d83 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/74acfa4fffaa4bb9bf5887f0481a9e8b : FAILURE in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1053e26065e44f179a77ac8c6d470c17 : SUCCESS in 6m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b1584fe32f9476e87fc6a8377a17a91 : FAILURE in 5m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9a8d13cfe6694282914882132a9bdf3c : FAILURE in 5m 19s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/779653c77a064089b7890ed64da70629 : SUCCESS in 9m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1dc4d18ef6cf496084daa897ffc18dae : FAILURE in 5m 43s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/9f738997cb4547ec96513a7ef0498c8a : SUCCESS in 10m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/951276c7799c484889c1b5dcc98b530c : SUCCESS in 14m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/033466e2d3ae43ae8482f0ad718f5c5b : SUCCESS in 9m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/40ad77ba260b4f0f90aa6e112ed4d2e9 : SUCCESS in 6m 01s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a0b4e1da375c469084ae693f61d29857 : SUCCESS in 8m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/eb47fb0cb2b44cd3879f58a683830011 : SUCCESS in 12m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49777a5186bd489baa21a63dbb398223 : SUCCESS in 8m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/77c0ab5f015b40bda98bf511b6029728 : FAILURE in 9m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7d87f1d91c7e44a58d211511d1f8f702 : SUCCESS in 15m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ab9f4ed54ca04607a62c57fa69de6d68 : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f9d5224dee0c421e8d0cde1a7c7f0b9a : SUCCESS in 11m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/307653af60cf4b28aba638e9b188d534 : SUCCESS in 4m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/15a8517ab8284f90bcc3cc64086022ec : SUCCESS in 29m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/220d70c3892448e2b42a22e4184c7542 : SUCCESS in 12m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8ad0921e0b2448aea65f5dac932056cd : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9f14bcc5fdf24bca9492a73e77ab808c : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa8c0bcac96f44cf97be8bd89c2332f7 : SUCCESS in 20m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/593e190149db4f41bbd7688a33492d70 : SUCCESS in 8m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/25eb82b4427348d6b2b5af7b796d091e : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23a3fa4542d64d3a96994f0a9c952872 : FAILURE in 5m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c69284dd96024f2284bacc2414f6d3d8 : SUCCESS in 13m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/868ccb7db5f24257b4ee75dd8e42e948 : SUCCESS in 49m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0cba4732766941fa88e0e646667b4d17 : SUCCESS in 1h 29m 09s","accounts_in_message":[],"_revision_number":4},{"id":"6217665db13463911b7429c0b6a3b3cce6122436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-12 02:40:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"45cfd11e6ff1b9da491c1a2a6258bc2da9e99d7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 02:54:55.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01357aaab40b477cafda019653994ca3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/26fca1b1d1444f9e8adce8be16e7603a : FAILURE in 8m 56s","accounts_in_message":[],"_revision_number":5},{"id":"34b336d710c9e3f7e32e476b0b0c4ec8cf68f7d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 04:21:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66d7aa9752254127a37b3326cf443ae9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f5f0811e69f4fa0a97954d551bece43 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2fab67fb839428e949da894a56a7f8b : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cf7dbb757a84ecaa3ab7fced450e440 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6661ee6a1306460e843e5289ae699ab2 : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/350232c4cdbc40a4afa58e8c3c8dbccc : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/36c7ee7ec29342f9a4731546c836da1e : FAILURE in 4m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6465ab4cbcaf437c8b3590bd6248eb65 : SUCCESS in 4m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8054bf0a4ea4586b6ad1ac2b93c9745 : FAILURE in 4m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/952f978568e943aa9a0567ad312d21a1 : FAILURE in 4m 47s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/43e0c5dcb7054c68bc3bdacdd75b5698 : SUCCESS in 10m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d656154c8fcb4abf97078c806b918f77 : FAILURE in 6m 13s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/338441ba46a247e5a5b3fc4701820b3a : SUCCESS in 7m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e505b071359243588325d32a7a9d1266 : SUCCESS in 11m 46s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f58bb62b1cca491d8977a4131f1fa57d : SUCCESS in 6m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/101dcaacc69443b196a8455020e40acf : SUCCESS in 6m 27s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/07d1e8de73684fd0b99f7b8e7bbda533 : SUCCESS in 8m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/07ee404930404ff0b51e9427ae3f7ed1 : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/356b816de132418983051d49dcf6cc99 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8a36093b00a948e29b6b61ac50ae5b50 : FAILURE in 9m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8ad87f15f70e46e29ca163273d2a2f28 : SUCCESS in 10m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2e22ef7f768c4b55a6f031ec59c54d32 : SUCCESS in 3m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/436308c6e5404bfe86e5cedf1b87af01 : SUCCESS in 12m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6c96279c308643e0821dfd83d496e185 : SUCCESS in 7m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/208a6192cac44821b4c77ae2e4097bf7 : SUCCESS in 25m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5324f3550840454eb443a3d343ee92aa : SUCCESS in 13m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/aa9838cd570247df978d991ae110185d : SUCCESS in 5m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/26a985343fb04307bbf00416bee64185 : SUCCESS in 5m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/22e14acb3502480bad436e9891b86828 : SUCCESS in 24m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/660b5b38d6fd4e398e84ba35bb4cf109 : SUCCESS in 7m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/50248b1af24143bbb89593d64a1797f0 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15f30c0f98414ccb9ff44db2854c5a9d : FAILURE in 4m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e48c4a898ee54a4cb8160f64b32354dd : SUCCESS in 19m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34dc05f98dca4442aa4badcb161f77e3 : SUCCESS in 54m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf5e4f7ec95e49f3acc5b1c21f1d5174 : SUCCESS in 1h 33m 13s","accounts_in_message":[],"_revision_number":5},{"id":"0dd80f84310e7337f1face92b33ed3a3da2e0644","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-13 02:28:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1b180ee479bcc20733f502310ed5871b1c295fab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 02:44:19.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd3c95b085d4b1f878d9c7e42baabb3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6ba20cb89b0646789500b89ab188f9b5 : FAILURE in 9m 22s","accounts_in_message":[],"_revision_number":6},{"id":"b56001a90ae5b93a2157091dd6cf9d9ff527301e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 04:34:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/036609d662b34ebb865767e69c50583e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c16a7e35b504c33904ad0ae99c1ca62 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aa701c89aa147debc61f86ffba22f4a : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61f7e35870024588961c6f7c264b5504 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0a72315f6054c0b8400cc0639551512 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20d96f58f7cb4ccf94a65d7cb9f23d15 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c9d040e236244d5b69703ef01130e00 : FAILURE in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2fb7379bf58740799d68212533f98254 : SUCCESS in 4m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ae40085ca4824bb9bbd533bd38ab764e : FAILURE in 5m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c9e7b6b6e7cf4d70bf06eb760ff84eb4 : FAILURE in 5m 28s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/6857f41e432647a38cbf266fe84d8740 : SUCCESS in 10m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/edd1cbeef152430fafa8f52fdea38d5e : FAILURE in 6m 09s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/fff477063ca1454ea6855c84d857ea14 : SUCCESS in 6m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/56a2627a7d80463eb1dfe87a1e3e9014 : SUCCESS in 13m 03s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d308185470394e3797c16020d14ce80b : SUCCESS in 5m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/43eed0de406d4a258587e4402c6294da : SUCCESS in 6m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a7b07ee1271945b9b6c4a4d1e463f798 : SUCCESS in 7m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c9b912783d494f8ab02b4171b992ee69 : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f09078f72dda4f9ea498e650f6fc103f : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5ce0676dc09d4bb2b463a80ecba32938 : FAILURE in 9m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f089f2bc2ef24eda97565ef84cac1227 : SUCCESS in 12m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6acd3cbdcf1f48e69f306caa7854fa6c : SUCCESS in 4m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/78f7cb8ce2aa4f3cbfa9e5368bb63327 : SUCCESS in 9m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c733d2cffcfb4cc38342270675530758 : SUCCESS in 5m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7b5676330cc54ad9946422006defffb6 : SUCCESS in 25m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/27bc4a9d90934f96b8dd317586dda80d : SUCCESS in 14m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7cb80f26acf8451c9321f48ed7442395 : SUCCESS in 7m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/48681cdc8a0343419f53d2c0573ae0af : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ac070d5183540f193d460364bd269ff : SUCCESS in 25m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4d0b94934265492b9658b12a69c3fee3 : SUCCESS in 5m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dbf5abe7ca7743a68f4909cc94d1f335 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a948e135af32455b9b8a550320dfcfd9 : FAILURE in 7m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6c1e561a72304bba98860203d1f9b798 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a0f28f3b65140ceaa3b943a302d128c : SUCCESS in 58m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/518428ccfd2e46d386435384fb7539f9 : SUCCESS in 1h 53m 08s","accounts_in_message":[],"_revision_number":6},{"id":"ad0c8ae908fd4232e079ed637bfbf52af999a9d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-14 02:27:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1144a8bbbe644e3e5ee65977baee92a2e7a8b4ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-14 02:37:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1be9a75394e627a9370af7fad4ed2af94ba66347","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-14 02:37:38.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":8},{"id":"e0828df69588d549f4a0041070aeb17d1e38af50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 02:52:32.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57be285404214a1f92235b2ab75ebe09\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/77212059b1e44fa39aefe04a2d05bdba : FAILURE in 11m 37s","accounts_in_message":[],"_revision_number":8},{"id":"c8fcb728e964a43b46c2b24955e8036e9929c4db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 04:11:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f52bb40b9734f9d9cf305b6303a3cea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f405f125484d7d82ebefd8f0170993 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33835c9779ff4edfbc11559524383afe : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/857cc34d38904b6a966e248e5899706d : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8811275c207945cfa17bff2c4fb5361c : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ea0a31c2bf44662abe857eda86ef075 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2a08c051573648d0bb2d2379b23a38b8 : SUCCESS in 6m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5188081bfc24f819af8c2ba0c687307 : SUCCESS in 5m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38934563bf204f89b01bab70317b647e : FAILURE in 4m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/642362e80d954fd299c46983938b1b9e : FAILURE in 4m 33s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/7b30063fdace4b3397936dcdab69fd96 : SUCCESS in 6m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/da2990c454944776a47addf9fac51c15 : FAILURE in 3m 37s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c464a9d7d1ac46e080e27fe887164719 : SUCCESS in 12m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cfb4e9e3a8f749ebb29271b77127505d : FAILURE in 7m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/89542d833fd148eea8f2eb211c808cc6 : FAILURE in 5m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5415a37ca1ee442e9e03e3bebd8743de : FAILURE in 4m 22s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/22dfa1fff97e4b8d9fbecd35321dfca8 : FAILURE in 4m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fa4726cbd7f2476bbd9a90a31b682c57 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/791b50082aea48a0a8f0bb0209ab495d : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a4aa28e2edbe45df92938fc4a05876f3 : FAILURE in 10m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/912e466fc1f1443ebf6214a8d101605a : SUCCESS in 11m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7d8bf2a07c7e4c61854076fb0cb83d3c : SUCCESS in 5m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/78deb02007ca4fee8f77810c116647c9 : SUCCESS in 9m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a809987e68024b3da3ac2c1686385dea : FAILURE in 3m 49s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/7f513bbbc48b47db9c477c19b3733683 : FAILURE in 3m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/769e2b4c07ab46e1b821f161c50672aa : SUCCESS in 28m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7296744919254bbda51f066a072fd6cf : SUCCESS in 15m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0757871b32154c7aa90d6245b21882d3 : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c631619cf0af4392b7e12f46c1e510e8 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a0e0e497f4a44c99375f75c6f3f9fee : SUCCESS in 19m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/04341a6f9ef041dbae6291f39e31ee01 : SUCCESS in 9m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/87bf04b7744e418a9e8571e68c6efb3f : FAILURE in 6m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b3ecf12ad01b4d1f8b61398b1cd93793 : SUCCESS in 8m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7887689b391f4079ad222684f4438df7 : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f6a19250ae8465dbe2b72d83d9969c3 : SUCCESS in 56m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6bbcd7202003441096d9d8edc1e30285 : SUCCESS in 1h 27m 05s","accounts_in_message":[],"_revision_number":8},{"id":"6d3d6b41371c2f72c1ef80dce954dcd8c481e758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-14 04:29:06.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"833d19c4e7966364649d5b0d419204dbb9474629","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 04:46:06.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d58ecbfb2a4947c1b1cb3ac3da1762f5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/44f9d663665b4017bd58cee6602ec1ab : FAILURE in 11m 46s","accounts_in_message":[],"_revision_number":9},{"id":"f016daa93439afca4385944654ca8b21f3469205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 05:49:44.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d87a4d94a6624be991261dc407426f30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b3782032ea4db3bb1fedf0f9b85111 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027c38a52ff04a38b69a48ce86e416b0 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f5083fb661f4a13ac24c95bfdf7b06c : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1ddcbc401584b5ba49dc5752704629b : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80b33c3d86cb40c2b855d68d9cbcc97a : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c86ee6dcf9e1463eae7655308061f768 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/25f3c534f1be4be3bde2aefbd620a49d : SUCCESS in 4m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1cc75f4edfc24075b8ea1eaef8544318 : SUCCESS in 10m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a5de87edd65b497a996f17707293c41c : SUCCESS in 11m 07s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a19b3f60d64b4b27b664217da93870f2 : SUCCESS in 7m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/98e707e3227e432299894e95a5d101b1 : SUCCESS in 5m 29s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/28311a5bd6bf4cd08104a44939b87073 : SUCCESS in 6m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2625a8b9b0ed4e1bb9d02b0d1d029c36 : SUCCESS in 10m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/748d80a7dc304c64b46cf28444c565ec : SUCCESS in 5m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/50dbc136cc3b4c208873491a8cf41bfb : SUCCESS in 5m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e09907f55e0a4b5aa9b328b75b15f253 : SUCCESS in 7m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6b85ba792d7d49f3b4950a21f4e2d92e : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b0c04afc3b4e4d7ab8bc252a3e7d15a4 : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8f99101a52874b22acf7fe4798ad0734 : FAILURE in 8m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1ae7ab88ca8940a9a132ded2dae993bf : SUCCESS in 11m 04s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d61c05877adb450cafbf40bc9c212b97 : SUCCESS in 4m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/746aabd35b3e4b4eb68fcd29c177f87a : SUCCESS in 9m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/de4141db5a284bdba5863c97d3cabd35 : SUCCESS in 5m 01s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/85af2d0b814d47388b576722c131eba1 : SUCCESS in 8m 24s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e56dc1f240af4cb7a29484cd98beb911 : SUCCESS in 26m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/96fb1b0a1c014aa98958075bd8a2250d : SUCCESS in 10m 16s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/57ecada6052848aca9dbec4fcf71a901 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/19478591f5784b41b109b0bba3dfd75a : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7c28f5ef125f4f14942c6057f56236ce : SUCCESS in 19m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cfcdb9c0fea04ead9c22082005c3d935 : SUCCESS in 15m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3b22fc09ea344cd0bde1459bfe26ca5c : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1edb8a2d45b540b28c6644c4dfd66a12 : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e40aded4b074400998ca963946789143 : SUCCESS in 17m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73b1ba01721245d1a867b498e0d80725 : SUCCESS in 56m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b073cb5a5b74f86b685f49c20a5fccd : SUCCESS in 1h 12m 43s","accounts_in_message":[],"_revision_number":9},{"id":"7b62ae1083051a6f0624c55e9a349347ffc5f3d8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-14 05:58:09.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0a9f504752cef9f8a082c1ca436246b424a05b76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 06:11:22.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b676958e57c41f49b93d0f0f6d35f1d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1ade659c9103477e9a8ac1da82e4cbc7 : FAILURE in 9m 49s","accounts_in_message":[],"_revision_number":9},{"id":"7676fe3c3a53093b47ed4e599cd7974e0842d54f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 07:19:55.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51fe66328b0d4281a80697cdb47545a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/750b9cb22f1b4ce9bc455c522b45a249 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e82004e055748fb9905c2d4b18e57b8 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/637c149123a042a4b6ef89a34f422e7f : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5b3f90add6e402b9a48f9d5fef16893 : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ef4c79ba1384af7aad6845c9ab3f139 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f666cab5bfff4482bb409c2dc5a75c4b : SUCCESS in 6m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bebcec413ef5464e8b2b61838fb5c928 : SUCCESS in 5m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ea6afa1b30fa47768b68989dab1c2052 : SUCCESS in 11m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5dd57914320a439cad7cae5562740464 : SUCCESS in 16m 04s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/929facc843b043c7a6bee01464de3451 : SUCCESS in 10m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/408944716cc7472d8e6fccc7973e44a3 : SUCCESS in 7m 53s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/a0e006018d7946769b8e2ab9b3b71bcd : SUCCESS in 8m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c6c75592899244b99c4e291b7b2e0e4d : SUCCESS in 12m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9e45531eae4641e085185c815254f15f : SUCCESS in 7m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b0f8ecfe41ba436f8644ecfc40d935a3 : SUCCESS in 8m 15s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/04e51d88472f4a0db542894b3fde0aa6 : SUCCESS in 10m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/00da8589185d476bbaa309c53e1d1857 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a27983f8202c4064a01bacffb0bf37d0 : SUCCESS in 11m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f12421447c90487395368c6802dbd43f : FAILURE in 10m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7dea0829d96b44c49e7ae7f605a196bb : SUCCESS in 13m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/387ab0bd9bac4418b5064a8772560004 : SUCCESS in 6m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f87c3b3cb5fe477296a0fac9f2ffdb25 : SUCCESS in 8m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d7644c14225a4a389a61a05ffc445016 : SUCCESS in 7m 26s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/15a29a832b2741cabbf5f944cb9293b3 : SUCCESS in 8m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f4552b41f8e94ba486166c74f2cc4dc4 : SUCCESS in 21m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6e7950fa5dbc4a8ca0b70f8052837e3b : SUCCESS in 11m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3b6fc36ab15040f1b31f0a327071034d : SUCCESS in 5m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1a441b8196264698bc86297115b48c91 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bce7907f97e453a8885588ec8146244 : SUCCESS in 17m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c578982fb65d4442bc6f502d46bdf987 : SUCCESS in 6m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a9c76de3efc649a784f6ded721502106 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7a8f103f2afa489da9945016a1a35f0f : SUCCESS in 6m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d0f0345f42ce44d7b0c18facf45893cd : SUCCESS in 9m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/713cddc122c44cd4a1b20a2c4fd7b3e7 : SUCCESS in 39m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/730bb381a93340c5a830d9f66237f7e0 : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":9},{"id":"e3e54788cb642a913c28addc3aaf0a99223d173f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-14 07:50:12.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"a3475e68f64bb0923c6688567403b7b93df094ad","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-14 07:51:30.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"312565a180853f62b7b281a57c4799a6c65ef1dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 08:09:46.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db43a5c6c89b48938d09ff5c4187e98d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/84f5e0fc571948da9c3731fa8cfa1b52 : FAILURE in 17m 37s","accounts_in_message":[],"_revision_number":10},{"id":"f527bcf006d9f6a2b14e5135526a2eb98839ca02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 09:26:09.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ede193ed4b940189f6c239a2867e425\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb653fbc14b4bdd8608ba2de73e2ac3 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01f1f782298427ca1f800fb8e9aa96f : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a85930d39600483f84e5b82162b6e950 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac88e682297048b6abf639347bed8d29 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3e35c7e994a4d208ad03033dcf481f0 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c134f27719f14b08a70840af32588207 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/94efae1f26894e0883303e2902b0e0d5 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/de3f9937f24f41c886bcc4107712a0cd : SUCCESS in 12m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/57f5953827b9491283a6b33c7e5b44b8 : SUCCESS in 14m 07s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d269f53d3fba447db48903941512a8ff : SUCCESS in 7m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ac8371b95cf54615be2818e747549bbb : SUCCESS in 4m 32s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/4aceb63d76ff460aa288d2b08b5b419d : SUCCESS in 6m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/daf80d4583f04adaa0c2399ccd2234ea : SUCCESS in 10m 44s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f5794e955a214cb8a34a6e7aeaef5845 : SUCCESS in 4m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d42f593240914cffa336c703362404d9 : SUCCESS in 4m 35s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5f3dd4e5398544148a7aa9b777ed8726 : SUCCESS in 8m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/20d9e8719e2a4fd2b72fe305cf78dd0f : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a137d219a9bc4d2db3de50ba394f8940 : SUCCESS in 9m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/742dd5e0f5d94917b0a502d09a0fd681 : SUCCESS in 12m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c506ba8e8843400685150d4548b38068 : SUCCESS in 13m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/eb5cbee3ab0a4a48a85bd794bfaa9a55 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b7c67fba22e341a6b0a7e17ee2b2d424 : SUCCESS in 14m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6bed3e9aac67447fa77c7d8f972f14d3 : SUCCESS in 7m 50s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/ebd41fa5bdb64235adce1dc003d2511b : SUCCESS in 12m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/46405512024446539d209eedd574007b : SUCCESS in 27m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ce864c01d57040de850c237853eb681c : SUCCESS in 15m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d3cb0570f897426fbf944b32d0f044db : SUCCESS in 9m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/574ccefd64604356a13f9fee3c186656 : SUCCESS in 7m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8fddfff032de4bb19c5d60a0d60a859e : SUCCESS in 25m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9dd18679339143eca01453cc717bacab : SUCCESS in 6m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0f9427a264ec4b879bf11ed71aa919aa : SUCCESS in 8m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05f8b9e1fb0a4761b7505e5809850f65 : SUCCESS in 7m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2a8435ee2b524b39ab4d997276a0e02f : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9b61615a6468418d8885c049fd1c380e : SUCCESS in 51m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a858ce7b5734e9381d77e6abc31323e : SUCCESS in 1h 34m 27s","accounts_in_message":[],"_revision_number":10},{"id":"733aa3c784524dda731a13eef5b0287327f3f1bd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-14 13:43:59.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"ffd75aaec16e3e5e87bafc939cbd868c284a6b83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 13:44:43.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"d40049c0242dcb988a56a40c90bc7d1a61677331","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 15:24:14.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c379dc80dea844ba831f59e221637120\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5262040df764dd4a75b2a55a956c173 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df7f765d38f4cfa9c740cf62eeb763d : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/517cb7e68b2f40218b85c6c7252f34fa : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b69a50aff993449b8bbaa07631ebe6a9 : SUCCESS in 4m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/286f21641bcb45fcafe4ae3ddfae319d : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2e95560a5d684c7c89b716dda75ab52b : SUCCESS in 4m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/11fe924eb67c4335a8f3b4883e929c17 : SUCCESS in 10m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/13eb1edc49bd4a23a497488d7887b1c1 : SUCCESS in 10m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/240e4a8ea81043c6bd896e7a5c98fb31 : SUCCESS in 9m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/aca4d553501646a3958cb578719f42be : SUCCESS in 4m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f7e4befdf4ba4ab5a1e8f0b301f2f713 : SUCCESS in 5m 08s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/843dd10c92ed46679d236ed1e77cd773 : SUCCESS in 6m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8cc0fc4c21b147f78dbd241c6c4266ac : SUCCESS in 8m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b3b15ff6b33467da6b89b2e159fa2f5 : SUCCESS in 7m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/039ecf4a712a468596b34be24a9c5ce6 : SUCCESS in 7m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/098c962708ad4742b75b1e8801a68b5e : SUCCESS in 16m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0af2da9958074fca902ced6a179dcec9 : SUCCESS in 3m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/60cef0dc13a5486ba9870683f112f4d8 : SUCCESS in 9m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9cd4832ca052468980e5b008ee872943 : SUCCESS in 4m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b5552fa194f744b88a2bf1c5d4df9893 : SUCCESS in 34m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f1b4a559e27c46ada447ded6464157b5 : SUCCESS in 11m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f00f326401d4436ea2b8cff8b3c91d19 : SUCCESS in 4m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30d35cd24a044c429567ae1a92e1ad94 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f0dead261f1e47118510d0a5e8ea330b : SUCCESS in 28m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7d0334f3378d4713ab35e7d0f214a7cb : SUCCESS in 5m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cf1abb451b0d49459393761971b72718 : SUCCESS in 4m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2eab26d4b5f54afbaf7cc637779b7cf0 : SUCCESS in 6m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4f16f69125f945dcb658e72ecff6720d : SUCCESS in 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8aafdc6c3d624b37a21a3dca7083336a : SUCCESS in 38m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/647411031c354003a618f9aca9a32623 : SUCCESS in 1h 33m 42s","accounts_in_message":[],"_revision_number":10},{"id":"01fb984689dcf3f0171caa6d1ed5111c3372a7a4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 15:24:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"c79ced86194176436bf186e1c97ddc66119ec4c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-14 15:25:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e5e99b1d89845e6bbb88ac94913b1a3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eb85e51fcc14896881fca33457cc3f8 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea","revisions":{"ff77e175fdef04087e02984774e19ac5365de633":{"kind":"REWORK","_number":1,"created":"2022-08-08 02:31:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/852370/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/1"}}},"commit":{"parents":[{"commit":"da09b605c21450b0abf32fe7b64a3c38740b8a74","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/da09b605c21450b0abf32fe7b64a3c38740b8a74"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-08 02:31:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-08 02:31:43.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff77e175fdef04087e02984774e19ac5365de633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff77e175fdef04087e02984774e19ac5365de633"}]},"branch":"refs/heads/master"},"43fb41bce1f126564cfd1c53fb88d019794471a2":{"kind":"REWORK","_number":2,"created":"2022-08-09 02:41:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/852370/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/2"}}},"commit":{"parents":[{"commit":"d79553627daede71690782a98d6418102e4ad93e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d79553627daede71690782a98d6418102e4ad93e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-09 02:41:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-09 02:41:28.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43fb41bce1f126564cfd1c53fb88d019794471a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43fb41bce1f126564cfd1c53fb88d019794471a2"}]},"branch":"refs/heads/master"},"e5ba4eaf786ec98a2349c8945c2c7df0ac4319bc":{"kind":"REWORK","_number":3,"created":"2022-08-10 02:37:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/852370/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/3"}}},"commit":{"parents":[{"commit":"d79553627daede71690782a98d6418102e4ad93e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d79553627daede71690782a98d6418102e4ad93e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-10 02:37:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-10 02:37:03.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5ba4eaf786ec98a2349c8945c2c7df0ac4319bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5ba4eaf786ec98a2349c8945c2c7df0ac4319bc"}]},"branch":"refs/heads/master"},"2436cab0f0d9bb4be9e36f6333455c29b5b9bef8":{"kind":"REWORK","_number":4,"created":"2022-08-11 02:33:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/852370/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/4"}}},"commit":{"parents":[{"commit":"b12966c487c055f1830ae257ffd3489b691cc830","subject":"Update code to use a std. read_requirements_file()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12966c487c055f1830ae257ffd3489b691cc830"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-11 02:33:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-11 02:33:45.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2436cab0f0d9bb4be9e36f6333455c29b5b9bef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2436cab0f0d9bb4be9e36f6333455c29b5b9bef8"}]},"branch":"refs/heads/master"},"9ae52c76bc388b3650320757be3bb3033a43afac":{"kind":"REWORK","_number":5,"created":"2022-08-12 02:40:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/852370/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/5"}}},"commit":{"parents":[{"commit":"9f9647adc948a5dbb8334a7a55545af4ec4b167b","subject":"Merge \"update constraint for cliff to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f9647adc948a5dbb8334a7a55545af4ec4b167b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-12 02:40:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-12 02:40:33.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ae52c76bc388b3650320757be3bb3033a43afac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ae52c76bc388b3650320757be3bb3033a43afac"}]},"branch":"refs/heads/master"},"7aec9441b921f3197a9f541fddac7875c167eb44":{"kind":"REWORK","_number":6,"created":"2022-08-13 02:28:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/852370/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/6"}}},"commit":{"parents":[{"commit":"9f9647adc948a5dbb8334a7a55545af4ec4b167b","subject":"Merge \"update constraint for cliff to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9f9647adc948a5dbb8334a7a55545af4ec4b167b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-13 02:28:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-13 02:28:16.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7aec9441b921f3197a9f541fddac7875c167eb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7aec9441b921f3197a9f541fddac7875c167eb44"}]},"branch":"refs/heads/master"},"7423894d637962ade021feb9d0c8814b22328b91":{"kind":"REWORK","_number":7,"created":"2022-08-14 02:27:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/70/852370/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/7"}}},"commit":{"parents":[{"commit":"95bd13c40ad18e4fbb52c754dbeafcb6b408ac67","subject":"Merge \"Add cross project tests for Mistral\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95bd13c40ad18e4fbb52c754dbeafcb6b408ac67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-14 02:27:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-14 02:27:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7423894d637962ade021feb9d0c8814b22328b91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7423894d637962ade021feb9d0c8814b22328b91"}]},"branch":"refs/heads/master"},"34cdf2902e60a0c0d353ef7b0ba03130e5fa854a":{"kind":"REWORK","_number":8,"created":"2022-08-14 02:37:33.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/70/852370/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/8"}}},"commit":{"parents":[{"commit":"95bd13c40ad18e4fbb52c754dbeafcb6b408ac67","subject":"Merge \"Add cross project tests for Mistral\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95bd13c40ad18e4fbb52c754dbeafcb6b408ac67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-14 02:27:46.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-08-14 02:37:25.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34cdf2902e60a0c0d353ef7b0ba03130e5fa854a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/34cdf2902e60a0c0d353ef7b0ba03130e5fa854a"}]},"branch":"refs/heads/master"},"95e1bfad700a6b5a33b9f24a3e7f05011203dd41":{"kind":"REWORK","_number":9,"created":"2022-08-14 04:29:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/70/852370/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/9"}}},"commit":{"parents":[{"commit":"95bd13c40ad18e4fbb52c754dbeafcb6b408ac67","subject":"Merge \"Add cross project tests for Mistral\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95bd13c40ad18e4fbb52c754dbeafcb6b408ac67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-14 02:27:46.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-08-14 04:28:58.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95e1bfad700a6b5a33b9f24a3e7f05011203dd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95e1bfad700a6b5a33b9f24a3e7f05011203dd41"}]},"branch":"refs/heads/master"},"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea":{"kind":"REWORK","_number":10,"created":"2022-08-14 07:50:12.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/70/852370/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/852370/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/852370/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/852370/10"}}},"commit":{"parents":[{"commit":"95bd13c40ad18e4fbb52c754dbeafcb6b408ac67","subject":"Merge \"Add cross project tests for Mistral\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/95bd13c40ad18e4fbb52c754dbeafcb6b408ac67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-14 02:27:46.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-08-14 07:50:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I81c77f1775691e492ae410989713341aaeadcf5d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
