)]}'
{"id":"openstack%2Frequirements~852651","triplet_id":"openstack%2Frequirements~master~I177d0e457ddbe4c29cc926ef0f03043751903ca6","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I177d0e457ddbe4c29cc926ef0f03043751903ca6","subject":"[DNM] Test horizon with JQuery-Migrate\u003d3.3.2.1","status":"ABANDONED","created":"2022-08-09 17:30:51.000000000","updated":"2023-07-19 14:12:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"921ef0efdf0b2cd28e66ea877681cc1134812082","_number":852651,"virtual_id_number":852651,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-09 18:48:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":-1,"date":"2022-08-09 17:31:23.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 17:54:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-09 18:48:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76856e95c4054f375aa8efc614ad31e9d4693107","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-08-09 17:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"673f2640c7a6b2ed293cadd7ff371ab4de57e462","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2022-08-09 17:31:23.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"ef018436cf0508d60cc9fca130f2824d4eb7d204","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 17:54:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0b214d65d94437697f8ca354f972db1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/20f8cb065673484dbf78118a9e7942d4 : FAILURE in 20m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d4689504a7b3b8f1d1042d1212b20f184c657eba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 18:48:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954333c89bf140fdabd445eea2112a42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8853565f68470c84ffbe15a8dde70c : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ec5ed0c31349dab88cad1a646220ed : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8be9c266da184880a30f06b5cd2eb4ab : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fd0206f3a804f2596c8161188ef2285 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d7835eabb9f4f1382ca495a852ddcf2 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6739699c85b4dbe928006cdf64e182b : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7ae7fc11528046b59895e1d80262c0f1 : SUCCESS in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e733c3c7dcaf4df689ae8e367612a8dd : SUCCESS in 8m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b2425cfbd91144bd8b5e2167211a511d : SUCCESS in 12m 45s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/c93fc19123104e3b97e200d923340c7c : SUCCESS in 9m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/30cf890563d84c12a984cfc1b7b8d4fd : SUCCESS in 6m 11s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/680281562330431ca6b24c30aeadcd99 : SUCCESS in 6m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/305be60453fa41a9a79a3654cb062470 : SUCCESS in 11m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8260f2649ac94cc18354ff0b0eae39c8 : SUCCESS in 5m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6e9d76d0f93c464da83b4af61c796d64 : SUCCESS in 6m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/62fd53efd56644d883b46e72c3140e60 : SUCCESS in 10m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/071960823e5c446889331191f6dbefe5 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c207816d5fe849908fbc79cb79f6c266 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b5e71d8fa0014806a8e810280a7be50f : SUCCESS in 8m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c42a66be7e3e4e219c1a8a0a5b60e8bb : SUCCESS in 12m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6418066e55904a13a282072daa0e1f30 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f907b8a5150447d9a6f4ca9644d4d8c6 : SUCCESS in 12m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/da638b958e5a4fdbadacc276334afaa9 : SUCCESS in 5m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8068b46ab2c74b9c8aa65597d8efaf3a : SUCCESS in 22m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f27f0656fc814c1c98aa153710b69c13 : SUCCESS in 13m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1efe6551296e48c985d7ec813e53469a : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f72fa44a7a6d4a2aae574b0d982b7767 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82747f4f55d04fcf87ae7e7704f0cf02 : SUCCESS in 19m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/971b1ec444184d4b8674192308741ed1 : SUCCESS in 7m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2390caf174f241f1a57635c192fdd576 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7bc9f0784358403682768901c9f36516 : SUCCESS in 7m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/11e20369234745dbabd1311977fc4017 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb058efb039f48afa6feedb2adc4e947 : SUCCESS in 50m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6063b56aeadf43f0b045f29d51c18a48 : SUCCESS in 1h 15m 15s","accounts_in_message":[],"_revision_number":1},{"id":"921ef0efdf0b2cd28e66ea877681cc1134812082","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2023-07-19 14:12:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b38b614e9031336a5c2a6883326a2cc5c7de7bbd","revisions":{"b38b614e9031336a5c2a6883326a2cc5c7de7bbd":{"kind":"REWORK","_number":1,"created":"2022-08-09 17:30:51.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/51/852651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/852651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/852651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/852651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/852651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/852651/1"}}},"commit":{"parents":[{"commit":"d79553627daede71690782a98d6418102e4ad93e","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d79553627daede71690782a98d6418102e4ad93e"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-08-09 17:27:31.000000000","tz":330},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2022-08-09 17:30:48.000000000","tz":330},"subject":"[DNM] Test horizon with JQuery-Migrate\u003d3.3.2.1","message":"[DNM] Test horizon with JQuery-Migrate\u003d3.3.2.1\n\nChange-Id: I177d0e457ddbe4c29cc926ef0f03043751903ca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b38b614e9031336a5c2a6883326a2cc5c7de7bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b38b614e9031336a5c2a6883326a2cc5c7de7bbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
