)]}'
{"id":"openstack%2Frequirements~852840","triplet_id":"openstack%2Frequirements~master~I030eb18e723c4d479990a4e2e30b77a4d2712a7d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I030eb18e723c4d479990a4e2e30b77a4d2712a7d","subject":"update constraint for neutron-lib to new release 3.1.0","status":"MERGED","created":"2022-08-11 09:47:14.000000000","updated":"2022-08-11 14:30:13.000000000","submitted":"2022-08-11 14:29:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852840-new-release","meta_rev_id":"f30281785b3750b04a5ee2b95bf438dd1505a170","_number":852840,"virtual_id_number":852840,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 14:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-08-11 10:47:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-08-11 12:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-08-11 12:54:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-11 09:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-11 10:47:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-11 11:23:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 12:54:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3d31a838a29451bce7f097d79bc6010fc20b2d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-11 09:47:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f095cb0beb00d5c57573e7457885eb50f10431c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 09:59:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82582394bd40430799aebb2895dd2d2c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f71c8b5e086443649bdcaff32ff11c5c : FAILURE in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f485b781583ebd6fc1e59d0f02ae58ae050fae66","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-11 10:47:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0b50da6423b3dbc76974e6578edb3c01625fcf62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 11:23:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0d29b9a6e3346a49c63304831df64bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99faf490d4eb4c5b99c0859f6711f776 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b290475807d49059c8485294136d44b : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf04f8b38175456a9abed0ca7afc63a1 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c97ecd57552d43f7b792ffb1e177030c : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7245807383d84ab0b468175b7da36f6c : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/88a6d725ee0f45618c9c8e40473f80ef : SUCCESS in 6m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cf05ff28734140628483df18d836d763 : SUCCESS in 3m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce3145b3f2394f679842278f56764e5a : SUCCESS in 9m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/503aeaa654344e9ba4ae71a4573b4a27 : SUCCESS in 12m 11s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/2957eac4c87a4e5b8a73f028e000697e : SUCCESS in 7m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4b96aca8f75b448ab76fa05e21e92868 : SUCCESS in 5m 50s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/b577beedb8c245b593a9c7123d513413 : SUCCESS in 5m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d8dd1b4ba1574cc08571de33d26ca286 : SUCCESS in 11m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/75e01f95b2944fcf97528178ab891530 : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e832ab401ca549c0b1e79b1af3adb0c8 : SUCCESS in 6m 10s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fe78baef43944ff7a8bc0eb50172581e : SUCCESS in 7m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8f197e5cbce143fd9238107392b3695a : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/119882f00e034d94a5cc50d45379ef01 : SUCCESS in 6m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f417a4a90c6548799aa1340b0068d31c : SUCCESS in 8m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fcd67382663448deba6ff14733655fd7 : SUCCESS in 9m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/44a53b7bf1b1406b8565d4296d1f5095 : SUCCESS in 4m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4796d8e41822416faedbcf681e5cfa71 : SUCCESS in 8m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ad99dd9de54448d5abfc40a9d0b8733b : SUCCESS in 5m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d2fde8a88e094101b331ad3d6f1eed4c : SUCCESS in 24m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/529894f677fb441d8068c0a63d41a92f : SUCCESS in 12m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f31edf6361e74b74af59cc4e9ba42d90 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d0b19ec21c4c4a6689ecd04cbaa71663 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/373d8b1ef7d84d1eb554ee2d1789908d : SUCCESS in 21m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d06c8488da5e4ab58d68754adf123bec : SUCCESS in 7m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/feb8e1810baf43219738834ff78a3572 : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2b6715a8ce14f5886327e0bb2aa5447 : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/20fad12096f04654830e407e040462b7 : SUCCESS in 13m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2e6293f78ee4d65a8e76c051daf9287 : SUCCESS in 45m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfcc162eb4c74f63a143f1b8454a6edd : SUCCESS in 1h 26m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0d63b2b169bf3bcea4965c4ed5ca757de2a4d90f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-11 12:54:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3aa9b7511a186637cb7e2fc3903dcc0975f27208","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 12:55:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6a77ad843cde0a5a2fa0b85998f265a5a06b2b6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 14:29:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/741067893af442269f66bcb7cb898033\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a9aee699b37437c829f8a6b1f396b8e : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c2c4359d9f94591b68beecad7d5be62 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1371f835d96d4f7a8fbb2f4b544e4271 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2684a8efd814ec98926f56752f37933 : SUCCESS in 7m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dde40b0f5d08415ca5accd6f46e906e9 : SUCCESS in 6m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/53d8fb87a1284d60b9178926ea325d62 : SUCCESS in 6m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/edc30ba60e464230b9dc4949690bdc2f : SUCCESS in 11m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/370cb5d8e37e44a9bf84d4dfa8237a93 : SUCCESS in 12m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0e503c88b7994a998689d545fad42a33 : SUCCESS in 10m 08s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/df6633669258481684dc1d5bb844195c : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/875ef25e27e0409db2518c6362372902 : SUCCESS in 6m 56s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/475affd090084697925f8baea047583f : SUCCESS in 9m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a672fef8d8ab48a187f0708f1729a2ce : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78db85b8b0f54bf4bc8dccc0e9586555 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/67f9032b68904a53858ae649dbd24fe9 : SUCCESS in 9m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b875c8e87a374588b77685104d062bbc : SUCCESS in 13m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3c4ebe30bdd14df58088e8f025a7ccda : SUCCESS in 4m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0a7f93dc1ac1424fb70a1f0143008b3a : SUCCESS in 11m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/54bfc23a6c864d5991f75343fb83cdf9 : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/47e73efbb10c472eb0599bad345e92c0 : SUCCESS in 23m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d6a7d2534fb949d5b71ea04210d22193 : SUCCESS in 12m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ef3c072e10664a369b0a5ed62ce25a2a : SUCCESS in 6m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa93b8929e1c4dff9c43df3f3c761f26 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/75526c18c51649dfade076767194baf1 : SUCCESS in 19m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6790b64546e94514b113faf527355775 : SUCCESS in 7m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6877653a3a9345f0b2d9e56ba352e21e : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eab260a65e3945c6bba6b3585d7c1349 : SUCCESS in 7m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3f1afb86bcc94f9b8efaa98792bcd81f : SUCCESS in 14m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/129a0e9027f64eae8081fcee6055a519 : SUCCESS in 51m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1e05463eacc45c9b59f0ec1f9c69590 : SUCCESS in 1h 30m 35s","accounts_in_message":[],"_revision_number":1},{"id":"da40cef6c29d88cc25a0ea7cbf89c0806fa3be91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 14:29:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f30281785b3750b04a5ee2b95bf438dd1505a170","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 14:30:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f6407c2980745dfab07bd5d162445f3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8d054c518da4c8291f8709791b43c2f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ecef1c67c60ad4f8082c973d12ee75f72f38c5d","revisions":{"9ecef1c67c60ad4f8082c973d12ee75f72f38c5d":{"kind":"REWORK","_number":1,"created":"2022-08-11 09:47:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/852840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/852840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/852840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/852840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/852840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/852840/1"}}},"commit":{"parents":[{"commit":"b12966c487c055f1830ae257ffd3489b691cc830","subject":"Update code to use a std. read_requirements_file()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12966c487c055f1830ae257ffd3489b691cc830"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-11 09:47:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-11 09:47:11.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 3.1.0","message":"update constraint for neutron-lib to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I1b32306ac47e9a49ae16581f376a156b6a65255d\nmeta: release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Code-Review+1: chenwei \u003cjzpcoming@163.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: huang liujie \u003chuang.liujie@99cloud.net\u003e\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I030eb18e723c4d479990a4e2e30b77a4d2712a7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ecef1c67c60ad4f8082c973d12ee75f72f38c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ecef1c67c60ad4f8082c973d12ee75f72f38c5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
