)]}'
{"id":"openstack%2Frequirements~852860","triplet_id":"openstack%2Frequirements~master~I971246893fc0d528b16e76acce4b33d2f1e1aa95","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I971246893fc0d528b16e76acce4b33d2f1e1aa95","subject":"update constraint for pbr to new release 5.10.0","status":"MERGED","created":"2022-08-11 14:10:21.000000000","updated":"2022-08-11 17:33:13.000000000","submitted":"2022-08-11 17:32:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852860-new-release","meta_rev_id":"2aef08406729122f93735b5946e04c3d0fd0a646","_number":852860,"virtual_id_number":852860,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 17:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-11 15:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-11 15:52:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-11 14:22:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-11 15:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 15:52:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f9f1051a1634755bbba8273a4d494b2b49924b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-11 14:10:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16f662fa3c74bea7179e27f048cd7452df08388d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 14:22:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9b01318281049d8b097272c02e2b5a7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a9222c948d3844a3bae69db74529b158 : FAILURE in 10m 55s","accounts_in_message":[],"_revision_number":1},{"id":"664f20ae0af814721da348dde395be90b787ce9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 15:51:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88eb5047cbcd485188cc32f2023304c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cdb6fee71ee49b18732ccd6ce326af6 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8036803654c4bb18a84afa5ee11da27 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb566b01b7144198a2cb4bc9f5abdf62 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe87758803f14e07a69de79b4118c0a4 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/342af28521964c8ab8dd676c73083330 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9462588edfad44a7838b41e2d5df02e5 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8dc262fe1f35400997a42595697eee3a : SUCCESS in 4m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/644a8e544a1c42508014df2054cb9275 : SUCCESS in 8m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8616b95514414737992dd75b5c75109a : SUCCESS in 12m 08s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f7fa66a2d9f2471380d2d455db28cc3d : SUCCESS in 9m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c79540ba3063490cb27bf5e77f758943 : SUCCESS in 6m 01s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3490b1b586ce41419d66bc324585f3ff : SUCCESS in 6m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8e99eb9f05c1487e9d4bc0dba1946e77 : SUCCESS in 11m 36s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/922ac951d0064344b68b5bd70b714369 : SUCCESS in 5m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4748d2aae5d64140b5d6d13bac2e66fd : SUCCESS in 5m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/08acc5df17a34cec96cd6cf425a33c0f : SUCCESS in 8m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dc64b9ce6558420685e0be4727c2616d : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08bbe21939fa4d8da9b300e121df0c29 : SUCCESS in 10m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fb4019601cec4fa2a146dbe07af0d01d : SUCCESS in 9m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/901915e35c984dc9b690996a2cac3c81 : SUCCESS in 13m 07s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b8c799a7efd5441db3b0cad29fd2ef87 : SUCCESS in 6m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7858b681e5da4270ae220260d85ffe83 : SUCCESS in 12m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0e2a69cc8d5146218b2202f13823b27f : SUCCESS in 6m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b62f292ee3a249b6bf4921becd339441 : SUCCESS in 30m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6d2d19a30a6148a68d2b7cb6aeda20c4 : SUCCESS in 11m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/20e89a46277b431a90e933ed29ffc968 : SUCCESS in 7m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6973cec2809346e086aa8e3bc065c478 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/21d54e02638c40b88c8cf1a5cad96fb5 : SUCCESS in 21m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/71982de6397845afb21b9cdf18741e39 : SUCCESS in 7m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8d52ad8a93fa4d9e95ac8f898bc32bc8 : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d5d4ba943caf4569982d5b161335af6c : SUCCESS in 9m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/398cf0546a6b4ffd86ab97544d2b533f : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47a1708227bb496f92d4fdb1f0be2dc5 : SUCCESS in 45m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bcfe7664a2494312b125d34e57c1eb58 : SUCCESS in 1h 38m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9b36add4a1a1b808f33d3bf364b57635702e49c7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-11 15:52:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fcadf94dd380528dfd615af1e83545d6cb31be11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 15:52:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7713c76de22a81b3cc0c0cc9000e1b8cb0175a1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 17:32:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86f6fdc30db9471c87a7ed52934d767c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fecd8e510ca248048e31257e6900a61a : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/058d83a5807e4401ab08ce31ed77971a : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17d1bc4e44e943f581d44fe8796979b4 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ade00dbf4bc40089379f6d142032a28 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/682e39480b5046d8ba3f299dff81eb9c : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/73ced778d2ab4b8493447b982db95d7e : SUCCESS in 4m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/598165941f4642518671ec5d624a9b1b : SUCCESS in 8m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b382503b92734e41bee7f7ad07059573 : SUCCESS in 12m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a725a570c08c4d95944e8ca0912f5df5 : SUCCESS in 11m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3d37cdc111174cf78d149de8caf3aef5 : SUCCESS in 5m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/556b8a51d85640058ef0ff0483042f11 : SUCCESS in 6m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/51f4b53c55cb4fbc8c6eb86419528503 : SUCCESS in 8m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2432344e2f7e45119447f254ce9a019a : SUCCESS in 9m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/811da75b87d9479f8f350f0695c12b79 : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f90a70ffb285415c8ece01f022d7c8d0 : SUCCESS in 11m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8d600587c1074d9bb3a5f65f37e4047a : SUCCESS in 14m 07s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/572b878d1bb94e57870fdc333290d12b : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eebed0e5b35f46da97a1fc6c2bbb73e9 : SUCCESS in 12m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4500c55d66714fa188b4d6e0f1d42ece : SUCCESS in 6m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/db5df44dec354bfdb69a5bd6d5c6f9f9 : SUCCESS in 24m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6d2d120755ba47b981fb22048250be23 : SUCCESS in 13m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/be755376b95e4070bd5f21207a3e16ca : SUCCESS in 6m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a768d50cd3ac479fb12f36f1cffef4a3 : SUCCESS in 4m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ee3cacb531b945b38098e767a4b4eae7 : SUCCESS in 21m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/038cbf65ef62437b8d490171e9a45346 : SUCCESS in 7m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2144f3dc8bcd45069a5d229301f05aec : SUCCESS in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d783fb27c26425a82e441a71af228e5 : SUCCESS in 7m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b7bf1bccb9e7428eb1d6d1a2d7b63c04 : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f06f87c0b5234a2c9450b36bf1b35d65 : SUCCESS in 51m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/556760280b534864a9f1f9a239c7df1e : SUCCESS in 1h 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6ee42ad5511ab5e881c57fc9cf731a91342ebf79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 17:32:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2aef08406729122f93735b5946e04c3d0fd0a646","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 17:33:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd9cd0d0e77044d09cbc5a9fdc97c351\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9750793bed48cf9befde83c3be344c : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a441b28147799996b06a41dfecb6d1870a3f4cfd","revisions":{"a441b28147799996b06a41dfecb6d1870a3f4cfd":{"kind":"REWORK","_number":1,"created":"2022-08-11 14:10:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/852860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/852860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/852860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/852860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/852860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/852860/1"}}},"commit":{"parents":[{"commit":"b12966c487c055f1830ae257ffd3489b691cc830","subject":"Update code to use a std. read_requirements_file()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b12966c487c055f1830ae257ffd3489b691cc830"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-11 14:10:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-11 14:10:20.000000000","tz":0},"subject":"update constraint for pbr to new release 5.10.0","message":"update constraint for pbr to new release 5.10.0\n\nmeta: version: 5.10.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nmeta: release:Change-Id: Ia129b7562e5efeba831fa33f7f79d4e02bb8f330\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: chenwei \u003cjzpcoming@163.com\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I971246893fc0d528b16e76acce4b33d2f1e1aa95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a441b28147799996b06a41dfecb6d1870a3f4cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a441b28147799996b06a41dfecb6d1870a3f4cfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
