)]}'
{"id":"openstack%2Frequirements~852935","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I76136e02b2311fe9fd71d3eada4d2020d63cf550","project":"openstack/requirements","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-08-30 17:49:27.000000000","reason":"Reviewer/Cc was removed"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-08-31 14:54:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I76136e02b2311fe9fd71d3eada4d2020d63cf550","subject":"Constraint setuptools for Python 3.9","status":"ABANDONED","created":"2022-08-12 05:06:21.000000000","updated":"2022-08-31 14:54:06.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"04092b0cbabec4dc76113b4fb576ae569586d389","_number":852935,"virtual_id_number":852935,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-12 06:28:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-2,"date":"2022-08-14 20:26:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2022-08-26 02:56:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-12 05:17:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-12 06:28:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-12 06:35:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-08-14 20:26:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-08-25 19:11:38.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"CC"},{"updated":"2022-08-26 02:56:05.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-08-30 17:49:27.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"REMOVED"}],"messages":[{"id":"1f2e0ad3c30ae6127d9aa6943a4f34572d3a509c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-12 05:06:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49753a2db5e8d00f631ecaae9f2b4f61969ccfd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 05:17:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f661a57a8f8340ad87938e8dca42ba26\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6d80f90451374fdeb7ec1959f060090e : FAILURE in 10m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b6c4eca0ef1b0b4bc2a559440db8c44f8d671212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 06:28:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9e975692ec74811a2fcef6e1fc1c20a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a6082dfe6749c6844f81447f8178fc : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff590c07ae05430580f1a1064e04c733 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94db5d2b0c31477e9707d2fcfce18a8d : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1727adbd2fbe4c558aca27ade6d7d21b : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa619df33d0b4c198016b17c465a7599 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ea666e7cd76549bebb2cb6619cffb3c9 : SUCCESS in 9m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cd2a203813944012bc849b775db9f4cf : SUCCESS in 8m 56s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/82c9c91a35ec48c8ab8fd12ea9e9b722 : SUCCESS in 5m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bd6a2a04636e413a8463318b905e1c0e : SUCCESS in 10m 43s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/694b1dff9d6d4f3fb786077d09c67a93 : SUCCESS in 4m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0d5653a763244e43aa423c950a7781e0 : SUCCESS in 6m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ea460b1ed28a46bb8a0b7a2ef4801b56 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2728d1388e7e4cb4b8d0ff29c444e594 : SUCCESS in 8m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d7577f9a45564086ba005a7e92bae86f : SUCCESS in 9m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/692d5f0854d645ed9f7ebfa4c8ec7776 : SUCCESS in 25m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9e4394bcea1f4a18920d3227c89b8190 : SUCCESS in 4m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9afc1d1ec7574db1a8ba81a205ee6a25 : FAILURE in 23m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bafd98566b42452582696292f3d8698c : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cee9e67b77bc4fd6a06857df6af06124 : SUCCESS in 29m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5f55b38112c34783a55a6cea0827c544 : SUCCESS in 12m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0cc415db5c4f4a0cbcd2aa2704065e1c : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b86985f8df04aa29555c7d6a79c776f : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/abf2c3897af140dba3db578306067250 : SUCCESS in 21m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2608bf0c332d4a9ebcecb8c12104972e : SUCCESS in 7m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b0bc6617757b4f7caa548b131eea1640 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/329a32c49e6b47c48691d2624e643aec : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fdcdf80b82f344b5a3a2aaaf8fad250b : SUCCESS in 15m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcf14228fd044e41a6a8ba3024242e26 : SUCCESS in 44m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c1a03c6074e5498db39f057bf2ed989a : SUCCESS in 1h 19m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a80a5c587aa7581832fdf25ad794bd62e0d47d4c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-12 06:35:48.000000000","message":"Patch Set 1:\n\nfile is programatically generated so will be remove in the next update, to get py39 in upper-constraints we need an image with all the diferent versions of python...","accounts_in_message":[],"_revision_number":1},{"id":"747bc7bfb1c8d53d49cb011c59c00946f2210891","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-14 20:26:38.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"f1fbe73798872247cba11bf3fe301660c2e8228f","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-25 19:11:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1d98b0d1c20b383bcaa27ac424b52263c258e130","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-26 02:56:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"65318080cae2e1ed073842f975bc4d14ecce8699","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-30 17:49:27.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_13177\u003e.","accounts_in_message":[{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}],"_revision_number":1},{"id":"04092b0cbabec4dc76113b4fb576ae569586d389","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-08-31 14:54:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"24da5881c46b608ea1853fbb091b042de2a176df","revisions":{"24da5881c46b608ea1853fbb091b042de2a176df":{"kind":"REWORK","_number":1,"created":"2022-08-12 05:06:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/35/852935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/852935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/852935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/852935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/852935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/852935/1"}}},"commit":{"parents":[{"commit":"b10648cba7395323492b9360f6023fab9dd42cc5","subject":"Merge \"update constraint for neutron to new release 20.2.0\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b10648cba7395323492b9360f6023fab9dd42cc5"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-08-12 05:05:39.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-08-12 05:06:05.000000000","tz":120},"subject":"Constraint setuptools for Python 3.9","message":"Constraint setuptools for Python 3.9\n\nChange-Id: I76136e02b2311fe9fd71d3eada4d2020d63cf550\nNeeded-By: https://review.opendev.org/c/openstack/aodh/+/852769\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24da5881c46b608ea1853fbb091b042de2a176df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/24da5881c46b608ea1853fbb091b042de2a176df"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
