)]}'
{"id":"openstack%2Frequirements~853214","triplet_id":"openstack%2Frequirements~master~Id962baa2032c22f3d09e3985b9f3253d1994d48d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-08-22 18:20:31.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Id962baa2032c22f3d09e3985b9f3253d1994d48d","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-08-16 02:23:05.000000000","updated":"2022-08-22 20:04:06.000000000","submitted":"2022-08-22 20:03:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853214-openstack/requirements/constraints/noclob","meta_rev_id":"a5ac7135160f8ba63a56acf595675782f3ee95e0","_number":853214,"virtual_id_number":853214,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 20:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-22 18:20:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-22 18:20:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-16 02:52:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-16 03:56:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-22 16:09:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"877ce1ad4d103f021bc9b98b667da4fd1d79e12b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-16 02:23:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61bd6ecd949a9ad8fbd6ec5334e88cdd7fa8d808","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 02:52:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce074f4715524e6d98e33a72b60f89c3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/841ed27c4f094de5a9936d6ce969ed0a : FAILURE in 9m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0b9beca40001c9ab940ff3890136554a77f7de25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 03:56:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/220025e487d943e5890262f3f5d8359b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5d905ead9c4059afe1f1d8d82854fb : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb6b37aac3d4446af077c27eae805d6 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/798f51afd85f49ca93b0d06283b66c15 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d00ecb4f15c2496799865e15eee51e86 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04701b0bb0c24634bb145e9dc577ffb8 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50b5b0b983c64180b5fd4895db354fb6 : FAILURE in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/572be65ebc5e4b539b9746f49673b8d8 : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1be7c7310cb54c1ab7f66cccfaafcc9c : FAILURE in 3m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3b717e88bfc24af8a37d64f664b71e57 : FAILURE in 4m 35s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/44027566e7714c078d8e8c2e2ba2f52f : SUCCESS in 7m 36s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d07d8c8ba91f45428ae73027653d9e39 : FAILURE in 6m 19s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/960cc120f85740ad9f0cc05329082e16 : SUCCESS in 6m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2a6096f313894ecdb2ca36d7b6e9a773 : SUCCESS in 10m 03s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9df58ca4b13d4d28b69ac59003d1e7db : SUCCESS in 5m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e0a335edce334596acb848bd6ae0b0e4 : SUCCESS in 5m 53s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/24486497e24741ef9567498a8d185656 : SUCCESS in 8m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/06271772f76f4be795a5fc7135009391 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4659678b3974a6ca683228ba915a998 : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f39ed51d678e442fafef08676135924f : FAILURE in 8m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ac076f129a4349c3a498cabd75f1916c : SUCCESS in 14m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/644f57451bbb452cb95b256b2365bdd2 : SUCCESS in 3m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/48ec27ef3b4f4b5989aa8076b8552074 : SUCCESS in 9m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f663940255c848f5a1ad5719a2a8c0ee : SUCCESS in 5m 34s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/e9e1abb11e5447bebf875d56f92254ed : SUCCESS in 9m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/54e91b7123f44f1fb797d6d6050c786d : SUCCESS in 26m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/47a7b2a22eb347d29c9d0b96c9c3d55e : SUCCESS in 15m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0e2a4aa95a0a4a4dbbc26139f4b4da5a : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6a01a6ec117e4735aec266648533fca1 : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1935e92c25014bdcb436a6dc285d7688 : SUCCESS in 20m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6b5bf4954188464ab05f4b2bfb2c88eb : SUCCESS in 8m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/65ae8a157817446a9c01cd4b2bc4f166 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/48813599684a478c8ac1d24ff97e9440 : FAILURE in 5m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/45c7c00d848c4ac7b836b2181a9f44ef : SUCCESS in 15m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba29c89885cd465c8e7f0bcc8a4c1cb6 : SUCCESS in 56m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b92141f32e494ef18ea6246a55169f49 : SUCCESS in 1h 29m 20s","accounts_in_message":[],"_revision_number":1},{"id":"45ccee0dadb4c7b5ff34c809a0bca42b39909f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-17 02:28:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d98526109f27bee580015d252ee17d5f3d8c2ef1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 02:47:11.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ac52647475e41829025a854eda26614\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/32c3d630355a4a88b4035d195b6844cd : FAILURE in 10m 45s","accounts_in_message":[],"_revision_number":2},{"id":"70ec4f7e106f76fe5dd95f528c45f7328be35550","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 04:39:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3e4867ce8be46b38ef2e08ea55b8076\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/662688714b6e4cfcb78455d9dd3d6a66 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32126502d5f747979d0595d30cf1671f : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e4d6bb7990e46b3bd4f3764c90a2aa3 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b04ee779fd1f48d8a23d080a46f0c021 : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78b0fa8dea874c2296dfaa4d87e7c2dc : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d10b04497e074688afdc897e57744c79 : FAILURE in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bf9fb8a57a8840b0887975be20f9ac64 : SUCCESS in 5m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9f50888551884c68b1ef98ef42cee23b : FAILURE in 5m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c71524d8a30b4d34a4ddceb030bb05a8 : FAILURE in 5m 07s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/c87c768483534186860cf480817b9227 : SUCCESS in 10m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f11defab29474c8aa5a337ef4c0ad828 : FAILURE in 6m 15s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/19b6d09e1adc47089f543e7f11369732 : SUCCESS in 6m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2dc33f395db348d7b95422706c6f19f7 : SUCCESS in 12m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5499fd998e9b43c1b83fae1f7057dc7a : SUCCESS in 7m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a35e9581e470499bb62f110180be6c29 : SUCCESS in 7m 58s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/456d2f8765254305b9b15aeb38924619 : SUCCESS in 10m 30s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cd6a4541951b4e6f87dec86b71b09d69 : SUCCESS in 11m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ecf14876b1404cd0a3aade0233b7e469 : SUCCESS in 8m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b8d5e99a1d34425a9eae541ca618b292 : FAILURE in 8m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2c061f987e0747ddbaa3a1945da36eed : SUCCESS in 11m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3e2e777be490408bb75836823bed21fb : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/825448c0d08c4681be5c65f98f3407a6 : SUCCESS in 9m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/69922167ba044c2297043abaa11840b1 : SUCCESS in 6m 40s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/c2533b2a9772480f88a85ce9ca798fd9 : SUCCESS in 10m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/00635772a1694f6daccf5be61a1cc85e : SUCCESS in 24m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7dfb62cdf15b4c6ab5db57873f6a3f5f : SUCCESS in 15m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2a2881f14add4b03832e601aca6fec85 : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/808ad0824acd47538097f32c523591ef : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34cd5aa9397446ef874da7096f041e6a : SUCCESS in 20m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3ffb4177db0e417ca5e53efa44ac7a51 : SUCCESS in 6m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d9119129837c489a8638da8cf52a8c30 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bce768ca2dc4474f8b48bddef56c4cb5 : FAILURE in 3m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ca542ddee81a4a9c8ba8f308e84e96e7 : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7119e1c6847440049dc00e4599f67d28 : SUCCESS in 43m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a17f2ef60df4d818de600748ee7ae7e : SUCCESS in 1h 44m 20s","accounts_in_message":[],"_revision_number":2},{"id":"624d281d0c19d1064f1cdf904849487f7a701f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-18 02:36:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e1f014e9a451d26d9e955ab9bb24d13aabb68ce8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 02:53:22.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef294408e58a4adfb33543dcf758f8fa\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e3ed287ff05a41ba8e05312f013a9585 : FAILURE in 10m 42s","accounts_in_message":[],"_revision_number":3},{"id":"e1e069a58f32f78d10f9cdf1eebb3e58dd62a73b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 04:32:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d21b46fb1602420382ed455423c8ab52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427cbbbe27244398ae8a733b0b72055a : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd5f5cfd85a446b68a640d6ea76d7299 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9861be9cc00640658750466fb1d69050 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d248f9e8c5448e6a4551bf041a27864 : SUCCESS in 9m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f40e6d02faa3422f9ff2261971070650 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/48bdcaaa5bd646e3980a0f5edbf349d3 : FAILURE in 8m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e598a76ad7f04e2bbeeae66407c03e85 : SUCCESS in 8m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a5d3a4d9c954730a15a22c87dcedeed : FAILURE in 8m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/801af183021c4cb38f09c0114be2900b : FAILURE in 18m 06s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/25ac0d0c3de747ce9cc345def2c71853 : SUCCESS in 12m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8744bb4f6cb947018b162670c0701dfe : FAILURE in 6m 04s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2fa5b081ba9945b4972536b975aab925 : SUCCESS in 7m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/16214d1bb9d442bab8c94b38bd89fda8 : SUCCESS in 11m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/da7b813e4d9043e59cc79dbf28d05aa1 : SUCCESS in 5m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9d72506360dc448fb56dee56c074c9ed : SUCCESS in 6m 24s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c5df0ff5d59a499e8d5ede2c85b189c3 : SUCCESS in 8m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/93ee6b31fc004c5a9b7464414e6d64bd : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/11a6b544ba344a178245e20119850baf : SUCCESS in 9m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/96cdee0317e34cb7a676d388c7bc35b3 : FAILURE in 8m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3cfc812e6ab649859128bc48d75fc0d7 : SUCCESS in 11m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/247d1f5bed234fb8a88a0fcef5638e1e : SUCCESS in 4m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1ae07d7eed2d450398576029f8527ba6 : SUCCESS in 9m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/33a02b6a4d72477faaf9f45c5ba5e711 : SUCCESS in 10m 17s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/85c942f40cda4b3eb38f0fbc5bad9d38 : SUCCESS in 10m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/06a646f4f7744bedbfe302b5549d41de : SUCCESS in 25m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0f37cf91cc81420d95630e1b7dec0f3c : SUCCESS in 12m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bb4e3e79463b468ead540865749b8346 : SUCCESS in 6m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e51d66d052b4808bb33e50207a6ea72 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7fa29efb18b5401281fc679d7f0e47cb : SUCCESS in 23m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c70647607c304ef6b2c51a46989913ae : SUCCESS in 7m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a60490988dd040e7b4937f7c5d218e95 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c92fd765047643d593caf98e8ea50e1d : FAILURE in 4m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c5bd46bb20944671a1f1866ac4258ca2 : SUCCESS in 13m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f1b53368db8470c8155dbd12e7f510b : SUCCESS in 1h 01m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2c62ede5b9b4de3847df28bf014046d : SUCCESS in 1h 43m 23s","accounts_in_message":[],"_revision_number":3},{"id":"cbff81df9e2ac7be1a4fe0c6d35237f18f2d0494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-19 02:34:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9b0ed9982b7c98379722047f592ba6ae1c7efdef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 02:50:06.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f996276f194cc7a76a6713da6810e7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/62df1d84ab9b4ce5834b991ed59dc252 : FAILURE in 6m 50s","accounts_in_message":[],"_revision_number":4},{"id":"b3f444357fc0c3774e59c28e1dcff6bdfcdfa5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 04:12:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cba8880401846ebb396c9b4febc2241\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f1114231434dc6b3c608a1d42ef955 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08754fa9afa84904a7825db0d4591d37 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89bed87997b443b3a6672f79d28555d8 : SUCCESS in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/316504644fd94a70aee4df09b36a6de4 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e4c13df89a44c558f8c2b6ea8e40021 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5622e93b5fe34ef5bc4280732e38fa94 : FAILURE in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e152aaee06f04155b9acd0c575a5468c : SUCCESS in 4m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68271629eb1d4652aaeaa0d1baaa55b0 : FAILURE in 6m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/11d655e313824559978a287992f6bf90 : FAILURE in 6m 27s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f7ae18de81a14c12afc1247627c008b5 : SUCCESS in 10m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/642de210065b4f13b719185cbc436ee8 : FAILURE in 6m 58s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/38c9e4544daa479b839d597a15e986db : SUCCESS in 6m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3c6e21fb76d6415daeb2491c0f1eabb9 : SUCCESS in 10m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1b9fb45defcd40dfa1b8aa2659aae718 : SUCCESS in 5m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/87bceccf90b44a9497fd949cbea7ea5f : SUCCESS in 6m 15s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/183333e07c114114a329557dd3b71b28 : SUCCESS in 11m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f15ad938986b4cf8a0c4b244a7ec3d73 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/022b0300efad4d74b7856db9b54f511c : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c7563915b33411c9a9d286f00228f0f : FAILURE in 8m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2b6c14fd85c843afae90eaff78c82b82 : SUCCESS in 12m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/14ef7c224bf743a2a17105600ed046ca : SUCCESS in 3m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c23f446ed1864385ac799254cb2fa361 : SUCCESS in 9m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4bcec2c6081a495ba780061c8b339d66 : SUCCESS in 5m 44s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/111ecffe80ea4e0dafb3c45932496eba : SUCCESS in 10m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4fe429ebf69549bb9cde892bba3220d0 : SUCCESS in 22m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/216992158b1b410ab81c29dbecd1d594 : SUCCESS in 13m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/093cf39c5cdb4841a8996eb1da9a56a7 : SUCCESS in 7m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/af4322e1628e4cf4b596411363db20c7 : SUCCESS in 8m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f3413620f8304ec49ef04dadfea418fa : SUCCESS in 23m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/aff80f58744b45aaa9b16c4a26463c49 : SUCCESS in 9m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2b6d95e14eec435d870542ef8a361659 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae67640641b2467db7cf4803f93f6832 : FAILURE in 5m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ee00295fbaff440d95c5c7d56b51d7aa : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba2ea2eeabdf4ac3a4413700ac3c6340 : SUCCESS in 1h 02m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68b863478c4047bbb4626352ed9dd08f : SUCCESS in 1h 31m 47s","accounts_in_message":[],"_revision_number":4},{"id":"cc8ee92069297831686d53f22dc6f2c2bcb3d909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-21 08:34:46.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a7746c604a2328de75b9c119a1c0284260f6bcfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 08:45:28.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d0faa297d84df88deeaf30e1775297\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4cf34262240b46ba8f4843cd7777dc3e : FAILURE in 8m 45s","accounts_in_message":[],"_revision_number":5},{"id":"04737b3bf6195340ed27f4c383e4a4549a12a644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 09:50:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94974f8600eb4125a4d0b11f2b3669dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288f2caca1054dbcaec9c10502a6591d : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c2bd8d8b34d4be99ba48b9fec71a109 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5eea33a82434d289560477347acfb57 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/989c2ab277914415b6258fe3eceefc44 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/691a46a27b764edd955549c55a7cac3c : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58bf706168ca4a80832d137ef8ea532e : FAILURE in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e57057d5827645b3af62dec304c677a4 : SUCCESS in 2m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb64938c38f34132a30672eaa33a627f : FAILURE in 5m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/05e3ef5c27c34a429950bb4244ecb00a : FAILURE in 5m 00s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/279035768a254192bee548384a982412 : SUCCESS in 8m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ce8bbf40333463da52ad7b74faff378 : FAILURE in 5m 05s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2e652cccf9b746b0827245170dca3baf : SUCCESS in 6m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/10a0a2394cc144f7982c13d42b368154 : SUCCESS in 12m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/775dfc7a477042d9a3d7ac7d53bfc7f0 : SUCCESS in 4m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/281bd32ab0384eea9a295cbb44db26d6 : SUCCESS in 6m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/100de37f2e6e4d4891d83a0c18439a8f : SUCCESS in 7m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0938ac8d661e419e94b1c436038b0967 : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/848cbbf142324d1b9fa8ae0c52b1abbf : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2760f9f14c774db49243139fc6c58334 : FAILURE in 9m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2103f4d5c0784a2390476832135f9d10 : SUCCESS in 10m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b55dbf310df2446eb82f525ef1d1dd4c : SUCCESS in 4m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/728db823e9a34cee97369beed1e8f9ce : SUCCESS in 12m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d37f8447dc5b42f79d2f9a267c7f03b0 : SUCCESS in 5m 21s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/d46ce81d0038424e870ef6aee3d0d5b8 : SUCCESS in 9m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fb81f977243c4af7851cd4ace3fb4734 : SUCCESS in 27m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f2c5e071443d47b59f83d8d8b60dae53 : SUCCESS in 14m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/35e368db835c4ea7b29993473aceaa8d : SUCCESS in 5m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf1de11dec2b4128949e988afc62183b : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8eafd6213ae649eaba52591af8b0deb3 : SUCCESS in 22m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5d6612553681445ca2c11dcdc7f12df4 : SUCCESS in 6m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1cb8003d5d1c42ddb42e80a887bc0da1 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/874be96336184983a762e136eb851bd9 : FAILURE in 4m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/84d859550ae7460ea43e6f15563df296 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9517f55e9188408eb1b52c3a1d2e0383 : SUCCESS in 43m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5dcef09c560a408f8d5e2857b9699923 : SUCCESS in 1h 11m 51s","accounts_in_message":[],"_revision_number":5},{"id":"209657e19f44df66e692c8955690b6c90ec643fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-22 16:09:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"55f6905b02e619ec4896dc8d55660bb4cd168456","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-22 16:10:00.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":6},{"id":"8b94c63574b9d351e8a8a2db981df2f567259eb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 16:28:18.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f489f81b12b4e188951e122f181b26c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a3c26b85b4024940ba98e3782ba35522 : FAILURE in 13m 17s","accounts_in_message":[],"_revision_number":6},{"id":"13c816aec888c4547d8fda98221facfc31ab05ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 17:33:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9af85284553c46419edabf1393f43fec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b792990d201e4e8dbf417eca3a909834 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6e869c13c4545458208ba0e8e1418e8 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38b642cda90140f7b69763d1048c7b5d : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d4eaf0b2a7d4134a44796035c83f4c4 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3083425dae5a48efa0648ac932bfe470 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ecda169484bd4a30b05c01195bf979ad : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/13a23b131a2f42a9bdee2321e0027e9f : SUCCESS in 5m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/06669d0147f149c59090b306a456f4aa : SUCCESS in 7m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/340679896b244ffdbc490c2becc0d032 : SUCCESS in 11m 28s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d0fd33cb24ca48d7bbcf9b564a019e5a : SUCCESS in 7m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/524471d4bbfd4634ac491664bb647c35 : SUCCESS in 5m 08s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/b445260e53c54e758967693a1c483f2b : SUCCESS in 5m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ade55b2d75fb46b4b253b19cb0937a93 : SUCCESS in 11m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/746ed095afb44b32af0fede0c94e99d1 : SUCCESS in 4m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ef6a3dba0c174c38b3a59b35bcec3c54 : SUCCESS in 6m 31s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/759d3e5f188f4e91916feff401d6806a : SUCCESS in 8m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1bc48f71c1ef41d98b03ccd9f3d9e66e : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c95edaa07a804c10964e7d261a90a854 : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/51ae7306447648a39d49b69ac27a9b3d : SUCCESS in 7m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/020f1e6012564a818b612865de464e0b : SUCCESS in 13m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/962d38d5129d43f9b7890d24ea28b846 : SUCCESS in 4m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a3f345b2fa384ada85d43b1582275760 : SUCCESS in 11m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7a024e99b2834c62bead597713560c1a : SUCCESS in 5m 06s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/ca5b86b44a32481986ce7856bba05943 : SUCCESS in 9m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e1d7d4a0a9db49d49b1219fa1c3f7e93 : SUCCESS in 23m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7748519a8da54116ab8126779d42138b : SUCCESS in 12m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/344f8cf0f8d84437bc6ab98f5b0e7b05 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54d551d892c24a4b9d7deb88770de3a3 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e964971c0734b58bc0b218222614cc3 : SUCCESS in 22m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/aff1b8d46e4d43e5b8af7229d0eb0851 : SUCCESS in 7m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e1f71f39108946cdb33e58073af4111d : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ac2076eb0b6d4c3d9dd46b205b13e87d : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/19d274d4f393421eb2b021bdfba79fe5 : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/754f94be437a4921ba85ee36d2f85ea6 : SUCCESS in 50m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b795bba5fe041f0b96d2ba62aaf4ae4 : SUCCESS in 1h 20m 52s","accounts_in_message":[],"_revision_number":6},{"id":"f62107356994a0603806a0ca355f48aa6489b5d7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-22 18:20:31.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2ad465c6471e1808dc16e42e0d027536780cc03e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 18:21:17.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"92c4ff7b355fe7bed43d391512fc466227c348fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 20:03:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/423e16ccb1ad42f888d61ba48cd2524e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e4e11887e734dfe87818e6c5967b42c : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09529ddda0de409bbd4749eb7b2dce59 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01a2ef84174f49498406fe5e7f002c2b : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4617c6092944623a327912b8cd91d77 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aadf3fd805ff4b458a873bd716077b58 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d56bec95ae347ae950829f588c584f4 : SUCCESS in 3m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5e83c5d76f39475e982a1e9f8be49999 : SUCCESS in 10m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5b01b8d7b7cc49ec83d2e3906c50a503 : SUCCESS in 16m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3cd42795fe5c4d6386f172b038996b7d : SUCCESS in 14m 30s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2f3e58795f6b4c22a222cbee93c22476 : SUCCESS in 5m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5afffa2a2a144aa9a22d04e47b9e896b : SUCCESS in 6m 53s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/93f9e29f8784489c9353adc80577a77a : SUCCESS in 9m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e3422a49732c4437a9b19f336c1eb926 : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b916257645749c98ab786397b60a5a3 : SUCCESS in 8m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/16f48036e3904f31a57ae1e044f8fe11 : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bc1bc5e63f2e44358d9931d5a198873d : SUCCESS in 18m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8a7feaf9767143dfbd74805f48444521 : SUCCESS in 5m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5571acd8e5c34a6ba38387b40844c9ed : SUCCESS in 16m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c2b822bae4b24d75b584068eda94789a : SUCCESS in 6m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ac22c03056cf449ca6586a9c9a7c39b9 : SUCCESS in 26m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dd7c6e4b7ce14720b79b209989b5ea37 : SUCCESS in 16m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/52412213ed1e4f5599ade83dbdf083ca : SUCCESS in 8m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ec7aa781543f476c95663d3684e00024 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6cad8cb4725e4bd5bd01f7592aeb4648 : SUCCESS in 18m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ba8c6b24ca104035834334ae950b2388 : SUCCESS in 6m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/808187cf9418460590303a502cc182a3 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d90f0f418584c67af66c75eb6cf5bf1 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/455e5785e12f428182b79575436ea491 : SUCCESS in 13m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1ae6c77a617422e816b954f3d08492a : SUCCESS in 59m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/796304b728974e5eb36d25385e779d4a : SUCCESS in 1h 41m 42s","accounts_in_message":[],"_revision_number":6},{"id":"9a8d0abf099afaefb8afc211d0344dc625e4dead","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 20:03:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"a5ac7135160f8ba63a56acf595675782f3ee95e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 20:04:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0adb7f02fd1147c89ec130d524b532df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd19cd1ecc5143db9985e4e0264d47db : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f9677e5fca12fa9ceaf239fa841d578d03db5e5b","revisions":{"ca9d4b6395bb0b822c53742faaea1a995f2ebc72":{"kind":"REWORK","_number":1,"created":"2022-08-16 02:23:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/853214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/853214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/853214/1"}}},"commit":{"parents":[{"commit":"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-16 02:23:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-16 02:23:02.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id962baa2032c22f3d09e3985b9f3253d1994d48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca9d4b6395bb0b822c53742faaea1a995f2ebc72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ca9d4b6395bb0b822c53742faaea1a995f2ebc72"}]},"branch":"refs/heads/master"},"b31a9fcfbf19a754d198bc7f85c2c811cc62f6a5":{"kind":"REWORK","_number":2,"created":"2022-08-17 02:28:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/853214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/853214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/853214/2"}}},"commit":{"parents":[{"commit":"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-17 02:28:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-17 02:28:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id962baa2032c22f3d09e3985b9f3253d1994d48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b31a9fcfbf19a754d198bc7f85c2c811cc62f6a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b31a9fcfbf19a754d198bc7f85c2c811cc62f6a5"}]},"branch":"refs/heads/master"},"149692d97bf5646a8910233d31d4c42ace18ca12":{"kind":"REWORK","_number":3,"created":"2022-08-18 02:36:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/853214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/853214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/853214/3"}}},"commit":{"parents":[{"commit":"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 02:36:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 02:36:19.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id962baa2032c22f3d09e3985b9f3253d1994d48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/149692d97bf5646a8910233d31d4c42ace18ca12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/149692d97bf5646a8910233d31d4c42ace18ca12"}]},"branch":"refs/heads/master"},"70bd929f6e3741fa68fe7c9e75da606e9836ae90":{"kind":"REWORK","_number":4,"created":"2022-08-19 02:34:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/853214/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/853214/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/853214/4"}}},"commit":{"parents":[{"commit":"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-19 02:34:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-19 02:34:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id962baa2032c22f3d09e3985b9f3253d1994d48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70bd929f6e3741fa68fe7c9e75da606e9836ae90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70bd929f6e3741fa68fe7c9e75da606e9836ae90"}]},"branch":"refs/heads/master"},"f47dedcff01d1f3a2b230f078b77554274789894":{"kind":"REWORK","_number":5,"created":"2022-08-21 08:34:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/853214/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/853214/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/853214/5"}}},"commit":{"parents":[{"commit":"d4f0334e7ee88ae8712a9ee283916c0a2c9ba0e2","subject":"Merge \"update constraint for ovsdbapp to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4f0334e7ee88ae8712a9ee283916c0a2c9ba0e2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-21 08:34:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-21 08:34:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id962baa2032c22f3d09e3985b9f3253d1994d48d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f47dedcff01d1f3a2b230f078b77554274789894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f47dedcff01d1f3a2b230f078b77554274789894"}]},"branch":"refs/heads/master"},"f9677e5fca12fa9ceaf239fa841d578d03db5e5b":{"kind":"REWORK","_number":6,"created":"2022-08-22 16:09:53.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/14/853214/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/853214/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853214/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/853214/6"}}},"commit":{"parents":[{"commit":"d4f0334e7ee88ae8712a9ee283916c0a2c9ba0e2","subject":"Merge \"update constraint for ovsdbapp to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4f0334e7ee88ae8712a9ee283916c0a2c9ba0e2"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-21 08:34:46.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-08-22 16:09:34.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Id962baa2032c22f3d09e3985b9f3253d1994d48d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9677e5fca12fa9ceaf239fa841d578d03db5e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9677e5fca12fa9ceaf239fa841d578d03db5e5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
