)]}'
{"id":"openstack%2Frequirements~853612","triplet_id":"openstack%2Frequirements~master~I8e6433bc7eb3ff0c0261c65ca3e672769fc88ac5","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I8e6433bc7eb3ff0c0261c65ca3e672769fc88ac5","subject":"update constraint for python-openstackclient to new release 6.0.0","status":"MERGED","created":"2022-08-18 09:41:07.000000000","updated":"2022-08-20 02:59:25.000000000","submitted":"2022-08-20 02:58:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853612-new-release","meta_rev_id":"23f816d059d9bb966b662f7f26dda0f0b04c8bc0","_number":853612,"virtual_id_number":853612,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-20 02:58:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-20 01:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-08-19 16:15:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2022-08-19 17:41:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-20 01:18:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 09:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-18 11:15:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-18 16:08:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-08-18 17:40:17.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2022-08-19 16:15:46.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-08-19 17:41:10.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2022-08-20 01:18:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1c40fc5d93e4b954a5ea7309ce99a9fe2d19a642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-18 09:41:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78afe5c5725a44f3819d984d18505bc1ca994236","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 09:55:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4869bf532ca446586b6588ec924a312\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/985b86a6ab154dd29fbffe33f28e1762 : FAILURE in 9m 09s","accounts_in_message":[],"_revision_number":1},{"id":"96f03194cefde2f911955dfbf79a089ca7d47d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:15:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecd5bb2c7cad4ca190c2dce8ca112e97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b720aba40cc4bd2ab9f699069f9c3d9 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0cec85636014daba48209633afbcdd1 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6be37f839334f089f575624543e7ca4 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/199e946430884c908d4c9f71d47788e3 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad91b91ce92847ce887ebb0fd75f6818 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c24994d27aa94c1d9bf85aacc83d7c9e : SUCCESS in 6m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0f40e7f69c0467da9197c7f933daf17 : SUCCESS in 4m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7f85732cf85247bf99e12f9a1a8c9b8d : SUCCESS in 8m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ffc6b184925140bfb24b123ee50aec37 : FAILURE in 7m 20s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/30c3e1ec8d8b4f2f9f85ec1fd35e92d0 : SUCCESS in 9m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9e5ceee878dd4ba4a230ea130e295971 : SUCCESS in 5m 17s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/51762dbd0f2145cbbcbc6688215b66be : SUCCESS in 7m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6ea4fad4c6574285a0f13f82ec3ad51c : FAILURE in 24m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0097dbae53aa41e4896c445075ebdbb5 : SUCCESS in 4m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a950a6fb4db145d191c97bc7046d34e3 : SUCCESS in 7m 10s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/0ea1887462d143328303d0c50769f518 : SUCCESS in 8m 53s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1e08ddde2a72446ca6a7972f15d5fbc6 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/730f71843b4f40faac3f1de6ac986edc : SUCCESS in 9m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a5a0bc3895d94fc8bd6e75ef5267538f : SUCCESS in 8m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e2df56a986ef4f64ab399a94df42a2a3 : SUCCESS in 14m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ccf4ac5f51d8462899e1431655d110d9 : SUCCESS in 3m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/41003c10d3a64a518ff0704a7ce33531 : SUCCESS in 12m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c5c6f4187b8b4068977756d84303c51d : SUCCESS in 5m 57s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/73c2e5d3660c429495f1235ce6724ced : SUCCESS in 8m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7196eb6cd34a4fac93db9e5c2ef84dce : SUCCESS in 23m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b5f1f0a3d33b456490d2e09bbc8a2a93 : SUCCESS in 11m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f132400f976c405ab2240411c72c86c2 : SUCCESS in 9m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/99ca9c37f9fc4b1b84e19006e81a946f : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2dc8c6469a5490ebf0273c7aecec3b0 : SUCCESS in 21m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3ad5405fb9b44d579fdb540373a82bc9 : SUCCESS in 6m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/346b479379ba471c947ddb631bfd14eb : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f23ed82bf1f46eead2129f5f60ce2f2 : FAILURE in 6m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/350cd961368b4f108daa48a32238ce17 : SUCCESS in 10m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78ccc850b9244b5e94e67b74ccf6b077 : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa2772320f2b4846850024055c03424a : SUCCESS in 1h 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"880cc65cc7769188228e23941e1aaa100a569bd6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-18 16:08:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8136f9d997014846f57c06b71fec1468e9646f65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 16:18:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d706fda069a4b07b18d01486e792338\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/676b8314cc724c6f9794b55acf61cc72 : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"bfdfcb8299a39e315fb3c6599fbec8a47acb2230","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 17:29:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fd4e98db7634c2eaa16f4b72cdfe400\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ffe2cf0716491e82ee6049613215d1 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68abca09ab7e41dca2abebe3d8ec7d35 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1314389f150d458c9d66e929f0e4f441 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e04f94910bb54f5d8834f939e74f912b : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a15c1afca18413ea14d1125e3ad78b0 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bbbc96f1ee9f4b05a11512e4301af68a : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d68ea34740a4fbaabe7d9a04f7d303c : SUCCESS in 4m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/421bcdc21ed34e9d9f6d3b1194b29b09 : SUCCESS in 7m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dc9902e0c458463f91c64f2ebcb6ec96 : FAILURE in 4m 43s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/7276860d5f4c4a67b7d44ee8c9f5a22b : SUCCESS in 8m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/79881e3c8f4a43339eb3c01e29a9d002 : SUCCESS in 5m 24s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/9220a66bf6854ab6bab6afb9b4fab834 : SUCCESS in 5m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d8004e85f8074bfabbac3eff6fd38b2c : SUCCESS in 11m 14s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6f2671a418f1481e89d2157fee50fdea : SUCCESS in 5m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/26768f23200a4e539bf5732885786c87 : SUCCESS in 6m 26s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5cd1e96fbae0492d9573e72ff41bab81 : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3b16e6271b63441eb29c58b8fadbfa50 : SUCCESS in 9m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b86f02734e244840a5b2315b16f22036 : SUCCESS in 8m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/91170f7ced8a474d993fd10fc2b2bdb1 : SUCCESS in 7m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/731c0cb2eb1742ec903a78efca17fa18 : SUCCESS in 18m 51s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/22169017f91045108741f52f4caf22f1 : SUCCESS in 4m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/213e070245214a09afd3fb6767bdf343 : SUCCESS in 15m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/83175c759e994bad9467dfbabf7a9115 : SUCCESS in 5m 10s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/e20d355cedef47579d3c99713969129c : SUCCESS in 8m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9b7dd8c6fbe44e6e88f69f82faac4ce7 : SUCCESS in 22m 10s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/588111b1667a40faabac1686b3928c35 : SUCCESS in 12m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3dc851b547da47b2b90dd6154d5a81b0 : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b86f076969444cbea165fa57f8c0858c : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/80f965092e754347ab79e465f3d7282a : SUCCESS in 23m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c6ba8f0d630b4e5b90c731d3a442a142 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3c6cdfe26f8a43a5ae857f34d6e76e07 : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/096198620da3404ca0d0ab07d016216e : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0d433f83533a410095254053494440a5 : SUCCESS in 9m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1fab925a738437894c79e5ba7f01206 : SUCCESS in 53m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1795c807cca48e3812e108536524b67 : SUCCESS in 1h 19m 20s","accounts_in_message":[],"_revision_number":1},{"id":"58244a049e7cf4ec3713adfe5cfcaeadc6f9396a","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-08-18 17:40:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26edba781bad22d441741d113b49b382e4cb0444","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-18 19:56:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f71b1780ae92874a2da4ca85a4cbb814852df029","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-19 07:23:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e49f0f44a393fa6f0988637d9925063af1c8cbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 07:38:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c59d3b7be1134cc3b2559368df55f630\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/99630b016f6146d5b47530c83e97dfe2 : FAILURE in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"188da5bee49086540318cbfa8f96ab858b39758f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 09:11:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9bc89d4a70447c482b6884fda890401\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c087b4043df54bacb208a90e5fa4d690 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f960f66ac3640cda29d5410600dfac1 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8eba04dfcd5486cb209513e93e0b498 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf20aad5f384d54b4c607c70137dfeb : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8f827450b11345428748bdac5dbd180c : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/710650dff771432382d7f3231f91a19f : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1cdd91800e74704a8e7449494233780 : SUCCESS in 4m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/186fa2c365274db7b2c5c0755f87f8ac : SUCCESS in 10m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/73169ab4a8b643be86601ec1c232291a : SUCCESS in 12m 37s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/b2895bce24954c75a677b73183a6ecb1 : SUCCESS in 7m 17s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1cae8120b1074968bd4568e6844639bc : SUCCESS in 5m 39s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/9ed55f77471c491caeee2a584e00b9bf : SUCCESS in 6m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2f29d2ca860b4a83aa5d4a20b448641b : SUCCESS in 11m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/23ec2fe4aa7f4b68bd98273b71245e6b : SUCCESS in 4m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/25b95224e31c4375a9e15a2547ce37ce : SUCCESS in 5m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/da87e76c12a046768b15df0baa097d81 : SUCCESS in 9m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b38259182a6642dea777f2d194a155b2 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e9d34a1cfa249ac9cbad19e42cb46c7 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/075b8766c68d46fd805ced47ba6a9d27 : SUCCESS in 8m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5b0544a7597a405a83e818f6de4bc4b9 : SUCCESS in 14m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7e5071e5002e40ea98f41a8592ac7e9a : SUCCESS in 4m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6fe8d54030234757ab2ba755f58c29d1 : SUCCESS in 10m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/19edcf9e9ae1496094a3e544dd0b4108 : SUCCESS in 4m 47s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/07cd0e9fb774416bb53de426b229d165 : SUCCESS in 9m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2edc2a21efe244e49b0a81fc86078f63 : SUCCESS in 28m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c0d913aa18e847949c470ff7bf040767 : SUCCESS in 13m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3267596158c84619873d0d7ae6c0fc33 : SUCCESS in 5m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/15933db419414b1a8d9a32d4b331ebbb : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/964727da1ef340e6bdea912b2716038e : SUCCESS in 22m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3f25b5d2266f4b039932bd86e97e7512 : SUCCESS in 5m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f38700ebb1044d82809525d1622f0762 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e228b98341064a5bade6d2bed6a47146 : SUCCESS in 8m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/182695de481146eda9b2fa148144a4d0 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b92991497c2e4b249c0f12e41920721b : SUCCESS in 58m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0afee3ec6d6a46b2947e7cba1734b3ac : SUCCESS in 1h 41m 58s","accounts_in_message":[],"_revision_number":1},{"id":"dee9c796c78300878bc6f402c9e88fc36a3805d4","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-19 16:15:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"385a0f15019530bd8a3a2035102f16a6d5bf57d8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-08-19 17:41:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9b9e1e5f3cc1fcbdf91d9676e2028145dba10805","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-20 01:18:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9da86ddc96da46695f99c72b1c57231c196442b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 01:18:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a783b5b4358ec827e4a9b6fe5eb710cb8a511f57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 02:58:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eca4ed1d5fb34f9291e876f186e47116\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/490c8e8cebae40b2b1dabcf3f0c2fa67 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c9098eec5e4a9fa00ad955d5c8b001 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46b2a79431c74a3eb06a613bd7e93632 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f5c7a6776d7494eba11ee995fefdfc9 : SUCCESS in 4m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0d635058699e4bab90e91c5a6f0e6dcd : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c95ea8c9c1c4c0a9c7a010fe6ac4185 : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0cd23805cbb4485eb6c98a65828f9f2e : SUCCESS in 10m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3934bcb459d24229a5279b7131b3e44a : SUCCESS in 14m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a5c2487538504d6f94b7a1f2266a72df : SUCCESS in 11m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e46e34b4210a4f08904b5f419d98572d : SUCCESS in 5m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7bf8e11ad350420aa7430f21c726c84a : SUCCESS in 5m 51s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/832bb37c5a3e4288ad1b7bfc5b2f44e1 : SUCCESS in 8m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a4683290e7aa4fcaa9ae3154a86d47c5 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63a987a67bce45359f12e92a07d2b59c : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/614b05787932454bacf19a6bd290f93a : SUCCESS in 9m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7deacf2573fc41cdb29f91c032177a62 : SUCCESS in 13m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2dbade18fbda4c138333d043642c8920 : SUCCESS in 4m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/094a1c43b63842d98c5d8a08e77fc9a1 : SUCCESS in 11m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/91b12432ba0f452aa326e065533815b8 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f972a69973544c37a5318cd2aebca74e : SUCCESS in 23m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d413ebd127ed4dada861c1589a873843 : SUCCESS in 12m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d0134cd65eab402090bbcce0459aaf75 : SUCCESS in 10m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2bbd9b1e61fc41aa8f1f98763d614f86 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f722d767942c49d1b1626dad2c508ecf : SUCCESS in 19m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c502a7649eca44ecb39eadbe545348f6 : SUCCESS in 6m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9029bfdfd6aa4daa93ebbd7dc69deea3 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/36f5d3ac7fa949d386eb89babd1cbf6a : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/85fdcfbbe20145a6a1629ecb9949af7d : SUCCESS in 14m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b8417017d2f42edb349eb515fb5f352 : SUCCESS in 49m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2ec0b3851db74f44baf7c19ad7a56597 : SUCCESS in 1h 32m 55s","accounts_in_message":[],"_revision_number":1},{"id":"67b5aab5630a311744b102fe3df51cfcc8e99c06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 02:58:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23f816d059d9bb966b662f7f26dda0f0b04c8bc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 02:59:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adf4638ce187438695ed868acc2d823d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/753bf9ab21044237a974d3749658ce9e : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4dbd15d3b80027fdeb1c36286633570d82376de4","revisions":{"4dbd15d3b80027fdeb1c36286633570d82376de4":{"kind":"REWORK","_number":1,"created":"2022-08-18 09:41:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/12/853612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/853612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/853612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/853612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/853612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/853612/1"}}},"commit":{"parents":[{"commit":"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 09:41:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 09:41:06.000000000","tz":0},"subject":"update constraint for python-openstackclient to new release 6.0.0","message":"update constraint for python-openstackclient to new release 6.0.0\n\nmeta: version: 6.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Id4f8b917056a00e4dfe99184daaac0d8c3a605e7\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8e6433bc7eb3ff0c0261c65ca3e672769fc88ac5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4dbd15d3b80027fdeb1c36286633570d82376de4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4dbd15d3b80027fdeb1c36286633570d82376de4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
