)]}'
{"id":"openstack%2Frequirements~853614","triplet_id":"openstack%2Frequirements~master~I058787be47d40395fadc87b1ec1ed8845738d7db","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I058787be47d40395fadc87b1ec1ed8845738d7db","subject":"update constraint for sushy to new release 4.2.2","status":"MERGED","created":"2022-08-18 09:45:50.000000000","updated":"2022-08-20 03:05:57.000000000","submitted":"2022-08-20 03:04:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853614-new-release","meta_rev_id":"39694dd1ef5fbb86120f2a4c6dbbf599351b3de5","_number":853614,"virtual_id_number":853614,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-20 03:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-20 01:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-08-19 16:16:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-20 01:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 09:59:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-18 11:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-18 16:07:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-08-19 16:16:15.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-08-20 01:18:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4629f231877286f16af685bc4bc47d06607cdf8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-18 09:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe3d9fffba1b81a6745055407f18dbee2732054e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 09:59:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26821705facf40839f18a38465797534\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b9918e3bb1704d90b53251bd665daf70 : FAILURE in 7m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1d07b11a56ad5a94f82fcefd8e295829197ead9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:25:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4735352c3e694e6d8f6d933c6fbff92f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ea3f50bb7b4c7a979340b63bf87086 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2418058d6c4d84ac3b7b2a0bc26690 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f65bb0df2ad24c71b9a6e9f0e84bd412 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e24533026b74ddf96fb80d67425c3d2 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a8a9c4fa5ae41adafc9061a48877515 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ad0db9710934e6c838c77402d26468a : SUCCESS in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4d4b2fab3836437c99017e11ab058d8c : SUCCESS in 4m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4adaaa3d39304c0dbafeefadfabc05bb : SUCCESS in 8m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/719500f262ee4558bbca3b801e63664a : FAILURE in 4m 20s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/7131efab070f4d729fd27531fc25e870 : SUCCESS in 7m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d641253de3844cb8a40e2929f00de5b7 : SUCCESS in 5m 56s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/f78684a9dadf40dd86fc025b9b80f6c6 : SUCCESS in 6m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ba8c3560e36242b9b0d70f65cc7ef750 : SUCCESS in 10m 43s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/df210d8198904880abb608e89c367e5c : SUCCESS in 4m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/52d1f965b1a44547896e0d6adc23849a : SUCCESS in 5m 55s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/aeacde0eac494553b6acdab1c033df05 : SUCCESS in 7m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e1be618976e047cd99eb4225980f2937 : SUCCESS in 10m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/afed978a7c6745cdaab7c188b7438c53 : SUCCESS in 8m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1cf64eed4d4144438487d1bec70591be : SUCCESS in 7m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f9a1c2ec0a9e4a7a9f24d8fdc7fe594e : SUCCESS in 11m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2805217e4cc24edca88d04620d2332ce : SUCCESS in 3m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2ec3951230354865abfc8253d02aa803 : SUCCESS in 8m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2e894c1a7cc541a899652303d5cedb82 : SUCCESS in 6m 22s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/5d096688768e4c949762c9e6fc6ca3df : SUCCESS in 7m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1083949fd0dd46ebb23ece2b8cfe986a : SUCCESS in 26m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4971f8a791c04ef08d521bfa0f5ee477 : SUCCESS in 11m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bcab34ce757b45c18e717d4ce8bd1196 : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8b449504a683403cba637a384fc7788d : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/366035ae8f694d6f95054b151c691ad7 : SUCCESS in 21m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/035ec121ba7940e1b77444f891fed7c4 : SUCCESS in 5m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/514ae9ed48134c83b2f22db94013f198 : SUCCESS in 4m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f1c8fd49ac8f40b2a63cc275122f611e : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f8421ea4345c4382952972bd8b8e4b9e : SUCCESS in 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d6ad323ba3b49f88d81376327180965 : POST_FAILURE in 54m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fe43e20b0fd491f9db67dd7aa70aa81 : SUCCESS in 1h 34m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3802d17f01001f17f8a3ed89dc65bef80fd1ee73","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-18 16:07:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7cab4ade17e656c9e47d38e4229f3ae8fad43d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 16:21:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9436cb822793403ba7ff8475f12409a1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3e42437f27074d80828d9b044d80285e : FAILURE in 11m 38s","accounts_in_message":[],"_revision_number":1},{"id":"799de31cdbf0c55a9969bf9bad8ba8f939c50f58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 17:53:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/42c7168d98344269878706fea8f542c4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe771d6a18e496fb464a4d74f70ea2b : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28ab44015ef74bdeab11a028d417deac : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01c4d5a0877e4e29ba3d7d1a995d60dc : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab1c191ccfd5407b8c518e506bda86c9 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/192878bad5464ecd80a030a72a7e315a : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f63b506fc8244c23b53782484470aa91 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bacea68e8a640388ffbcc10285e4441 : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f859ff5992364b6fbd1c2554c2cccd8d : SUCCESS in 9m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7077d45d468f4267af0ac9799d2db091 : FAILURE in 4m 22s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/2e5cd407a18748729fc91ad91056ef9b : SUCCESS in 8m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4326553ff82347089db553644f7d0a11 : SUCCESS in 5m 48s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/033cb1f8669f4caeb67444f038111015 : SUCCESS in 6m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3060e0039f174cf3a08045d4118c6f2e : SUCCESS in 10m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8a4f218b07d24b9b89f07dff68eb9818 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/627097abea294fe3b04d1b8d28b17953 : SUCCESS in 4m 58s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/8c5638eb21be47fdb1903a0e4865e02a : SUCCESS in 8m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fbc19d10cc0e4921b14273eb8fa95cb3 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff78f76721bc439cb37db79f596ef464 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/70e73bcf63db4028a75d4aef3c04f60c : SUCCESS in 7m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9646a88317e94fd2a08f43f0d6893474 : SUCCESS in 11m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/77a6017b15a648b3a30e9125fe1c105f : SUCCESS in 4m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2b36eedb2db74257a32eea3293dcdc8b : SUCCESS in 10m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bf3226db4c9b4ea68b34a1f2a92509bf : SUCCESS in 4m 12s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/ab6b6b772a354c80bbd0f1b082be0aae : SUCCESS in 10m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a3431e368ee24088990de2c7a22b4d75 : SUCCESS in 24m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8bf24abc8e574c1b8ca94acb225fe71c : SUCCESS in 13m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5838546edbce4b6bb7829f38d60187ed : SUCCESS in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/275668bb3b1f4ec8bbcf11e5be058f10 : POST_FAILURE in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e1ccd31993a549bcb0d904e6e1cc09e0 : SUCCESS in 19m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4ddf28619b374462ab75cae8f686deba : SUCCESS in 6m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4044f0fd68e24de8972535dcf6a7a318 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2d441260234949c5a4c6cb0f7102f95f : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/673a19c84d5c427a871f0e5f86f46b20 : SUCCESS in 10m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84d910f679784b518aa94edb1a252ae5 : SUCCESS in 48m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0468cb7d2d83422aa45dfde5304279ff : SUCCESS in 1h 38m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d7e373c41cb76d4377b37beb95d3e51882a0df29","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-19 07:25:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8e297abd1d70effdf6aaffb4b40d8e84851d3ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 07:36:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f341b94830724687a3cb0fdc9e5b68ed\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/64c5fb47bf3642d2819f4fe77abea234 : FAILURE in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"d12e7c90d97384067ea0ba4b96c5a3c1f703c359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 08:50:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f3af84ab9844d1b8fe89fb2d7212b1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737eb1bc00da42db8d17087be543a4d0 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23859db4dc704765968aa3e1297708f2 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/397744b7c67f4e3b9c458da694fcdd9d : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/116027cf7a9d4b989523c7edba4df2dd : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64deb660f7e0484f86e429e2299d5a19 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2785c3faf7b7403faeb993e0734d5fad : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/abfd9dd354f8482289b7ee689658cf87 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb2a68f0d83249559b53ae528d701617 : SUCCESS in 10m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2793d4407bc24b55915460c4cc590e48 : SUCCESS in 12m 48s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/95a2b5527b8a40f48bbdedee306199c5 : SUCCESS in 7m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5b07ad4f2b0e4ae294c1749357dccc87 : SUCCESS in 4m 49s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/f678be61a0d04f04b7e72d34fa09db99 : SUCCESS in 6m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b00397114d194e67bc2bc7ef973ab88a : SUCCESS in 11m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2111f478947848edabaaee5f08c0bef2 : SUCCESS in 3m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/55417a30c9af4f21b6ee2f5253f47cec : SUCCESS in 5m 30s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/9402f23c00ec4966b328b3fef2f14da7 : SUCCESS in 8m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a0baa56c7d3b43fea59dfaa28726293e : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9f41019a6dec4e6790e2d6800142c21b : SUCCESS in 10m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/64c444168da74e8c94eb62eb94fa82ee : SUCCESS in 8m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f73c7274d0354d85adcf60c95273c9f0 : SUCCESS in 11m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/68e68acfe7f944fb89bdb7e949dec5cd : SUCCESS in 4m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e791d8be33534ccb802baf1051fe2603 : SUCCESS in 10m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/89557b852ee44514a28ab9f1a61857e4 : SUCCESS in 5m 12s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/b737bbc9050245ec99ba08b2a28d7ee8 : SUCCESS in 11m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eb227232d28a4dcf842fda70165d95bd : SUCCESS in 22m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4d23cd34a3b549f18301a658f3895725 : SUCCESS in 11m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0ccd46283b354841898ed05c393e19ef : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ada5884e8d434d6384adc207bafd4b55 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ba0a1f27fe2462e9119d9d6f29e65d9 : SUCCESS in 18m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6beef30fb2a347ebac7fc533a1372afb : SUCCESS in 6m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d5e134cc440b4a49a4bc795ea2e9b3ba : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee88db6fce6d4499aaddedbe29e85feb : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dcdc7ae955a541558efc869ae33882b1 : SUCCESS in 9m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4b1b7d4de2e4c28a546a5551e29ed74 : SUCCESS in 42m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/247955d0b10d4721a46e20ca6fb3b448 : SUCCESS in 1h 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a65dfd02a33358e0997271f76239b700f6944547","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-19 16:16:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c39f21f29ddf52399267424c22055d5fd2e8014","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-20 01:18:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ed64b89a1306548c52ca6a8b8b3ec547c4225639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 01:18:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad201bfc32e321853e41cbd54ff571fa8081d73a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 03:04:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cabedb45e7e24f6c96211f4370c5cb9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ba9affff7944c5e88158b4cf3fc3094 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9df98623e1a54dc5ba46d2ebe6e92947 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e485b6efb3c24b6a9e724cb7239fea0d : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83509b8d7fe84b51a04c488d661b597a : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01d7f2a625764e27bdad7ec73bcf0b87 : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8ca140c580304c1e9b3e0b85c7684f75 : SUCCESS in 5m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d568b56f0ce847c9aea194da84a930eb : SUCCESS in 9m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d3dad86ba6404fb8b4250a4b99d573b0 : SUCCESS in 12m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0c641a56520b4d5e8e5471286fef6226 : SUCCESS in 9m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/437fee6774b64baf808245203a947892 : SUCCESS in 4m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e4db6aca079342a7bf0f389d0812a77f : SUCCESS in 6m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/98983b1eef3745688c3b506de4f817e8 : SUCCESS in 8m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f63c53404eb24ad48d1d6b57d4399527 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ebcbcee3d98b4cbd87f2581d0e13d235 : SUCCESS in 9m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dc8b332b42b64ff7be034e0a146fa420 : SUCCESS in 8m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ceed857313a6458cb4259b49a8e36bf9 : SUCCESS in 14m 07s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/48c5d83547b4496f9c14770e3ec12a50 : SUCCESS in 5m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7bb36ed871874700851eafb782c16c71 : SUCCESS in 16m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0a6ca0009b5247c9a86b26b17bef4666 : SUCCESS in 7m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d5aab66d590c4ed8b732947418d61537 : SUCCESS in 24m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/65f79de10cd546b8a153bab9236caf0f : SUCCESS in 14m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/00bbcebbcd9c493e89af2ba62a90c7a8 : SUCCESS in 6m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3627b8f6ceed424081f4942c6beb32e1 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f2c4445ac354cba9592e26170e812f9 : SUCCESS in 21m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7780db2bf0004f06bb83a24374790afa : SUCCESS in 8m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6b3354f9465540189b2463d90761968e : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/aaad36272ba049ee9431aeab0e5cf625 : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/817ab8fd7f544409a437df03c45b8fe9 : SUCCESS in 11m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30bf3958cf29433da1aef0d64b68e3cb : SUCCESS in 52m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/630be176e9824dcdb81a786b0f27395b : SUCCESS in 1h 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7db2511e7517b1d912fcdd18a736e57078526abf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 03:04:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"39694dd1ef5fbb86120f2a4c6dbbf599351b3de5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 03:05:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a01c8761b72440d97977649861aa91e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e680df3ecdbe4687b13776c6de3d36c1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80cd294d1965d9063e62bbcb92deafa3495e51b2","revisions":{"80cd294d1965d9063e62bbcb92deafa3495e51b2":{"kind":"REWORK","_number":1,"created":"2022-08-18 09:45:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/14/853614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/14/853614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/14/853614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/14/853614/1"}}},"commit":{"parents":[{"commit":"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 09:45:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 09:45:49.000000000","tz":0},"subject":"update constraint for sushy to new release 4.2.2","message":"update constraint for sushy to new release 4.2.2\n\nmeta: version: 4.2.2\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Change-Id: I03696c6afe5375f809c2b5e1332c027d09751ca8\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I058787be47d40395fadc87b1ec1ed8845738d7db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80cd294d1965d9063e62bbcb92deafa3495e51b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/80cd294d1965d9063e62bbcb92deafa3495e51b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
