)]}'
{"id":"openstack%2Frequirements~853615","triplet_id":"openstack%2Frequirements~stable%2Fxena~Ib448bb3763198bb1012e57ffcdb4b8c12f787c7b","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"Ib448bb3763198bb1012e57ffcdb4b8c12f787c7b","subject":"update constraint for python-blazarclient to new release 3.3.2","status":"MERGED","created":"2022-08-18 09:47:20.000000000","updated":"2022-08-18 19:08:05.000000000","submitted":"2022-08-18 19:08:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853615-new-release","meta_rev_id":"5faf9a828d9c4a14e96d65633fd5e95355518162","_number":853615,"virtual_id_number":853615,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 19:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-18 17:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-18 17:18:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 10:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-18 11:07:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-18 17:18:27.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c9b833a939111136be462a64d16b67155d9fcea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-18 09:47:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aadd4d68de4a7dff3459505a77b6bd700c1d75e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 10:01:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16028a6c6d8a4871bd30d456e3360cf9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/319cf13315ec46a39d2cc0a24eb76efe : FAILURE in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"2b180e520085c7fbf728b01a64b7aa257f5b42a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:07:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b78293e4b29542368bc07370d1b1190d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0931f4a16074470fb7f60c6fa0cd8b62 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a3e2c78e8604c7086d152072ac46d4f : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af75f20c692a49d0bb4edfa0002d163c : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c841d185e34f4a57bc87ae031a521a9f : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32b94f0a307b4d22bf6401ee30fa5e00 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/59181ef880fa47299374ef841be5907e : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/584cceaad98c44b6b50bb9b5cec1b3fe : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a2f2fb5b1e324bebb3337f7b5c99f342 : SUCCESS in 10m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7a8557cfb314474083ce4c87121eb2d2 : SUCCESS in 5m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/beeefdd380274d42845625e13a33e89a : SUCCESS in 11m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a983294c2a56455b9016be123e9a2aa4 : SUCCESS in 5m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d643f9bfd96345f089490305a18876ee : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9437f21f33894ca69c8bdba74e2192dc : SUCCESS in 9m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f7607d7ff91544629adaad3cba5b9c36 : SUCCESS in 9m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ea32cd49ea9d47e9b935be34ca697a1a : SUCCESS in 17m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5b36e725baf148afa2a88436d0f2a15e : SUCCESS in 11m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/437dc83879164f5595472d0647e7fb93 : SUCCESS in 5m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e71bbad4ec954cd5a6f9a31a93d89bd3 : SUCCESS in 21m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0537855fad2344949f3d9cdb2d86a602 : SUCCESS in 12m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c043d6df64fd4ed08b3e1826d0507471 : SUCCESS in 9m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/088c6f4bccb549edb7bd9b03a7ebc73a : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b086ef9f36d840bebbe00f85bab28cbd : SUCCESS in 19m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/970f8e7c721546e5b09c576c2fbb3449 : SUCCESS in 7m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b444865f83184052b4cb8a3bea241233 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/29d2e96fbc064781a64366f2c5030f76 : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6fc2952564d84a21a639c614fb4b0a0e : SUCCESS in 10m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8a9cb5dcfb24b8f859bf7ad8d01fef7 : SUCCESS in 50m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0bfd96641c84e81a00bf65b02a44469 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f3e61bff9ffbdc058e16fd9dc031e872bcf5d4c6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-18 17:18:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a39fa733c27baf2b1f08ab69cdfb81430839cc02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 17:19:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e038a2a7d8c832c23c2aa463dec9496ca633b8e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 19:08:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f1f68547694a5596a21dbbe4f0e45a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e78a7c9bc3f449cbae549ffdaa85ef1c : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1962b212fd54488a62cd0095b05e6d9 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/018c58e4f8074c5691935f519c90686a : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/137f278fc44b4c8b82ce2cf620fdc592 : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14e32df1b22947b8a7747df9e9cdb7f1 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/23d71605aaab424d8ad73cb98438ad6f : SUCCESS in 8m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce0640cf419b4d4993e4649285a1414e : SUCCESS in 9m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f0c5af0b09044fbe83aab162097463da : SUCCESS in 9m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/26dd209d5ad642fcba60205281fb72b5 : SUCCESS in 6m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b207a74c8425444ba200e1ecea0573bb : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4956bfaa38f940efb0a662a68e955175 : SUCCESS in 8m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23c9f00446fc40b7944bd7073184c84c : SUCCESS in 8m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/daaffafd63524090b7c95f901b7da63d : SUCCESS in 13m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f42dddbcbff644949383f98038e6612d : SUCCESS in 10m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a8886e11e2274011a810b48b590229b2 : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e0b178f1eacd4ce98111754628e3947f : SUCCESS in 21m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b538002f44cc4181acd9d04928788bb6 : SUCCESS in 12m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/49351d53eb684ca8bae7d76b3bdab039 : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/46bb890ff1d14f168eae4a5efd3ca90b : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/add293b2b9cb4bda83cedbff6eb63ece : SUCCESS in 17m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/71df44483a4242d28e30ed8bbec55fb8 : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/efb850eaf5bb48debcc6ebdf04c90ceb : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/32c923e75cde4a73812a1a72715d41b1 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a65de8e64640431b907ae5b8e583f0a1 : SUCCESS in 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e43d9105b4748748d25f0b66ecee3b7 : SUCCESS in 1h 02m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34a73a4b0784486f96fae1b093966c0e : SUCCESS in 1h 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5faf9a828d9c4a14e96d65633fd5e95355518162","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 19:08:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6be7f21cdc8002f797b421303282ae3fd03af487","revisions":{"6be7f21cdc8002f797b421303282ae3fd03af487":{"kind":"REWORK","_number":1,"created":"2022-08-18 09:47:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/853615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/853615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/853615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/853615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/853615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/853615/1"}}},"commit":{"parents":[{"commit":"4b0025275267e3a24884376e22710e0452e9b8d9","subject":"Merge \"update constraint for stevedore to new release 3.4.1\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b0025275267e3a24884376e22710e0452e9b8d9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 09:47:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 09:47:19.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 3.3.2","message":"update constraint for python-blazarclient to new release 3.3.2\n\nmeta: version: 3.3.2\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Change-Id: Ibc4a5fb7b16164e359bece9169ae87884d53bc6d\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib448bb3763198bb1012e57ffcdb4b8c12f787c7b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6be7f21cdc8002f797b421303282ae3fd03af487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6be7f21cdc8002f797b421303282ae3fd03af487"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
