)]}'
{"id":"openstack%2Frequirements~853617","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I7d3214cd5223eab1f204fd8cf05b68179e925ade","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I7d3214cd5223eab1f204fd8cf05b68179e925ade","subject":"update constraint for python-blazarclient to new release 3.2.1","status":"MERGED","created":"2022-08-18 09:49:30.000000000","updated":"2022-08-18 19:08:14.000000000","submitted":"2022-08-18 19:08:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853617-new-release","meta_rev_id":"d6976c8deb96317bb61ce17753242aafbddca023","_number":853617,"virtual_id_number":853617,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-18 19:08:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-18 17:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-18 17:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 10:05:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-18 11:22:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-18 17:18:29.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bfbec4d9de3ad2abea1a3fc16cd0066900533b63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-18 09:49:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4e1840bba698c8911dda8170412717d1216fdc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 10:05:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ad21f877c6d4c68be79b92eac4d3490\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/34740752c13e4626b6a9884359860a4d : FAILURE in 14m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cc4c917b855c739e5d7e129fedb7298512b4dd7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:22:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70990b292a5446298478d1aa7640b74f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4285dfdfdb3a43c1bf0c5c36b79ce937 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b08e640e87374a67a6ce26449a20fe10 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/375f77e72cb34e0b9026a75c70b392bb : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dda079eba24b4afab8e6e250e036bd52 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5441a54f59864562a03c11172c119baf : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e530caf715ed40baa453578231fa3ba1 : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c239d766f1954004b13ad3ff9c345ebe : SUCCESS in 9m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/00e9016b5e17426884797f658203d245 : SUCCESS in 9m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2f2a47ae02944550878effe294bc0c94 : SUCCESS in 13m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/42d1bab582334415b53332a258ae4ef9 : SUCCESS in 5m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/23642c5c3de54d9fba29a8b0699a09fe : SUCCESS in 12m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dfa57bc52dba45d790597c3e56dc9571 : SUCCESS in 9m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/26a2792c3fd7458b990ca663e17e19f9 : SUCCESS in 11m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/acda194908c34a92b6e02e2fee375b10 : SUCCESS in 13m 00s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3cf340f1600a41c2a9a85d7d8345cf1d : SUCCESS in 5m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8c246d0eafab47c0b60afaf6a7679d64 : SUCCESS in 33m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b6bcea4feef54c5486293f305f8e8ed1 : SUCCESS in 11m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/791702649c8647eda1ff4f8db1f96495 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc74710a34684c4cbb860280035ab623 : SUCCESS in 20m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/56d094765e5841a1944f2e4505cdf512 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/94bc81288b22446fab40104a7bfcf82e : SUCCESS in 8m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/04846d09d77a4bca92cdf55a3b4a6e69 : SUCCESS in 11m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ba7be5a2ef84bcbb50b33981ac77229 : SUCCESS in 59m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7dd945ef1bd140ac9bfadc6144d7e700 : SUCCESS in 1h 31m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b761e86fbfbc54f9386fe37ea1eb3b6daad603da","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-18 17:18:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92481f68482d569d6acfe8d0c055f5c488c6f4e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 17:19:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f5ec1d92af96e7f4cbd325f683da871ddafa932","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 19:08:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b393aa077eb145868336209584e4e1ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f993f10a277a4b64ba51ba257f8a9f29 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/012e476f177746998185fda942e6f5f0 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f594ca4343fd4f13831b6a7059fd07c5 : SUCCESS in 7m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/178c8d037f2c4f5e84c4a3795fb33cb3 : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b80b2069e854ac59970f4e8f7987775 : SUCCESS in 4m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7f72ceada3b64e02a2d4306b7a56f447 : SUCCESS in 9m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/90e52c2cb2c1486bad492b0e64c8fe77 : SUCCESS in 11m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c3c30afe078d465e955c6828cb57b388 : SUCCESS in 9m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/066daa9b95e74b81b2ca69cc0258cb9b : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/766fedbe723e40cca01dc32befb70aa8 : SUCCESS in 9m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/173db9beb5c7484486b3f26f7f7db6ae : SUCCESS in 12m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/db3e2d21a8ba49e7af498d57591ee990 : SUCCESS in 9m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/78b397fab0b444b89e10032b2de0d812 : SUCCESS in 11m 36s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/d6fa99e7432d4d02b29c9aa401a18543 : SUCCESS in 8m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bc5e6f1485c34894aeaac3b40a643e8d : SUCCESS in 23m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e98b7e743e741908d82e046002a3b89 : SUCCESS in 14m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d512526010e1426cb4cc499de22b3d6b : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/306cb61dc98947a9a93deed1cb3d52e3 : SUCCESS in 18m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d52e09abd01f4e93a0b3d54dea8ff631 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1a6f6c91841540c2b46e6387e6fb5188 : SUCCESS in 7m 07s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7efd6c472f3941889fa6260ce81dee47 : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b951311fe66742188172b0905732f93b : SUCCESS in 56m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6964029edb9e46c0a3503ba38966fbb5 : SUCCESS in 1h 31m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d6976c8deb96317bb61ce17753242aafbddca023","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 19:08:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa7c262ee118ffcee36df68a518b18b094ec207d","revisions":{"fa7c262ee118ffcee36df68a518b18b094ec207d":{"kind":"REWORK","_number":1,"created":"2022-08-18 09:49:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/17/853617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/17/853617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/853617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/853617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/17/853617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/17/853617/1"}}},"commit":{"parents":[{"commit":"e6211a254b6a86ec29ee22f6ad943cbc1650f3b8","subject":"Merge \"update constraint for oslo.vmware to new release 3.8.2\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6211a254b6a86ec29ee22f6ad943cbc1650f3b8"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 09:49:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 09:49:28.000000000","tz":0},"subject":"update constraint for python-blazarclient to new release 3.2.1","message":"update constraint for python-blazarclient to new release 3.2.1\n\nmeta: version: 3.2.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Commit: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nmeta: release:Change-Id: Ibc4a5fb7b16164e359bece9169ae87884d53bc6d\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7d3214cd5223eab1f204fd8cf05b68179e925ade\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa7c262ee118ffcee36df68a518b18b094ec207d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fa7c262ee118ffcee36df68a518b18b094ec207d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
