)]}'
{"id":"openstack%2Frequirements~853625","triplet_id":"openstack%2Frequirements~master~I1a3082c5b77a860e73c2c856d6ca31e9c6e922a2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1a3082c5b77a860e73c2c856d6ca31e9c6e922a2","subject":"update constraint for ovsdbapp to new release 2.1.0","status":"MERGED","created":"2022-08-18 11:06:30.000000000","updated":"2022-08-20 03:10:42.000000000","submitted":"2022-08-20 03:09:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853625-new-release","meta_rev_id":"864b6edf4ee775f46043e256d1fde34e20319313","_number":853625,"virtual_id_number":853625,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-20 03:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-20 01:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-20 01:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-18 11:19:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-18 12:24:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-19 16:20:05.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2022-08-20 01:18:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e5563f11882c572843f38d61b3b98dbb815b73de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-18 11:06:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b32226952db274627c31e0ea0bf74c7e1a2ab335","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 11:19:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac9f349aa5324d819c1187e03a27e3ce\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b1c1ec929ef44110aa32c779cda751f5 : FAILURE in 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b2cbf20040caec4507eae9c77b010c48a1b0cb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 12:24:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3a92e76a76f44b48ec37d778dd76391\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a97797b55214e369ed026ab2f995cc0 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55df93178edb4868b6607efc08b2b0da : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd48a8a48f28479999e85bbfc0f39d6f : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16a7cc90ab2d4b0284fb32399881eee4 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b64ec95218e4e14adb3ed185d69e9e8 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f3fd593ffe374a02af10cd232eefdc2f : SUCCESS in 3m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/541bba9629d24a289b3dabe768b2dfa7 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fdde8a3025a142ee9cc1ac307e13bb51 : SUCCESS in 8m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d492ebfa65ee4abd8e086e85d7a03e19 : FAILURE in 3m 28s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/bc0fbd606644453caf22393b146aa285 : SUCCESS in 7m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2a0e2dbaccc84fcc957efa28d79dbeb4 : SUCCESS in 4m 47s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/ec0e9632e07a40d4abc4cb9b1171bc18 : TIMED_OUT in 40m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/31a0fa1ed58e45c0bb263aef20fae597 : FAILURE in 10m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a522ee6ae2b949b0b001c5ba6a2a9b1d : SUCCESS in 4m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/144c8cf62f394eec88e3d5cd8725b81d : SUCCESS in 4m 51s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7a2a281e36f2462a855274345c695977 : SUCCESS in 8m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ffd025189ada4e40bbda6ee523659abf : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e04374ce0f464744a97bc9d45e93ad42 : SUCCESS in 7m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3c47eca138a74da4a6a51c61ee78b586 : SUCCESS in 8m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8148435f9f1e44b78d90480362b72060 : SUCCESS in 11m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/120a639357374361bbf810c5b87d653e : SUCCESS in 3m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3cd0e8475d9f4ecd97ed841bd26c2c30 : SUCCESS in 8m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/306636e766624f06bb31e0c01e406c8f : SUCCESS in 5m 40s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/360f1e917d4f4115a1bf8ccacecc3829 : SUCCESS in 9m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6969b78d0d724a25be319e36c935d225 : SUCCESS in 26m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/949bdab1220846b09a521a0ad6df777b : SUCCESS in 16m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a6e04df4247c44d19c3cdaaec6361571 : SUCCESS in 6m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4de59aae2c52425b8b9601bda1bfb5af : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74c87ce72d2b46ea932205981f3ed131 : SUCCESS in 21m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/49379c1497d2429cb5e175619d5ce389 : SUCCESS in 8m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/18588623568d48eea78c1fa090d581c7 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d0fbb949d744812a230b2397509d78c : SUCCESS in 6m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4fc5e51b5ea64bc78213dfa66f401144 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0240967251364876acf0f805e15acace : SUCCESS in 46m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a47bd05fa70845f2a854e842db85c1b6 : SUCCESS in 1h 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9078ed024d996625267b938b5b85e8ff02241267","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-19 16:20:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91c2b664290d675134183d6c5ff95512d58b09bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 16:35:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b512513042d6499c84293b7fe4bf2952\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e6d5c281e4604db695f1a365a62206c8 : FAILURE in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"208404a8022e4820ed405e950144a230bc88c9be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 17:54:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c146097ea4a04a37bb46db41dfd7f9a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24dfde0a56a84b98a09516c524024dcf : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421e6f3e28a14384aa00bd27bf1e417e : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbe29e2a99b44e8181d382db600e773f : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e95437fcce24ea5af4947ad16fc29eb : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7b3f10918724b59849b9051013bc515 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c8d93fbd9075424e91b71183e16e5082 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44bfccd719254e5c92f1efaac8037efe : SUCCESS in 4m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fe44c3dce16b4f43bca8cdea98a41089 : SUCCESS in 10m 11s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fbdf0b400d134764867ebe5de53cb479 : SUCCESS in 13m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f66d1bb377e84a54a1c6bdc42899b501 : SUCCESS in 10m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8daf53839e5f49b8a8472347a82f2c8d : SUCCESS in 6m 09s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/02785ad1b0fe4c569432a2fb4a4b1017 : SUCCESS in 6m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/37d90a743d4f46869737a0f5f787a43b : SUCCESS in 10m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3166156a250348eea8abf60736466b3c : SUCCESS in 5m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/439d7be0f8614a17a97857ad34cad52f : SUCCESS in 6m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/743ee1e416bc4fb7a59766e31d6b1446 : SUCCESS in 8m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b723c524b11946eb8eb24f698b130509 : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64a2f40fdd3e42bd91df184de1e771e6 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ecf4cd1df8384633862518e73402a687 : SUCCESS in 9m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d97a0e588e99465d9ae25accd4cb001e : SUCCESS in 13m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/19cd36405e06420bb40d24084f540d97 : SUCCESS in 4m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d5dbe3dc690f4ce6b8cd19409738544f : SUCCESS in 10m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8fb09a91ace14f92b5d3b3f6da234df0 : SUCCESS in 5m 29s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/74f69641edf24eea900679678aa8f723 : SUCCESS in 9m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/91dc2803820a49b6ada1cae8cd4d0247 : SUCCESS in 23m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/21d3d588e286499799a5fa431e88634e : SUCCESS in 13m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/06f89d4bd39e48d988e44bf24c22af63 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b4d6d017f1cb4d26b011c3caf723bfd2 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fe2448a15460455d9b3286e9ee0458b3 : SUCCESS in 23m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/49ef1970e3184e859ef102bc3db62ece : SUCCESS in 6m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c446d257cf8441f09a1cb34fca0e927b : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3baf1f19cb26455897a8e7fd10ab0ccb : SUCCESS in 7m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0df9f26bb9264a1ab5ea14704708a652 : SUCCESS in 10m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9919f6bb3cf14b6aa30ce779dd89b41b : SUCCESS in 50m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe7f4704ac9e4046a37607a430dc5f49 : SUCCESS in 1h 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"771df30eae733774eac6cc50e55c9cec1b80bb18","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-20 01:18:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"84f418fb240b96c67bb545f9069164895f274f5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 01:19:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d80d3dba7c7cb42b1ddffcedd4bd537385e8f14c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 03:09:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/021ee835d7ad4005a3c5d552224e7f87\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7293e830f750437790bebb7106b18463 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782607e5663d4bf9bc191ff59e6f93e0 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78d4d0bfe68944d09c7e006da4248871 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5636987bbea4279bc944f4eed0a79a6 : SUCCESS in 7m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d21d814b91d4b619ed44e4bd656dae9 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ab67cebb5b3f4aa98cd610a91838be9c : SUCCESS in 4m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e3022ece0695414c9f8f7786765722c7 : SUCCESS in 10m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/465e54ae5fb243719daadb2fbf2bca7b : SUCCESS in 14m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3555c38253884cc5877bc9fdfa311964 : SUCCESS in 10m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6982a19f955c4487bf000548f6d7458a : SUCCESS in 4m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ee19757ad4e347cc98a6c8db0479717c : SUCCESS in 6m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7adb5d63d2724bc99230ba3c360c7678 : SUCCESS in 8m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/37f86b3f20e54a4897a835ce8337d8a4 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/04d0edfa37d647549daba64700f23e01 : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d7d51b4a13ae4837a65695e8b9e1c416 : SUCCESS in 7m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/56a2e492b4684a2ebd7205a1e5616ac9 : SUCCESS in 26m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7e7df5abcde24fa5b69aa968615927df : SUCCESS in 7m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/56a89d851a7d4dcfaaf55979dc20615c : SUCCESS in 10m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/83344bda84ea422d90e5154779dc8c1b : SUCCESS in 5m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7aa8b8cfb9024a169ab8ec180eadf864 : SUCCESS in 30m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b038aefe78754034a9d31294bfc33e00 : SUCCESS in 13m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a23d3e4dfd704a4985b35257a8335459 : SUCCESS in 8m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4bb45914f25c42f68d53a8e16b8abdcd : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11c3dd39cc3a40c2b33d87474156922e : SUCCESS in 19m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8c33f431a98d4d549c28323ddf089c3c : SUCCESS in 7m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4d7d9b6686bd46ffb836196a8856db4b : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9087c7b5e83a45c1b76a253cf49c367e : SUCCESS in 9m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/33b5863911dc45beb8054728a93a7ce3 : SUCCESS in 8m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83cc49b0a3e74f02b0c7da2b3bac8995 : SUCCESS in 47m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bed1b9ad23a439982174e978de88350 : SUCCESS in 1h 47m 30s","accounts_in_message":[],"_revision_number":1},{"id":"01320bb5f6ecd1e7a7201d6b84223cb701478cfb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 03:09:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"864b6edf4ee775f46043e256d1fde34e20319313","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 03:10:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55b0bd5671f44148bd8048bb0a20ca1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759e6727ac144531901b7577e17464d2 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2111c7748e950c7de963a36490bd63ccedacb970","revisions":{"2111c7748e950c7de963a36490bd63ccedacb970":{"kind":"REWORK","_number":1,"created":"2022-08-18 11:06:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/25/853625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/25/853625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/853625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/853625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/25/853625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/25/853625/1"}}},"commit":{"parents":[{"commit":"8da71feebcb4dc9e8df3be6bb44f69774cbd02ea","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8da71feebcb4dc9e8df3be6bb44f69774cbd02ea"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 11:06:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-18 11:06:28.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 2.1.0","message":"update constraint for ovsdbapp to new release 2.1.0\n\nmeta: version: 2.1.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I68d3d4dff5882a289204f80c1ae0f5f9b9b5fa07\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1a3082c5b77a860e73c2c856d6ca31e9c6e922a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2111c7748e950c7de963a36490bd63ccedacb970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2111c7748e950c7de963a36490bd63ccedacb970"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
