)]}'
{"id":"openstack%2Frequirements~854038","triplet_id":"openstack%2Frequirements~master~I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3","project":"openstack/requirements","branch":"master","topic":"constraints-py36","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-08-30 14:36:46.000000000","reason":"Change was abandoned"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-08-29 23:13:08.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-08-30 14:36:46.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-08-30 14:36:46.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3","subject":"Update requirements for python36","status":"ABANDONED","created":"2022-08-22 18:52:36.000000000","updated":"2022-08-30 14:36:46.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"552642347bc34848abfa339f1bc0a441a745cc15","_number":854038,"virtual_id_number":854038,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-29 19:40:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2022-08-29 23:13:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 19:09:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-22 20:11:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-08-22 20:16:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 10:18:08.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-08-29 10:18:08.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-08-29 10:19:27.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-08-29 21:37:17.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"}],"messages":[{"id":"ffeb621897b83abf64471b2149a7d52a3af20840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-22 18:52:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c908245b575657a1528319ae70d8baafd95bdf0b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-22 18:54:33.000000000","message":"Topic cliff_constraint_py36 removed","accounts_in_message":[],"_revision_number":1},{"id":"d28a9ad6ed8961d5e29f929f4e83f8c009502bd6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-22 18:54:49.000000000","message":"Topic set to costraints_py36","accounts_in_message":[],"_revision_number":1},{"id":"0c00df52a08712f13e6caa7d574c488b8d1a6411","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 19:09:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dee0201161a4ce2b16c724eb70c862c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b58a9b7fa9444b05908564dbe6c8e3d1 : FAILURE in 16m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1ca86defdd915dbc192a65de0954ee33c6e26a6d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-22 20:11:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11430a0cb4fa32359e188490e5231f9f5fce4201","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 20:16:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aef11975bd6f4a858787d27147346b05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2244fec72a46dd977f924885106242 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2025f6ba9eff4a2bbae788a2bb62addb : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98a409b6405f450eb0d4ca5141c05837 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c53cef019e2846f5a125ea80c53d3d9f : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f371e840965049008f1c8ec53065830b : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/168bda832b9843ccb423b8102ebf9eba : SUCCESS in 6m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22b3ff7e9006453580412ffdbb17230b : SUCCESS in 6m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/22898d54fc1d4edbbcb6bad002ccfa39 : SUCCESS in 7m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fb63162de88245828a0a9171dca7b586 : SUCCESS in 12m 09s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/0159108c7df84e55bd083676554c5a1c : SUCCESS in 9m 40s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cec9f9961d27495b9f2c8257d9dcb838 : SUCCESS in 5m 27s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/aba7c7cad870446ea86eda7ea52948ac : SUCCESS in 6m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/21728f53e3084d0592bc8aee93bf2276 : SUCCESS in 10m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b67529a6b0ff43e1b2a195379b867048 : SUCCESS in 5m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e76093382d7d4c85af07d613851e404f : SUCCESS in 6m 18s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/18c26f2ba927467ead3d95374d59bfd7 : SUCCESS in 8m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0d214591f14b49eea11ee15861d61485 : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6930b9b183ff4fc88c284e10ab31d8ad : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/54acc569b4044fdbaa465dc5707d85d9 : SUCCESS in 8m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c0ff48c186004d63895eb87566402cc6 : SUCCESS in 14m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c63cef6b88c746c1864cf248a6c7d566 : SUCCESS in 4m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d687296d4b7349258d01afdd5c3652a0 : SUCCESS in 11m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/13c2e8f7d36348719d45f526d13b7987 : SUCCESS in 5m 30s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/9acd79f2c30f44faa5a0da786a4610ca : SUCCESS in 9m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d5e6757bb5d340c2b6e17ff436e8c902 : SUCCESS in 21m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2acd8b75cb1e4a8c85d57a1407a15c02 : SUCCESS in 13m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2035070b50a84b429790543b15bcf090 : SUCCESS in 6m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72f9d7de23304f1888511a4df7f689ad : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d80cd4934b194eaa8c28a8384e9f0338 : SUCCESS in 20m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/da9afb059cd64fa2a6e3a471d9c600a2 : SUCCESS in 7m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c666487bd34b4dc49e932f58ee0e4360 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5e33abf9e816459a909ff0a7d7ad78c4 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6d5f8761974c4b0ab7873192775ceb1f : SUCCESS in 12m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/577c6b63f7b148c6926a87873764bb43 : SUCCESS in 52m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2b1751dd3734f8c869d66a844f152dd : SUCCESS in 1h 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"90b5e95dc981c2086f45e0348492cc9396df8b6b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-23 09:29:25.000000000","message":"Abandoned\n\nProposed on yoga instead: https://review.opendev.org/c/openstack/requirements/+/854130","accounts_in_message":[],"_revision_number":1},{"id":"54b4c7197367098242c349cf6e88f38f1b59f573","tag":"autogenerated:gerrit:restore","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-29 10:16:17.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"f9e2f88f0f3a4f8a1df5d18106cbb6e8defb9511","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-29 10:16:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24eadb13a6f0c9a5a4969820aa29e2a0336f5af5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 10:30:08.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ae7b12912314fce985396824697468a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d3e87361a024816a6fdefc933d44e5f : FAILURE in 11m 25s","accounts_in_message":[],"_revision_number":2},{"id":"cc4edc6d7c460f6ee996d4c9aab7fdff51a2c8f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-29 10:33:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e8032c55c0ee7dc6785fe93616ec2586d910fb0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 11:20:19.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1f6387758a54e80be1657dc3b7685b6\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/68d3dffc5f174871bb51858a44284951 : FAILURE in 43m 36s","accounts_in_message":[],"_revision_number":3},{"id":"7b777d949a0f3209af3d99a10414e28ae54a40f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 11:59:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b24a67a2661f4eb1ab33250bccb4063f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91662a8b1085408195cc324a68579ffb : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec3c20df0b3f4182acb129ecc1d6d520 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45a3379d01994d38bb7608ba207c4837 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc4e80a3b96d4ea0967b0086e2e1cc9c : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4117f3be7e04395995d865bd3949372 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/736c62cb8e9f40a789145481de9155e9 : FAILURE in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/697d57528a494ec8a3d3f281854a5bab : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6995d271b84d4a449b3cc3170a725b3b : SUCCESS in 8m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/41ba14934af54c61bb37932afe14077b : SUCCESS in 11m 51s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ac3615e0733c4d38aeedcc289537dcf8 : SUCCESS in 6m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d8e6ba8d504249e3b8aa41d159a3c038 : SUCCESS in 6m 35s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/284544945e6d4753a78ae8bcbb18cbc2 : SUCCESS in 6m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a4453c714d51483882d84805bac0f6cc : SUCCESS in 10m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a0e756bfb62f4c5fbefe20169b8de54a : SUCCESS in 4m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/20b948dae0e54058808114828ffd96e4 : SUCCESS in 5m 40s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bee5153576dc4164a7ba1a1f14e01c01 : SUCCESS in 8m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/80dadb09b3d442089812f8bd5ccd2913 : SUCCESS in 9m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fafdfa91c44d4922843bccc33a589603 : SUCCESS in 8m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ba4b2b02b9be44bea7a2d0c900f2ebcc : SUCCESS in 7m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/444187de7964459b929cac8ac4743395 : SUCCESS in 11m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e4d9b7f3af184dabbcac7fddd344d56a : SUCCESS in 4m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1086b8cf7e90470095aa0f70612610f6 : SUCCESS in 9m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9930c290ed0c44fcb6880baef3c9e057 : SUCCESS in 5m 12s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/9753024ab5de4a139e3cf97371c11549 : SUCCESS in 9m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8b09c18227b54811a6de2f14dd9c4715 : SUCCESS in 26m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2ddfdbf8f2924035897f0314b657ac88 : SUCCESS in 13m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c712eb9388a64f02bd1169134052f633 : SUCCESS in 5m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7bddc66591c84a298f45d7079bd1bbe0 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2fbfb8c70b6c455cb1b71b16d1a95c04 : SUCCESS in 22m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8c952c02dde4486d81a3bd347bf15c3d : SUCCESS in 5m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/49e8c39b266f4009bcad7d6a1b9e0662 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/27a9c63f6c444f7ca714bc2b6f4e4048 : SUCCESS in 8m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7f921558922c4d58b699c7ebacf4449b : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61f6950fbdde44409a426c6cee9125f2 : SUCCESS in 46m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4051d235644e43028fb7097ed93e8eda : SUCCESS in 1h 24m 17s","accounts_in_message":[],"_revision_number":3},{"id":"2a54581e9dedbe1ac3a470f9687a6ea68809aaeb","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-29 14:51:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"90de1c90e0845d7d80d14f76a685529104f815b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-29 17:59:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a8f6007ea0a3ff3d0ba6a41550e297f94695c44a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 18:59:45.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/540d9000831c4126a15999bf2169c2c7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c7583630645043c9a2b0998a2cba9a29 : FAILURE in 13m 04s","accounts_in_message":[],"_revision_number":4},{"id":"fbff135237fc09ed25351d4cf089911b0449e90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 19:40:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d5916bf644348ab8600b42528d92f5a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/604b75dfe45842bdbb8680d080c72313 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/262fbdaf4c8349e2adab4729bdc6f9d6 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/235e8576763349f8b5c9a7c332316ef3 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/711ae780178349a9b9c95a32b2214638 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df6a495792824be082c82f186bda5542 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7fe79136ccb34424ae0255023c9b6f8a : SUCCESS in 5m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71009459d70945cd9612e8518a3a929c : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f3f14fe0d8604a9db6d662c6d5c8e566 : SUCCESS in 8m 38s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f9d2bd19b17d4b03a3f786fdf4a21c7f : SUCCESS in 12m 48s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/900e2580ee8a4ec7ac24c81824e937bf : SUCCESS in 7m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/35ce6c6ed6824689bae7e6f190843d0d : SUCCESS in 5m 08s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/8becc2bd9f8045efa50cfb7b502d195b : SUCCESS in 5m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bf6de332b3db421d932414cd256b9e15 : SUCCESS in 10m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/655624bb2afd411bab81124dd165852b : SUCCESS in 5m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d8d51073e20e4629a546d1e38948e44a : SUCCESS in 5m 04s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e5e02033384a47229c5378c66fe4014b : SUCCESS in 8m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fd65a87854254bf494041ea5e340f246 : SUCCESS in 10m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/041d911e5a6c4ef7be58a922eef5486b : SUCCESS in 9m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1f192547249e498dbcca376cf102d0a3 : SUCCESS in 8m 35s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2c6820bb5bec46628bfc8d90283a8081 : SUCCESS in 13m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d7a1ad13f89846ea864c51cc326007e7 : SUCCESS in 3m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a3c58e276ed0406689e8c11123ecc4eb : SUCCESS in 9m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/59d088384e50404f9a228e106bf9e9b9 : SUCCESS in 6m 18s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/29aab441a199439fb035bf83d4851e02 : SUCCESS in 9m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b29324fe5c0e44c381adcbbff08b8095 : SUCCESS in 32m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c2c32dc081dc49ada6470c4e3f03a6e8 : SUCCESS in 12m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7f43e02328424b2ab7a575a474ab1ad8 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6df0824cb2442f5b423f96aad30157e : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b7731f46033547a7beb57828b564b0e2 : SUCCESS in 23m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/86a5b35f4a62495c9fcf3d95ec7ac81e : SUCCESS in 5m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ab1df7a78a1f4227af2772a505124da6 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad7600c72e6d4defa4407702ebd5c82d : SUCCESS in 9m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bd08d68b6c7a4dfc82e6a898b130545b : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5a58f2892bd49639488a8183bd8d5e7 : SUCCESS in 1h 06m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a8a6beeea2a487eafc922202446001f : SUCCESS in 1h 35m 19s","accounts_in_message":[],"_revision_number":4},{"id":"ffd9157c4031efbd26fe8c1d871b669e319f3f73","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-08-29 21:37:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac3cee135d3de8008e933be14f1d482c83dc78bc","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-29 23:13:08.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"552642347bc34848abfa339f1bc0a441a745cc15","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-30 14:36:46.000000000","message":"Abandoned\n\nChange is not needed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b929e61c64d7db0d5b04a137ef7a4f6e570623db","revisions":{"c0a45c45a97ebd6a23f2a264f65f340f2dd85db5":{"kind":"REWORK","_number":1,"created":"2022-08-22 18:52:36.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/38/854038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/854038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/854038/1"}}},"commit":{"parents":[{"commit":"d4f0334e7ee88ae8712a9ee283916c0a2c9ba0e2","subject":"Merge \"update constraint for ovsdbapp to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4f0334e7ee88ae8712a9ee283916c0a2c9ba0e2"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:51:51.000000000","tz":60},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nSupport for python3.6 in jsonschema was dropped in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0a45c45a97ebd6a23f2a264f65f340f2dd85db5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c0a45c45a97ebd6a23f2a264f65f340f2dd85db5"}]},"branch":"refs/heads/master"},"710b45f068dfb1aa7f2237c0c5ea029fb0aa74ab":{"kind":"REWORK","_number":2,"created":"2022-08-29 10:16:24.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/38/854038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/854038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/854038/2"}}},"commit":{"parents":[{"commit":"b86048c24d57b12d187d466ac9b603ace17795ed","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b86048c24d57b12d187d466ac9b603ace17795ed"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-29 10:13:00.000000000","tz":60},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nWarlock requires jsonschema \u003e v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/710b45f068dfb1aa7f2237c0c5ea029fb0aa74ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/710b45f068dfb1aa7f2237c0c5ea029fb0aa74ab"}]},"branch":"refs/heads/master"},"4b686e49c49049c8791e694fbab4f3bfc8147482":{"kind":"REWORK","_number":3,"created":"2022-08-29 10:33:13.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/38/854038/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/854038/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/854038/3"}}},"commit":{"parents":[{"commit":"b86048c24d57b12d187d466ac9b603ace17795ed","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b86048c24d57b12d187d466ac9b603ace17795ed"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-29 10:33:00.000000000","tz":60},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nWarlock requires jsonschema \u003e v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b686e49c49049c8791e694fbab4f3bfc8147482"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4b686e49c49049c8791e694fbab4f3bfc8147482"}]},"branch":"refs/heads/master"},"b929e61c64d7db0d5b04a137ef7a4f6e570623db":{"kind":"REWORK","_number":4,"created":"2022-08-29 17:59:11.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/38/854038/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/38/854038/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/38/854038/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/38/854038/4"}}},"commit":{"parents":[{"commit":"b86048c24d57b12d187d466ac9b603ace17795ed","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b86048c24d57b12d187d466ac9b603ace17795ed"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-29 17:58:05.000000000","tz":60},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nWarlock (transient requirement through glance) dropped py36 support in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b929e61c64d7db0d5b04a137ef7a4f6e570623db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b929e61c64d7db0d5b04a137ef7a4f6e570623db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
