)]}'
{"id":"openstack%2Frequirements~854130","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3","project":"openstack/requirements","branch":"stable/yoga","topic":"costraints_py36-stable/yoga","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-08-29 17:59:37.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-08-29 17:59:37.000000000","reason":"Change was abandoned"},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-08-29 09:31:26.000000000","reason":"\u003cGERRIT_ACCOUNT_13177\u003e replied on the change","reason_account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-08-29 17:59:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3","subject":"Update requirements for python36","status":"ABANDONED","created":"2022-08-23 09:26:18.000000000","updated":"2022-08-29 17:59:37.000000000","total_comment_count":18,"unresolved_comment_count":4,"has_review_started":true,"cherry_pick_of_change":854038,"cherry_pick_of_patch_set":1,"meta_rev_id":"732f55b89f7912fc04a36a8a4788099be630c54a","_number":854130,"virtual_id_number":854130,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-26 20:48:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":-1,"date":"2022-08-26 19:00:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-23 09:29:53.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-08-23 09:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-23 10:09:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-23 16:50:50.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"CC"},{"updated":"2022-08-25 12:50:20.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-08-25 13:25:55.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-08-25 19:48:14.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-08-26 14:30:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-08-26 14:30:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"91c43a69ebded42b77a9333c92bad54a708c45c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-23 09:26:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* upper-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"d81f35386d5484e3d88445edb6563c2c59c7c51f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-23 09:28:32.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8069d884c3cb148b7801a528c412fdfd3cd0edf9","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-23 09:28:37.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"74b400e87826a8fd30c54c5706dbec1f5c3ca97b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:53:38.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a429e7fd9ea24bc0875c7e5ccaa6f9a8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/05bf3dc935234fcd946b59bf07308aa3 : FAILURE in 8m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6c81bce7a6f6db82f23793d3b660e5ea5be4fd76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 10:09:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b81cba25306540989505ea6fc718053d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e5a4af304c4b959d135afc92e1af07 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1059729e0fc4f36b753f20e9106a7fd : FAILURE in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7efe5c99df094df99a484a570c13be00 : FAILURE in 2m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe5ce49920074157a980af911504886d : FAILURE in 2m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72b9fb429c364d618d837b1d0940a4e3 : SUCCESS in 3m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f3cf9377c53c4c8a85649cdf84d6a427 : FAILURE in 4m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0b4792f8686d408d81b2517dc24da515 : FAILURE in 3m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/55a6cb4ba9ef43ffaad9d6e083777fe2 : FAILURE in 3m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f2ee0ae0a9d4423983ed60ac8bcc413c : FAILURE in 4m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/277fff63019642339d00cbedeb927531 : FAILURE in 2m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4f68511c418e47e6849972e88d884648 : FAILURE in 3m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0e300c25a13d4a5aa2fc555c717b9e50 : FAILURE in 2m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/810d14c308d147958ff1f3a9b3713863 : FAILURE in 5m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca8f891ee36847d197250f416fd6b01d : FAILURE in 4m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2697a4a38fd7425d8ee097d62f28adf6 : FAILURE in 4m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/217089c5c06743218b81158afd73f372 : FAILURE in 2m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8d6c75bc88fc4d6bb3d15c3ba4aba816 : FAILURE in 6m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6cee48f309f748d8be32f43c17bb469d : FAILURE in 3m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/157e578ccaf941acbfcfe3ae13a86b43 : FAILURE in 3m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d17e180fa0064366a80f54ae8511ce83 : FAILURE in 4m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f6d6a3347a6b45acbfe3f94da5682f5d : FAILURE in 4m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4988091a5c104e64a641c5f03bed7a61 : FAILURE in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a20b2fa7587414993a8a340b8a3b973 : FAILURE in 4m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/efa18a016aae4ea2ba1d05acc0b9aa26 : FAILURE in 4m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/373e192333f04f8c956d9590ae9c3777 : FAILURE in 3m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/277eecb5a11649c8a12471f8e83e412f : FAILURE in 4m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c00a605bb25a42ac99f8a43728c84849 : FAILURE in 6m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0333e2dfd34044c0bc719fd23bb0dbd4 : POST_FAILURE in 14m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45293c13a1b04f46876bceef4ed38576 : FAILURE in 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1ec306d2dd1353efd45f4e9f026ab3e942235b15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-23 10:24:12.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c4027b6b66b6a88b80642d808a2383643509f9ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 10:34:01.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48d9831c2541453481e2c8db27a9ad5c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1150181631954072b13d657e655c44f8 : FAILURE in 7m 49s","accounts_in_message":[],"_revision_number":3},{"id":"b0b30d11f9cffe111e61d95ade785ab48a23afab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 10:39:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc9ccbb17c2b436e9b9ade1bf234c17d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd8288157974073827584439d337e43 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/252afcf169ac4cb99f80958feae9aaa9 : FAILURE in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a40ebaa09ede48a8af3e6c4707d6cf4d : FAILURE in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c7c77d5edbc419ea62d03f49d7bff8b : FAILURE in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7e5ef56364f4d5585065fc87b25a9d6 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1ac5ad623f4b4b11977387240d6c9e0f : FAILURE in 4m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/79ee3c657fe947b9b8c86ed388c93dc8 : FAILURE in 4m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b4099f97155247598c8bfa4d047764e2 : FAILURE in 4m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4b7eb729179542d6984f795a798417df : FAILURE in 4m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c16823b59060441abc57a2ecf107ffd1 : FAILURE in 3m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/818136d9ea4644deaa9e5a5a48ea53d4 : FAILURE in 4m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/449a1a526a0a4e1ab55575dbe799c08a : FAILURE in 3m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1f1721500047451fb55dd743f907e646 : FAILURE in 6m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ec01bb96937472ab5ccb1125c93bb44 : FAILURE in 4m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0a8a1d26fbe2460480fb9db69a9372ef : FAILURE in 5m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fab6814660de4bbfb68a461c5b07a944 : FAILURE in 3m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fdfaec9e180f4ade8c3e27fd8511edcc : FAILURE in 4m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/82a81d1f78b94182a42465c7d85d182a : FAILURE in 4m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f35810dea6b2439ab34b5e540dc7d8cd : FAILURE in 5m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8e35ea095cea44f88cc2523e517f1c47 : FAILURE in 4m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f575d9285be84eb3b5b7a64bd4e17369 : FAILURE in 5m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2ac0ec2e98b5437daa1a7d96586d54c8 : FAILURE in 5m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6fdf5d87793b49609fb352aefd390474 : FAILURE in 6m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0fcf37e27bbd4080b391914d8677d7a3 : FAILURE in 4m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/84b57b3da4cf4f3aae555e0252f74eb9 : FAILURE in 3m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec9fb425cba540558c46b6558c9146a6 : FAILURE in 4m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b021a95e2baf48e8ba67e896cbbfd7b0 : FAILURE in 3m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54235de7302f40eab847c0c82f7d2d34 : POST_FAILURE in 11m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0123284f0e554e48b4d2da7ee163e1dd : FAILURE in 12m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ba560e08651b2b26cc21f5d5b55112b7ca2d9b36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-23 10:53:54.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1fc9c285ac58e4eb30aa2761d63354e553200716","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 11:09:33.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed561b95fb874ef88baf8b1285674c58\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/691046783d264eb4bc4b34fe26b409b1 : FAILURE in 13m 48s","accounts_in_message":[],"_revision_number":4},{"id":"8717c53e8acd40ad19301268413a2b362bd1c26f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-23 11:19:40.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a43e2fd42aaf5f3fcd4f2244a1297c8affb4a3b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 11:31:19.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36849fe49cc048b896499ec63e989cd3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fe388034be924f8fbe1cf87d7c18e4d8 : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":5},{"id":"44bc26269c73f8519e9c9562dea408fa62dc2af8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:55:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b9d8163118f4b92ada3e119854cc3c4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84f51f0a7b6b4ddd9c92831e9076ba2f : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b5c34ff432b4eab9568fec04c68970d : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80c183599c4842e7aaade5997aef9098 : SUCCESS in 3m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7501d775a36e43219f6de6607d9bbc29 : FAILURE in 2m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ee3bf5094c24489eb19d6ddfdfd02c15 : SUCCESS in 3m 02s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d4e8fa26b89343b1bbd2d86543f2e317 : SUCCESS in 8m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/89bcca2343f14a49b1c6828c10da3456 : SUCCESS in 8m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/66247811163d4ebd89c579952e3da215 : SUCCESS in 5m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/81e6108a691e48468c28d65489283b3f : SUCCESS in 9m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5774efe20cc843c59c741a99f394eb7f : SUCCESS in 3m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/355333444ae5490f9d1f73b522661745 : SUCCESS in 4m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7b82cc6702234b6398923d19de0b1505 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2c32b0eb8c641de906b19d5af09cb81 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1e93e5a7e4ae4ec9986bda03f2a64151 : SUCCESS in 7m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a0da570d570e4f909ac695bdc5f15561 : SUCCESS in 13m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/16b2f65ff4e243db849815e61d1be1cc : SUCCESS in 3m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4bdfb07fef314b50a3f6bdaf01e88d24 : SUCCESS in 10m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eea17ce4cc6348d881954f94e7d572ba : SUCCESS in 4m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/411e5c54bd614804a6b2539bb5f62e9d : SUCCESS in 25m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7f39b7bd1ded41bab387d0dedf8572b2 : SUCCESS in 12m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2ec41eff416e4d389377cd1b27636120 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd57f01ea0484ab2a9db02fea88962c6 : SUCCESS in 3m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28a8948c9a98496392097af2911145c3 : SUCCESS in 21m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4f02960ebf604af3bafa7652a777b346 : SUCCESS in 5m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2e279b0a9af644bea628ad2d26bb1e5e : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72c1042356e441a796610c8923d68de1 : SUCCESS in 5m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d0eb7b050c274f6b9f8c85d24a4cf088 : SUCCESS in 13m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e60dd5b337f84c40831a7250a6393749 : SUCCESS in 48m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a901d5ccf0a9475a83fdc31f1f1705ed : SUCCESS in 1h 30m 42s","accounts_in_message":[],"_revision_number":5},{"id":"9d379c14e67af0c9d84834f49a91f74326de527d","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-08-23 16:50:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a1d379c61df2a66affb67dbdd4027a1b9b2558e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-24 17:09:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3f660463c6a1bbf7f9d37784c26e64185c705fde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 17:21:47.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab2a28352e54d47b8d2bc5bcfa0b9fc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4fab0dd0f9f54fca93c3945793ae47d7 : FAILURE in 10m 06s","accounts_in_message":[],"_revision_number":6},{"id":"5b8c64c49acd78e94cc013b5aa6285c3e5b0dad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 18:33:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da0434bca14d4c21934ca13e68cd8598\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34067aca7f1c4561b316c31b41f4a9d3 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bdf723a5d1c4ec08599a4c2a5ab6df2 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/432a8264eea647808cd4a9ef53bc59e1 : SUCCESS in 5m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30ad2e7469914c838f9653e68ed73b49 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d54c0c6311034978b0e5bc6a5a10afee : SUCCESS in 3m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c07b3dabe90f4459819a9cc82c717c60 : SUCCESS in 11m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/30e612b810f44e1db0bed93acc92f571 : FAILURE in 7m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/efe5385f0da640539d240638e0a05b8f : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dd22401b8e4545409f13a734e8929d7b : FAILURE in 5m 48s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/848cb8642d8a482982c88ee501fd951d : FAILURE in 4m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/02d9f1da3f00459bb78098ff192f75e2 : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/69944a4a017b4cabb17e5abe5f7fbe70 : FAILURE in 2m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2f33bcb2d874c68aa386634224e7360 : FAILURE in 6m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7d9286c1744d4afd91171f1ba9c3156c : FAILURE in 5m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/391c2d9d4be2494ab74dd310c56d1734 : SUCCESS in 12m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/3149cfc4818c44b68e8c11fb3e169140 : SUCCESS in 5m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e1f4caf4f7ee4d16b48b053040fb1c48 : FAILURE in 4m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f6ad31c041814e6881dab81698eb7b0e : SUCCESS in 4m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/42b1a7ed2f8c4077b2bad8f133cacca9 : SUCCESS in 21m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2bd218a4b71f4b63b4c948644709942a : FAILURE in 5m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/24205f20bdd94ba3b99e28d3ea0cd155 : SUCCESS in 6m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dfefab680f043f9b3570dd68b124248 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34e337bf80654c608f0cb7097d2e31fa : FAILURE in 6m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c5bbfb140b2b45b4a5bf7c35d308c1fa : SUCCESS in 5m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/33a7bf2d80fb4f07a16ac025d433541d : FAILURE in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2a97c588616453e9a146c293f33390e : FAILURE in 4m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/51318404367648b888bee76ff3005c1f : SUCCESS in 12m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e0e65909e9544d6824b732d45a62f72 : SUCCESS in 52m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/023211986ad14d74a82aef7bb81d6eef : SUCCESS in 1h 19m 49s","accounts_in_message":[],"_revision_number":6},{"id":"052c2f7ceb80239e366fbf9affe91bd80b0f80e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-25 10:09:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"cf4c81dc9fcab1e31bb29208772e42fd2811cc5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 10:26:11.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e47e4ea8b8ff4eaf8181b4f7e2977f3a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e2b40ab788584493ad50948566203da1 : FAILURE in 13m 57s","accounts_in_message":[],"_revision_number":7},{"id":"2c3710bb81396f9e5e305701b4c0653dcd30605d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-25 12:03:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26d3c4619f3647d1a165451624d9ed65\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3ffdd11ff0d415b8f0a7a38263dd67b : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27ace09436ea4ccf91f336e55ae6b662 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d827e9daf6964dab9c5071ce0c151996 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b6ccd3bdab8e4076960b05cc25894788 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3263c7afaafb4e3db47c501234a904f4 : SUCCESS in 3m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/665b2626edac457f9784823a57374f5c : SUCCESS in 8m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cb8ccde7ab62412aa6e29af5924ddd0e : SUCCESS in 9m 10s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4f407e67bf5d4c63a7638522c79f8020 : SUCCESS in 4m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/da3b8ef8b9ea4a1fb44716f81ff4c515 : SUCCESS in 11m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/42b8ba78ad7e46308f5f61ab87f13ad5 : SUCCESS in 4m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5aa43ad2bb5f4eb489e549a733811728 : SUCCESS in 4m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6b86ae9c07f24de1a94f1ffd0511bb64 : SUCCESS in 12m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a2b06d31e92431291e597deed3a76a1 : SUCCESS in 7m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/92ef922c7c5e4a8e83b87df6fe5132c8 : SUCCESS in 8m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fb5941faa24741b3bf7b35023992cfb0 : SUCCESS in 13m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/00e563491d6a4571bed02ae6acc483b3 : SUCCESS in 3m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/de5cbb1d4d0947978a2a2eab8e77d270 : SUCCESS in 9m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4c5ac1fee84148b899d6a841d1fd127d : SUCCESS in 4m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/edef67797c514f0a8ca5c13f434669e2 : SUCCESS in 27m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a60d68a717c74e06ace1f3ab7ff14f48 : SUCCESS in 11m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a035bef5519449cea6cf26a795cc440e : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/00ba1da214e1444c8ea766543e9a5310 : SUCCESS in 3m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6b18b28690f408c82c1abb14eb44356 : SUCCESS in 22m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/af754507505a46588f894cba7ba8ead3 : SUCCESS in 5m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/01d4a868c71c47b6b3fcaf6cc9c0f984 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fd4b79b9e1f94db18ab8ee44867b0194 : SUCCESS in 8m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/483d1e87af2948cbae4da6bd97978bff : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86d8e71fb7a44efb9a91e1059a0f16d0 : SUCCESS in 53m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5621aea80ebb4aca823918fdc5c7c276 : SUCCESS in 1h 45m 41s","accounts_in_message":[],"_revision_number":7},{"id":"45498bc937cc61d73d92f547bcdceb3b0a099b5c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-08-25 12:50:20.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c255b43329bd4056866705f83a9a0470cb1e3347","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-25 13:25:55.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"2a0cbbf09b44cf9788c7a49131645e6e226eef06","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-25 14:14:48.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"dbc4aef346111e018965bb2b1a23bcc9df3758f2","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-25 14:59:42.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"89bd25a56d75e1f0e51aaa83266a697a82e38483","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-25 15:01:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"036a78bdb952e6e29a9ba481f7a2110fda87dd2b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-25 19:48:14.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9dfded88d9e17619aec5e1a4d32c81349719d1bc","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-26 12:58:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"56a50c736b8da377b0a3d5dbaa3e0e96c21309fa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-26 14:30:43.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"50ede879737a13057f02356ccf9c2088913ff08a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-26 14:30:52.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"90b732ee63449e743f4fe3594cafbc449b4a0c87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-26 14:55:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"79e530f4552859ca32a091a82befabe9ec0ef06f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 15:11:00.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1695b23f1f2e4d368b39185e437d9b34\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9fd6c6c047d0439a878cbb80eef57f96 : FAILURE in 11m 39s","accounts_in_message":[],"_revision_number":8},{"id":"ab0bf12b2fc16e853e5f14d260b67edfd47aa830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:48:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/911b10f7add1406baa698931bbac3857\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b825d3411bf4db4865de5d2d543b7c4 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5555f2a1c86049cba43593f1d1229417 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4aaa43749d414ab8be36753e728ceb8b : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d7d364fa7c347ec9e0fd8e7aaa29481 : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79c9468cce6845828b1e0e34629d61de : SUCCESS in 3m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f2c9166e656c45469bf732a58f911e71 : SUCCESS in 16m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7b626b4b14dd4e8ca1e2015a1bf791f0 : SUCCESS in 14m 14s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6516c64ad3624ae0a0c815a42f822b69 : SUCCESS in 5m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9d22ed7c57d0473eb9c24864abb4b91d : SUCCESS in 11m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bf8af02b275f49d7a9949861b46f0640 : SUCCESS in 5m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6e1eb5507f0d40deb7568152716e2c46 : SUCCESS in 5m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/317c2d09ad1a47c38046417f659621b3 : SUCCESS in 11m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b5695d0c611e4bf9be142b7c6e38b613 : SUCCESS in 11m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0bb32c517bfc493586850243f6372558 : SUCCESS in 8m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6d3ff2dd1079401bb8e56418ee9d039e : SUCCESS in 12m 04s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/445ffa6dbb3147acacdace8bd11a669f : SUCCESS in 4m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4473fb5177a0491c97155cd062afd022 : SUCCESS in 10m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e2c02aded84e4387ae3f00d8d50a3896 : SUCCESS in 5m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4fd9a76aee89418ea7e67373cfe2cf58 : SUCCESS in 30m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4b0b37a4b3474532bf386db78ec7db26 : SUCCESS in 11m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/410e7da4393e4fbcbe7b31d136428ebb : SUCCESS in 8m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e7133d9b97a74888860f6500ed4c7ee2 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e3915a38e514ce0a41047328dd3e5ca : SUCCESS in 26m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/84993d08f792449984b52b425500934c : SUCCESS in 6m 54s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6826e8ad83f14e40afc84cc2112a02ad : SUCCESS in 7m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc5e65fcf73347d4abd5e489dd657cb4 : SUCCESS in 7m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0eff2f67828c4e91a93cbfab6da41e13 : TIMED_OUT in 41m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2673a9cca82a4713bf5a2b2737b13a63 : SUCCESS in 46m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c1f7e0dafce4f3b98816a47a8821a12 : SUCCESS in 1h 49m 37s","accounts_in_message":[],"_revision_number":8},{"id":"d5e9b00197425179504944eb7bbc644e2d4d282f","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-26 18:24:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f3567aec8a297a301e239ae0c16cd625202fde36","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-26 19:00:53.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2e5d3c8ae99baba0f1f961ad83296d5e66351199","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:05:48.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a09f52c156d42bc934bdf9ab071af0a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7ab50a5c936f4afcb31ade903d6733d9 : FAILURE in 21m 29s","accounts_in_message":[],"_revision_number":8},{"id":"960e9c8e309a260b0fda24087c633d268788aec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 20:48:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6622b163e33e418f9d2fb2030fb5082c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09edb4f42f784bcc9337a672fcdefe73 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dae0da3a6685407dbb86f126d9d2411a : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4d40646cd89463aabc6c6e98b08d581 : SUCCESS in 7m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e7c577f43725408fadc330637e38c3cd : SUCCESS in 6m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/230821fc195745be9805fad9c890e3fa : SUCCESS in 4m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bc1c0ade67c943f88f7254ee9e8d93f6 : SUCCESS in 10m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/31a1994a353f470d96233906335892a3 : SUCCESS in 9m 14s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e3e032491f1f4998aaaa49c675ebd648 : SUCCESS in 4m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/947fb7c120354832a0adc43443159fe2 : SUCCESS in 11m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/47034e0c997c487182130f249b18b8e7 : SUCCESS in 6m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ddcf0a4468074c1dad1bf91a86e2e2e6 : SUCCESS in 6m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/05c6371156e343e3a556eae163657ed8 : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08488a52e326441f9e3e0924e0c97141 : SUCCESS in 9m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/badbd1f272274fd780c4b3680cbb918b : SUCCESS in 9m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/baa9c55f993d4742b8ab3d28314c49c2 : SUCCESS in 12m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f68ef45d5f684b4fb85dd02dcc25e912 : SUCCESS in 5m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7c09a2b1a51245bba1c36e81aad668e6 : SUCCESS in 12m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cc6b02266bc84aec9035c6f672a441f7 : SUCCESS in 6m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dd87de9c69fe400bbc8ba70b1b6ab49a : SUCCESS in 23m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f5bd5d6715c248548eabf143c848344c : SUCCESS in 13m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4ac9a2b0f1d640eab6621094d2e89acc : SUCCESS in 7m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b2652c6f966b46c98251ad2efada9d08 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/38e83950a4a9436984a7530c1a1a7937 : SUCCESS in 19m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0dd9fbe43877491ab6f311fef9d9863b : SUCCESS in 9m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3e4374b1364f49a2be8c11d3c3306f0a : SUCCESS in 8m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eab1eb5645bc463a926e9ecef9c0c7cc : SUCCESS in 10m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8c7c5c22d8474f79aa38fdf86ee6ef0f : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc9964fa28e5449ebc2ed214ccc2eb6e : SUCCESS in 59m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2456ed12847a4db28adc45e72d27432f : SUCCESS in 1h 33m 11s","accounts_in_message":[],"_revision_number":8},{"id":"95be176a0df5bd4fc54a09ae23ad575ab729dc0d","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-29 09:31:26.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d19839afd68b87d2a9d194f0f0d181a8d22b1ba5","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-29 10:17:34.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"732f55b89f7912fc04a36a8a4788099be630c54a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-08-29 17:59:37.000000000","message":"Abandoned\n\nProposed to master instead: https://review.opendev.org/c/openstack/requirements/+/854038","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"c2bbb86f2b133dfa83a50e42426d1c87b5839874","revisions":{"828ee98d1a093d49c418dfe66c40b32c5ef44db6":{"kind":"REWORK","_number":1,"created":"2022-08-23 09:26:18.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/30/854130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/854130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/854130/1"}}},"commit":{"parents":[{"commit":"52cd6721466b6af58f62ae8a16512b566db887dd","subject":"update constraint for python-blazarclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6721466b6af58f62ae8a16512b566db887dd"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-23 09:26:18.000000000","tz":0},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nSupport for python3.6 in jsonschema was dropped in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/828ee98d1a093d49c418dfe66c40b32c5ef44db6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/828ee98d1a093d49c418dfe66c40b32c5ef44db6"}]},"branch":"refs/heads/stable/yoga"},"61acd17502634d714830c16e3e4bf2c4f7e15f56":{"kind":"REWORK","_number":2,"created":"2022-08-23 09:28:32.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/30/854130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/854130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/854130/2"}}},"commit":{"parents":[{"commit":"52cd6721466b6af58f62ae8a16512b566db887dd","subject":"update constraint for python-blazarclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6721466b6af58f62ae8a16512b566db887dd"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-23 09:27:49.000000000","tz":0},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nSupport for python3.6 in jsonschema was dropped in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61acd17502634d714830c16e3e4bf2c4f7e15f56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/61acd17502634d714830c16e3e4bf2c4f7e15f56"}]},"branch":"refs/heads/stable/yoga"},"f4835483da12c71f7bfab4b6fc2658ec8a5da53d":{"kind":"REWORK","_number":3,"created":"2022-08-23 10:24:12.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/30/854130/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/854130/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/854130/3"}}},"commit":{"parents":[{"commit":"52cd6721466b6af58f62ae8a16512b566db887dd","subject":"update constraint for python-blazarclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6721466b6af58f62ae8a16512b566db887dd"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-23 10:24:12.000000000","tz":0},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nSupport for python3.6 in jsonschema was dropped in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4835483da12c71f7bfab4b6fc2658ec8a5da53d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f4835483da12c71f7bfab4b6fc2658ec8a5da53d"}]},"branch":"refs/heads/stable/yoga"},"22e853dedd78b2b97da664ca12e1491ece1360ce":{"kind":"REWORK","_number":4,"created":"2022-08-23 10:53:54.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/30/854130/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/854130/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/854130/4"}}},"commit":{"parents":[{"commit":"52cd6721466b6af58f62ae8a16512b566db887dd","subject":"update constraint for python-blazarclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6721466b6af58f62ae8a16512b566db887dd"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-23 10:53:54.000000000","tz":0},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nSupport for python3.6 in jsonschema was dropped in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22e853dedd78b2b97da664ca12e1491ece1360ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22e853dedd78b2b97da664ca12e1491ece1360ce"}]},"branch":"refs/heads/stable/yoga"},"94f10e2fd9859c7adfba0d795e9abf50fd505068":{"kind":"REWORK","_number":5,"created":"2022-08-23 11:19:40.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/30/854130/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/854130/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/854130/5"}}},"commit":{"parents":[{"commit":"52cd6721466b6af58f62ae8a16512b566db887dd","subject":"update constraint for python-blazarclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6721466b6af58f62ae8a16512b566db887dd"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-23 11:19:40.000000000","tz":0},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nSupport for python3.6 in jsonschema was dropped in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94f10e2fd9859c7adfba0d795e9abf50fd505068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94f10e2fd9859c7adfba0d795e9abf50fd505068"}]},"branch":"refs/heads/stable/yoga"},"633f829afc7c5799f8359eb9e168f0951ac4944d":{"kind":"REWORK","_number":6,"created":"2022-08-24 17:09:29.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/30/854130/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/854130/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/854130/6"}}},"commit":{"parents":[{"commit":"52cd6721466b6af58f62ae8a16512b566db887dd","subject":"update constraint for python-blazarclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6721466b6af58f62ae8a16512b566db887dd"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-24 17:09:18.000000000","tz":60},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nSupport for python3.6 in jsonschema was dropped in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/633f829afc7c5799f8359eb9e168f0951ac4944d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/633f829afc7c5799f8359eb9e168f0951ac4944d"}]},"branch":"refs/heads/stable/yoga"},"3149940054a8fa2d765b053b62fe3c8f0b52c117":{"kind":"REWORK","_number":7,"created":"2022-08-25 10:09:14.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/30/854130/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/854130/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/854130/7"}}},"commit":{"parents":[{"commit":"52cd6721466b6af58f62ae8a16512b566db887dd","subject":"update constraint for python-blazarclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6721466b6af58f62ae8a16512b566db887dd"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-25 10:08:47.000000000","tz":60},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nSupport for python3.6 in jsonschema was dropped in v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3149940054a8fa2d765b053b62fe3c8f0b52c117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3149940054a8fa2d765b053b62fe3c8f0b52c117"}]},"branch":"refs/heads/stable/yoga"},"c2bbb86f2b133dfa83a50e42426d1c87b5839874":{"kind":"REWORK","_number":8,"created":"2022-08-26 14:55:07.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/30/854130/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/30/854130/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/30/854130/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/30/854130/8"}}},"commit":{"parents":[{"commit":"52cd6721466b6af58f62ae8a16512b566db887dd","subject":"update constraint for python-blazarclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/52cd6721466b6af58f62ae8a16512b566db887dd"}]}],"author":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-22 18:27:12.000000000","tz":60},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-08-26 14:54:13.000000000","tz":60},"subject":"Update requirements for python36","message":"Update requirements for python36\n\nSupport for python3.6 in cliff was dropped in v4.0.0, released August 12, 2022.\nWarlock requires jsonschema \u003e v4.\nSince older versions of OpenStack still require py36 support, we need to add a constraint for py36.\n\nCurrently, this is blocking telemetry-dvsm-integration-centos-8s jobs on\nyoga [1] [2]\n\n[1] https://review.opendev.org/c/openstack/ceilometer/+/852878\n(json_schema)\n[2]\nhttps://zuul.opendev.org/t/openstack/build/618b46aea2ad4fd49b481263e28130b7\n(cliff)\n\nChange-Id: I25725a58c6c4b2198f6b83c43a52bcd33a42a8a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2bbb86f2b133dfa83a50e42426d1c87b5839874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c2bbb86f2b133dfa83a50e42426d1c87b5839874"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
