)]}'
{"id":"openstack%2Frequirements~854388","triplet_id":"openstack%2Frequirements~master~Ie0a73cbd3778239de631b46e6027ebb804fdd75d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie0a73cbd3778239de631b46e6027ebb804fdd75d","subject":"update constraint for oslo.versionedobjects to new release 3.0.1","status":"MERGED","created":"2022-08-24 13:32:40.000000000","updated":"2022-08-24 19:07:00.000000000","submitted":"2022-08-24 19:06:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854388-new-release","meta_rev_id":"cff90cc459ab15e07ac7e2b25f5659bab10f82a8","_number":854388,"virtual_id_number":854388,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 19:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-24 16:11:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-24 16:11:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-24 14:27:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 15:50:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 16:11:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"69149b09ab29a852d3826c4a54940236a851662b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-24 13:32:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3c1b45aab90c1f0ac2404a96e47b72a55ca2f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 14:27:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/807e0f0fb7a04863a58120bf54939513\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aecb561396e0409bbba8e082703a9cab : FAILURE in 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"e7673a00e0f99d1e9c7768077b39453299a34c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 15:50:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b078e0ce141842dc96d1c7e898869052\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19252d27b6c8417a90aacf59d5f23a48 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833f994e2ea94243bd743a6bf0b12731 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/481bcb364155415c9600cd15cbe3a855 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d41c98a18fa4da0874076a42eb67907 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/378162f08eed492c9d76ecf1cf7aa98f : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/297eb92e72e645b28c779500473e693e : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c2a1a75557194ea488b6bec75eb5d9f8 : SUCCESS in 5m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/911cd570d94240358fac26b692e1b449 : SUCCESS in 8m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ee58575d5d84463fa180edfe6d3c8d93 : SUCCESS in 13m 37s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/4cd178934b1c4f35a5edb09cdefb4693 : SUCCESS in 10m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8e5d5796daa04a289710a8eb32da7a87 : SUCCESS in 5m 14s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/eeb01c73902346f5b6dc8bcd647e5ffa : SUCCESS in 6m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/11c1a6646f7945dbb2e5950f0ce1f4b9 : SUCCESS in 12m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b37fe5225a8c4b6980d10cfe132701c6 : SUCCESS in 5m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/24ea4a42a9b3460c88ac2458d016f5d6 : SUCCESS in 6m 06s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f225bf80fddf48bea005193d57986c0e : SUCCESS in 9m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3bb99e58b7a64214a8a3bb27399beb7b : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/804e133b704c48e58af8699ee1be32a2 : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f48c5392f0a1448dba60ff8258a072c4 : SUCCESS in 8m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/72968500a13e433dbf04bad51846e3bd : SUCCESS in 13m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/13bd6d3d07f5459c9ccd7f30805cdf7b : SUCCESS in 4m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/792d777be68f4693bd623c2698794ed3 : SUCCESS in 9m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a95839ce3b404963b1c1036cc75051ab : SUCCESS in 5m 49s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/ee1fef9667ff42ab91731f0dbe97e9fd : SUCCESS in 10m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/58a2cf80d80e48bfb3aea117274d6711 : SUCCESS in 29m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/429986a752124060bd37c0db8596b6e3 : SUCCESS in 13m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/abb5fa24e2e140668e08f8fae6b6c72a : SUCCESS in 5m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a4dc9ae0ec94178820ebd6db06a7ec8 : SUCCESS in 3m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/889f9bf86ec44ab5882e4553f7bb8f2e : SUCCESS in 20m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2c5d47a8e9b44067a19ea78c75a98a6f : SUCCESS in 7m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1d6d7da2df4746a3905b1c8e8eef0506 : SUCCESS in 4m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/44265f38b8fc4e37be822769cbe2afa3 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0e7af8e41e964862b64a277c62a682f9 : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/545cadd8aff34901900e36355bef6a64 : SUCCESS in 52m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3ca5e72da2f444aab45b03fa286239c2 : SUCCESS in 2h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"282bb9f5807b05e325ad43be4070cab3ee332699","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-24 16:11:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4602a5ef948a294381e9bf2d3d5d0032ddc2e5bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:13:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f692b520681d8db7eca60ffaa205c741ead2cfdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:06:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eba01b67c5b41debfa8a9ee9615689e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bd62ab048fe4004b98cc3c6aff87663 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7615019f25464436b5a96aea710e23c8 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c5228c8919e4cf197b780d773d80cab : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73d1e9d68b184387998ed54f8ada0ea2 : SUCCESS in 9m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6d3638be1c6f485481afb081430104e6 : SUCCESS in 8m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c0d2c5510ac541af8d080f4641ccbff0 : SUCCESS in 7m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb09fe5de1004186b434efadec777a5d : SUCCESS in 11m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1c7d424607e74add814604a8c7c91b44 : SUCCESS in 15m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9fce4421ba304c82af0cb5de310337c5 : SUCCESS in 14m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a02eecef8fc8470fb654b4b675ec3dc8 : SUCCESS in 9m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c08bc24f49474d088becb00c1ed2853f : SUCCESS in 9m 18s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/db5c6c0e563c4759ba6623cfbcf437e9 : SUCCESS in 12m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0ef4cd0ef9d64da79c94746961aff7ac : SUCCESS in 15m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1724915503d94bf48dbd3b3f38caf77d : SUCCESS in 13m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/704bc8ddaa2a48efa195d54edce49ee7 : SUCCESS in 12m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e94fcaa2274847449b96946a7170129b : SUCCESS in 15m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/81c55dcdc9904739b23a3e9d85ffe0b8 : SUCCESS in 9m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0aba48a983424f44b684a9dc153e8ee7 : SUCCESS in 14m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5950a324af644d0799e46ba461334378 : SUCCESS in 9m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/394b4136389546a3a8c58960808bc52a : SUCCESS in 29m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/de4339c2beaa4d4c817b5ce4f0e3cd6f : SUCCESS in 13m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7f2176d28ee54081bfc9d51ca87e8505 : SUCCESS in 9m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/41f8aaa1ebfb4bff83fe874145fd7dd0 : SUCCESS in 5m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/27a87a1b2e464549a98a4549004e2f53 : SUCCESS in 24m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/804bf50f05bd48beb3de3f14f73064bf : SUCCESS in 10m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/12219432a5b34ef8b9afbc9181f0910e : SUCCESS in 9m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76a88821c49940319de76c711f58f6b8 : SUCCESS in 10m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2d1ccbc8613043b7b19a94201bfe3811 : SUCCESS in 15m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d83a954e79e430a8eb43f8569ffc449 : SUCCESS in 1h 05m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/733e62c85c884ddd843346e390692af3 : SUCCESS in 1h 48m 42s","accounts_in_message":[],"_revision_number":1},{"id":"28ba00c5d608c7307bd2eb1db8c9fa4b5ea25b4b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:06:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cff90cc459ab15e07ac7e2b25f5659bab10f82a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:07:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeb77b1e88b248009c7ee61472be35ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4a8482f6a54f7aabd6d80c08205ba7 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bca2ba9d3c717bb01b6baa67c1e81bb852013e9","revisions":{"6bca2ba9d3c717bb01b6baa67c1e81bb852013e9":{"kind":"REWORK","_number":1,"created":"2022-08-24 13:32:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/854388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/854388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/854388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/854388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/854388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/854388/1"}}},"commit":{"parents":[{"commit":"f9677e5fca12fa9ceaf239fa841d578d03db5e5b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9677e5fca12fa9ceaf239fa841d578d03db5e5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-24 13:32:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-24 13:32:39.000000000","tz":0},"subject":"update constraint for oslo.versionedobjects to new release 3.0.1","message":"update constraint for oslo.versionedobjects to new release 3.0.1\n\nmeta: version: 3.0.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib2c0e782b9a8722eae6fbd6c8046247e37f7a914\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ie0a73cbd3778239de631b46e6027ebb804fdd75d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bca2ba9d3c717bb01b6baa67c1e81bb852013e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6bca2ba9d3c717bb01b6baa67c1e81bb852013e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
