)]}'
{"id":"openstack%2Frequirements~854411","triplet_id":"openstack%2Frequirements~master~Id06a7eb104ce5946d000c4c08041ee24fc136c89","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id06a7eb104ce5946d000c4c08041ee24fc136c89","subject":"update constraint for oslo.messaging to new release 14.0.0","status":"MERGED","created":"2022-08-24 13:33:06.000000000","updated":"2022-08-24 18:32:58.000000000","submitted":"2022-08-24 18:31:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854411-new-release","meta_rev_id":"738c382af571417fc5d807d8a4a4a387192cb7fb","_number":854411,"virtual_id_number":854411,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 18:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-24 16:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-24 16:11:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-24 14:53:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 15:21:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 16:11:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c72d3bf00a6e77657385f6186ffc42f533b99f3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-24 13:33:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"762bd22f2c65185abd4690b8dd4888b1c5894338","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 14:53:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b94d1d92e6ae4da793b6a8cd722c0954\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/301fc9d27db04c0cb0ad0273032bd119 : FAILURE in 1h 05m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4f7d387327438ceb9ddb2a72259555a1f0973df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 15:21:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8751874496b42379fa04a55d4072a64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3b56d258f64ae48f1921420298600d : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/259ada06d95c4944963171d1f49822b0 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdb3060fe5c744c1abc3595f53a11347 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42a6c3ea74e6432797c480d8a6187607 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21920db7472a45f08c0e1b51085d02e6 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34ed0d65506e40f89b484e634fa97bd3 : SUCCESS in 6m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a769f9ab4ca44319290aa99b9046140 : SUCCESS in 4m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/59a57447c8dc4ea295abc14b345b6b9e : SUCCESS in 7m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2942f5e4745248b89862adba1bb911f5 : SUCCESS in 12m 49s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/248790693ce542a0a8436b9089a7bc54 : SUCCESS in 8m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fec41914c6134c9aa5d1707188630492 : SUCCESS in 4m 41s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2fd751161a224b5caa3cc31d116efbb0 : SUCCESS in 6m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fafbefd630a0423c9497528b910172ef : SUCCESS in 10m 51s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fc4ac1a64f5e44aeb6c8184b98348eec : SUCCESS in 3m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7acc2a8a6f444a028241d78e974709a4 : SUCCESS in 7m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/cddf34a32f924e66926a050ab9addbf3 : SUCCESS in 8m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5b750bd9ffa5441abba552dd305157cd : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e64861efc7b3457992f3829abcdf564b : SUCCESS in 7m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f6be62f2cebc482582497c28a723921e : SUCCESS in 8m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b31d9f09311b4cfca374e52da4ce94e8 : SUCCESS in 10m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0c03e4e02292470c8d768b356294159d : SUCCESS in 5m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/95c4731b82e447899910c3e5814cafff : SUCCESS in 12m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8fc6443704af4285aa800a20cf8efd79 : SUCCESS in 5m 57s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/a5eca1f24d2f48cc9cc42a9054e1b73e : SUCCESS in 9m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9c3b88022c6648889e5435ba79851630 : SUCCESS in 25m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c3f4af0948484fb6a496e98048735120 : SUCCESS in 12m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a0966baf0b024e549e7ad051e6555c44 : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/164ac0eff36c4172bf76209fa2bbd334 : SUCCESS in 5m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8c1991634a234e8cb6459046e00ef75c : SUCCESS in 20m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/de03028430fc47bdae72e93163b36f22 : SUCCESS in 6m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2d91e3293c4b406289a30ca2dd715c68 : SUCCESS in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/09ec297a972e4da08967708d140e1c9a : SUCCESS in 8m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f6c6bf38effd41958479a3d864bef021 : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a645493c1b204f5090d2b47d77b3fb52 : SUCCESS in 1h 00m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c70dd92c351459495e44072b9fe0f2f : SUCCESS in 1h 39m 19s","accounts_in_message":[],"_revision_number":1},{"id":"c1f64d97848a8ac0c365c872fc09dec956b71533","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-24 16:11:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36c024623b4eb9c149ff6d30ee3864d1f30eb458","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e04a241a337bdc424d644bb74719b01de49553c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 18:31:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb71ab688c647be95f71aef9ba28cba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b95ccd23ea4e9495d7dbd5a1b3b10f : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f42700be5a84d7b8eae883a0548e9fe : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e7aa8c4307546e2b0030067690dc426 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e45f8188374b46f5a753b226a3d91df4 : SUCCESS in 6m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c6656610ac447adb80c555dd2422c90 : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4f3f4c68a5b4bd1b08736f2dcfb0833 : SUCCESS in 4m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4d87d5e17e814d82aed9a2876e5d74e8 : SUCCESS in 10m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/92fd3d18942b430f807a093562761059 : SUCCESS in 14m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8dc65439a4f54ef0bf1ded4125ce3ebb : SUCCESS in 13m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/788c9aa7443e47fd88e0861e00412553 : SUCCESS in 4m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7d0d8a629d6a42ff9c32419a4eacff55 : SUCCESS in 7m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/47f66783466b4de0934f28157cfa0a8f : SUCCESS in 8m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5778d8e6c03841a4961ff03acad433a7 : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4cac44545e2e42dfa4408772fd1bf584 : SUCCESS in 8m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7ae227318a5f4b0fb9a4d528abec91c5 : SUCCESS in 9m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/296991597e434ec4aa1afcabf7b3cdd1 : SUCCESS in 12m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e403d6522ac044c8a39274f237daa683 : SUCCESS in 6m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6020be4e46ed4d40ba2cf7c900026c97 : SUCCESS in 10m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/12f13b05328348e0b2ebad4ea9639070 : SUCCESS in 6m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e3d447fbe2fa424daa560a2329159004 : SUCCESS in 26m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6b95e0bfd1b145f6917b6c4fefaa7603 : SUCCESS in 16m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e8168ee6ebdd45ecbb42549b86c5569e : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1af82159966a43b6a78eb243644d344f : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/be81987b05234c0580ad6b36ebff58b5 : SUCCESS in 21m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/082ee1920ff24f33a9cc04fa7b9396c9 : SUCCESS in 7m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5439fcf6abe1441f8cd69a00d3e8b2cb : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/02034b9554c14d36a5c91c2a23d83f49 : SUCCESS in 6m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/feb803ed4a9348b0a43ae0fdfa6281f2 : SUCCESS in 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b285bdfe46864be785f107fcf40fcc0d : SUCCESS in 48m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12a1d4aca104422da3778a18f1155522 : SUCCESS in 1h 47m 25s","accounts_in_message":[],"_revision_number":1},{"id":"70e22c10ac0570cbaf8082398c2f6089aa658551","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 18:31:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"738c382af571417fc5d807d8a4a4a387192cb7fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 18:32:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a01e913a380341bc986ae18ec2d68bb0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e90e8909b44abb8aecfbd582119793 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"21d8c03c4ad3b80cbea256b57656ffb4c8b9cfd9","revisions":{"21d8c03c4ad3b80cbea256b57656ffb4c8b9cfd9":{"kind":"REWORK","_number":1,"created":"2022-08-24 13:33:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/11/854411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/11/854411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/854411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/854411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/11/854411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/11/854411/1"}}},"commit":{"parents":[{"commit":"f9677e5fca12fa9ceaf239fa841d578d03db5e5b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9677e5fca12fa9ceaf239fa841d578d03db5e5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-24 13:33:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-24 13:33:05.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 14.0.0","message":"update constraint for oslo.messaging to new release 14.0.0\n\nmeta: version: 14.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5756f9bbef9a8b728794bfdd44299579c07dce3f\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Id06a7eb104ce5946d000c4c08041ee24fc136c89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21d8c03c4ad3b80cbea256b57656ffb4c8b9cfd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/21d8c03c4ad3b80cbea256b57656ffb4c8b9cfd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
