)]}'
{"id":"openstack%2Frequirements~854416","triplet_id":"openstack%2Frequirements~master~I89aee627bac3b82188e55a725ed540a46de3fc70","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I89aee627bac3b82188e55a725ed540a46de3fc70","subject":"update constraint for oslo.privsep to new release 3.0.1","status":"MERGED","created":"2022-08-24 13:38:50.000000000","updated":"2022-08-24 19:07:40.000000000","submitted":"2022-08-24 19:06:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854416-new-release","meta_rev_id":"f164ab2eee1d6cca0704feb6667aa00e964b7b73","_number":854416,"virtual_id_number":854416,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-24 19:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-24 16:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-24 16:11:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-24 14:57:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-24 15:18:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 16:11:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"16ac6aa1e8c96e901cf05066fc83bf55c5ea3fd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-24 13:38:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"751527cf436e0904a961b77a3290ba15ff10cd3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 14:57:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5123dd5c5cf9494fb63676265d7e229e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c5869dcb69c945e28084d6020b3e5967 : FAILURE in 1h 08m 36s","accounts_in_message":[],"_revision_number":1},{"id":"f6af8fee2c9ef68be638630d3b4fa147379ab1f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 15:18:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04e5c18c3ce04f4091934c1ba0564725\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583609bc63ec4e2483081947a3e1488a : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d46c045bab84a81aed24f7243eef920 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07c20b8f68a2455a81c794be523cb0cf : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa114cc743c149ffb4b281513ee9c795 : SUCCESS in 8m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81aa6ca8ea8342438e4d316e81ea4b27 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f45bb9e61b2c4143b9c989985438b002 : SUCCESS in 5m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/227216aa0fde420882cb82de288b1e5c : SUCCESS in 5m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a6bcada3666467ba3b866986fa49a11 : SUCCESS in 10m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/51b9019fb41d472493d6abffdfc23a31 : SUCCESS in 13m 06s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/24e780403d30467d81d906d1fd6e962a : SUCCESS in 9m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a115c57c6e264859991338c05da631a9 : SUCCESS in 5m 52s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/7785899e6fa24b0fa396308ce4909721 : SUCCESS in 10m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5ea9543613364697b2898fe7d6fa1fc6 : SUCCESS in 13m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c00789abab4e47ee8ac92b90c61d8de2 : SUCCESS in 6m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8d3bce1b984642c18a56b598d8732993 : SUCCESS in 5m 03s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/98579ff7fd5f4a7d9ac7740aa6039cad : SUCCESS in 8m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6ddbeab544e148998f552351548cc28d : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fbfb347bd224b4483b28c4cbe4f31f4 : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3959eec6b4a34e34bec6eca113880431 : SUCCESS in 10m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d022a94ecee445cabad34a6126f7a21c : SUCCESS in 11m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6def52b5789e48a5a482591b682c4dd6 : SUCCESS in 4m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/63a243af1df649f3b267efa8bf929c21 : SUCCESS in 9m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cf7cc6ffddb74176b73aee5c44b26e42 : SUCCESS in 5m 56s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/f6a2f7b2548c481fa58d1dface3ecbe2 : SUCCESS in 10m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0d6cc73a50cd48be89525baaf7fbaff1 : SUCCESS in 28m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e69c8675c5c84e10b97153aeb99f408c : SUCCESS in 13m 35s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4084af003b874f0aa34ed544066689ee : SUCCESS in 6m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa56325f452d47c1a0e072321e239f04 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46cdd79262724e2cbadd7e8e5b34768b : SUCCESS in 19m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/744cbb47ed6c4626aa2654473bc43944 : SUCCESS in 7m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cb2fcbe02c3443aaa93a1a8d449cb9e5 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fcaea6ef9a24ad3947109dcceeeb676 : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2bf2e85fb979485fa32a8daa3ac2bbbc : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b8db99e2fb84932ae92e8ef36148235 : SUCCESS in 48m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a9dd8549df947ce99d17e91368c6608 : SUCCESS in 1h 28m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e09631d09b9a140c74650c918bb64a92bf4fc289","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-24 16:11:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c9d6e20b6a1c2a75663af7b9c4b35fa77f4ee0f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 16:13:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43269e6078d2ab5d3361aaca5f7da94c2784e689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:06:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/566ea4c7b7844273adbaea737cfed61e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e057a93b80d2454f8a5a5a02cfcbfd14 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1bef1385b5e402888a3cfccb66f0d34 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9133f1385e294dd9878d9f06afea1ba2 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71a6d1ed324a4fb7874f7c83c56b79fd : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/46d3602fc3334d9f92743c26afe9b78e : SUCCESS in 5m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a6bd2c5c717f4c4dbf9c27b6c773a693 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da6c5c3e90684c5080802a20be10e692 : SUCCESS in 8m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b8202633dfc5427d8b884a3c311bcfeb : SUCCESS in 15m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7226c83f3837451c9bd4e80623128ff6 : SUCCESS in 10m 02s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/621088fb4f324c5a82cb375c06302d70 : SUCCESS in 4m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f20e25c54d934b65b58069d71deb5776 : SUCCESS in 6m 50s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3248292c034a42bba958fd0ef48b7ff1 : SUCCESS in 9m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/476e3404f8bc4280bbb978997cf02ebd : SUCCESS in 10m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1a379c1afa1046ec86ae79ac8fe9de5a : SUCCESS in 10m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6a35c01ccf04774bb4572829971da82 : SUCCESS in 10m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/48cc822e1f48406fade060c5df09b3de : SUCCESS in 13m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4ddd744ad4f547efb9a278a10f3f52e0 : SUCCESS in 5m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0aad9fc9fe034379aa27844b53fd6377 : SUCCESS in 9m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/06497f4067ca4fa5866d0bf919490aea : SUCCESS in 7m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9effe8f6bae447f880269f8549f57990 : SUCCESS in 23m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5336bab16d0a4f46be769b5ffb2d8972 : SUCCESS in 13m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8606064869f04509b2488c7249c56b3f : SUCCESS in 9m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/016b960007994034b411cf7400d12cc7 : SUCCESS in 6m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a968e13ef4eb45a38261df5c65276a31 : SUCCESS in 23m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b5a3b0b3aab74b3f92659bc14accdd5a : SUCCESS in 6m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5e326cd079cd4d68bc457813ae0c8fee : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbd291e8784c422abd2c8feffab68d20 : SUCCESS in 8m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/78f28220d39141f28f5302dd316fc093 : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11a2671ed9f742dab628d9cd87bcb9d3 : SUCCESS in 51m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57fbc04455364e458c2428f7b5aeeca8 : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0f7c38b91f2a53e6ce6d1391000eb1aafc53e410","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:06:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f164ab2eee1d6cca0704feb6667aa00e964b7b73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:07:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3c175d1bd7a4a9587698786b640d88f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7f6fa7b22245f1a324ef8896b709d1 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9e63a464d57e7b0102269b50e4b1e75a6846483","revisions":{"a9e63a464d57e7b0102269b50e4b1e75a6846483":{"kind":"REWORK","_number":1,"created":"2022-08-24 13:38:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/854416/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/854416/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/854416/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/854416/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/854416/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/854416/1"}}},"commit":{"parents":[{"commit":"f9677e5fca12fa9ceaf239fa841d578d03db5e5b","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9677e5fca12fa9ceaf239fa841d578d03db5e5b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-24 13:38:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-24 13:38:48.000000000","tz":0},"subject":"update constraint for oslo.privsep to new release 3.0.1","message":"update constraint for oslo.privsep to new release 3.0.1\n\nmeta: version: 3.0.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I3a6733a07e2d9739f4379f961b1afdc0ebfdb9f3\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I89aee627bac3b82188e55a725ed540a46de3fc70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9e63a464d57e7b0102269b50e4b1e75a6846483"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a9e63a464d57e7b0102269b50e4b1e75a6846483"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
