)]}'
{"id":"openstack%2Frequirements~854826","triplet_id":"openstack%2Frequirements~master~I3f83314cc607ff238d64b7fe47301507322be295","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3f83314cc607ff238d64b7fe47301507322be295","subject":"update constraint for sushy to new release 4.3.0","status":"MERGED","created":"2022-08-26 18:14:04.000000000","updated":"2022-08-26 22:44:00.000000000","submitted":"2022-08-26 22:43:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854826-new-release","meta_rev_id":"273686159612c97c050a1c2021127edf5da4b8a3","_number":854826,"virtual_id_number":854826,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 22:43:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-26 21:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-26 21:04:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 18:53:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-26 20:47:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 21:04:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9bcf56c6aab2119af84232a96adc4d33c04cc142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-26 18:14:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85f16b048609715a843e17369815d91faa35c1bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 18:53:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93a8c35aaf6406a9d5d8737293787fb\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bcb5ed8d68c94908aa7f58cf4f9e95d7 : FAILURE in 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ed9f234649587fdacb2e14adc1f9be7c61348c89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 20:47:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43b1b3d4009440b48b20aefb448cead9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a5965ab6f9a448ab2be439e69726125 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e2dfc6852b7427792fe195296f6c337 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a510ab7c5cec4acca70b3f1528545a7d : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f54eb78ac2904efda3e27b777234ac8b : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f0a10c9b7e54334be48bf5612ea6b1f : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e76c7ccf1734614a724049b238ba5df : SUCCESS in 5m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb74fc4326e440e4baf5acb6ec6c4379 : SUCCESS in 4m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5aa2752f5bc34a2bb4bd2cb54d9f692f : SUCCESS in 9m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5cdb5f0b3c504f02ad366137fd204ebd : SUCCESS in 12m 08s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/8045bb3f70224b96802fec4c04b5c117 : SUCCESS in 6m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/43ccab21671d4f369475177f70748bba : SUCCESS in 5m 13s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/b48d8735a76d4560b7dfa94468fb4d31 : SUCCESS in 7m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/244904a3bccb4ac2a3121e61799dac42 : SUCCESS in 11m 43s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/772e283d9a59467e9a4469efbd288f27 : SUCCESS in 6m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/555249d625e8489c971b4f980c3a6152 : SUCCESS in 9m 23s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f393a424281e44ffbb52316f2a82ed28 : SUCCESS in 8m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1308def0b36648c8bd7e86a5b6f06d60 : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a9ff4f9365e940cbb9f24d76931d7017 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/477ad0e3d0bc4b76b005baa19525758a : SUCCESS in 8m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2786faf0e76e45bf8108068bccdbc6a3 : SUCCESS in 13m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/54ae96aa4d724853a710688b236c2f66 : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f35295fb9fcd469cb56870979a304871 : SUCCESS in 10m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f27d2e760334ccda3695d1e7fcdc13c : SUCCESS in 5m 36s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/069829e090544a6c88271261f043f81e : SUCCESS in 9m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/04e25aaa6cde400ba84c7399589c7df4 : SUCCESS in 23m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8dcfacce90b240b2940aee26de0460ef : SUCCESS in 13m 46s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/50479fd66a784a69891402430c5ff1f1 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/782c0401375b4cbca88834d1f8e6f47a : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a552ec2e069a463e97b8922fd305a605 : SUCCESS in 24m 33s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d5db9a4cb35543c4ba01a94c224270b6 : SUCCESS in 7m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6f61d759fc394f0e94f91fb70ee156cd : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c51d2cd6552e4cdb848eda9248e94410 : SUCCESS in 7m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cacd930a8a9d448588525b8063f430f9 : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae758454690645d180abd14f405a3a05 : SUCCESS in 43m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40c1403a335343a9b2567eea584eb8a7 : SUCCESS in 1h 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"78bdc2d879cb212ebcdc11236be0283e43d95259","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-26 21:04:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9a6d8c5b8402ada20e8c9f9572dd3789e327718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 21:05:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10e85c6cbac2cb1521d84296814c81a13b37d62d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:43:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8af560d6ea5041c69dd75c21bcfee163\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd95e42279de4b6ebb165fb9c3d5ac67 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f426808cd4d4f75b3a96c29b720621b : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb4dd1845ade420d911c97199404ab2a : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43c90ba25eae496688123ccee077b42b : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3bfb4162dbe841bca0c2e4a4acb25906 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6bc7902dc9434648a1269a98caa018c4 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ebc04f49d5b467b91da4a686b2682d7 : SUCCESS in 8m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7524e108876944818c9a3e242af2b801 : SUCCESS in 12m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/405c93e562694c1681e57b62f58a4612 : SUCCESS in 11m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7c7903b2ea3f4cf98cf49b572a57fd16 : SUCCESS in 5m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/00968192c5c74176a5cafb70723f2f3b : SUCCESS in 6m 31s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c747a57550824a3398d0f28ea82a7e62 : SUCCESS in 8m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b950b406ca444d06ba39be3f6e7c3997 : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a544baae834148f2a644e4d58762ee3e : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/141b6fcc4267412b8814a5dc146eeb42 : SUCCESS in 8m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/75829b727b544f26ba898ae0e963fbf2 : SUCCESS in 15m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/418b7dfb497f44f2987ba612654323ff : SUCCESS in 5m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e0b7743bc0f74a078d0cd149a7e4b290 : SUCCESS in 10m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7a82746eeefa4c458740bd21bf112394 : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/290c69139e6d4dd1af94fbe1834549b6 : SUCCESS in 26m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2c4f99d3c1f34831815d5ad0f69a1b79 : SUCCESS in 13m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/264eb991873441f69274fdf400858cac : SUCCESS in 5m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54f99ed034e7456d89420687df5f0f6c : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9de53d3359a546d4966a29de34ae35d4 : SUCCESS in 19m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2cf022154915467492443215d468a52e : SUCCESS in 7m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cf5660d1bac24927b975028dc8113ab1 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cba6c2571f0c45ffa5bcb20e41880784 : SUCCESS in 8m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6677af89fee94254872a16a88bc3366f : SUCCESS in 9m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ace3ef6c1b64866ae37afe89d040cff : SUCCESS in 55m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3f30740bc944c0e8a3c20b88e2eb896 : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5dc6287d8ab4cb14ca862ad9d6adca49b711f769","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:43:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"273686159612c97c050a1c2021127edf5da4b8a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:44:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77a2b9de72144b12a6755f9fa128dd1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2ef2b1cd954c91a2dd0622d621100d : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0f25795fee9e0269cbe27766d33a76afe2e183f","revisions":{"d0f25795fee9e0269cbe27766d33a76afe2e183f":{"kind":"REWORK","_number":1,"created":"2022-08-26 18:14:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/854826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/854826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/854826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/854826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/854826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/854826/1"}}},"commit":{"parents":[{"commit":"b2eceb9ba40a6b008b1331561cf9bd4267097343","subject":"Merge \"update constraint for oslo.vmware to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2eceb9ba40a6b008b1331561cf9bd4267097343"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-26 18:14:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-26 18:14:01.000000000","tz":0},"subject":"update constraint for sushy to new release 4.3.0","message":"update constraint for sushy to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Change-Id: I0a72ca1e4a68dc34601c834353ff4a6ffafd4e52\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3f83314cc607ff238d64b7fe47301507322be295\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0f25795fee9e0269cbe27766d33a76afe2e183f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0f25795fee9e0269cbe27766d33a76afe2e183f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
