)]}'
{"id":"openstack%2Frequirements~854829","triplet_id":"openstack%2Frequirements~master~I74e51c069c7625b856ad20c280d7f01a69512447","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I74e51c069c7625b856ad20c280d7f01a69512447","subject":"update constraint for kuryr-lib to new release 2.6.0","status":"MERGED","created":"2022-08-26 18:22:03.000000000","updated":"2022-08-26 22:56:48.000000000","submitted":"2022-08-26 22:56:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854829-new-release","meta_rev_id":"f1d57bc80a921471fcb2111bd5aa84fb5dfb4157","_number":854829,"virtual_id_number":854829,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 22:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-26 21:04:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-26 21:04:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 18:53:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-26 20:46:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 21:04:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"c636c8034825d323399242ef69e58876eab32fe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-26 18:22:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38934caf2cc973969d84b2a8ab05a641236a638a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 18:53:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d6106f8d32424a86b72bb2841927a7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/378e13b4f98541468206fc5900483d99 : FAILURE in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"3be118868e6dea75f6fbe29a106e18013cbccf48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 20:46:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df9ea3dc4bdc41389fcf1d1be5d14aac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dba48645a09403fa9eedfa59bef5883 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6236da78a154fbfbc03b381cc4bda9f : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c399d0047cb3424fadd7d405c125b56e : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c83d22f31c045129cc09681ccd09911 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0d908fb20024e53bc05b4c8e4e6d7a4 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45339c880b774d3db1524ce46eb1df73 : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3d2c73a69f3a4cc6a8cbacf3bc49207b : SUCCESS in 5m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0072c80b705a4d04b41e9cccaade1cb1 : SUCCESS in 9m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/658ad9a91e194f68a91dbf61c1662fc3 : SUCCESS in 14m 52s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/8bcecea85eec477bb856b0e4061f1dd6 : SUCCESS in 11m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/665f2799232f47e8a6c6d9fef0bcefb8 : SUCCESS in 7m 14s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3fcbed9b25624c4fa009f4a025558cbd : SUCCESS in 7m 59s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7336dc59190144048beab9e4e0b8be93 : SUCCESS in 11m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d6af06fa31e446a88d4ebd80c5aeaff6 : SUCCESS in 5m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/78fe2465d51449699b1bf7e49c318845 : SUCCESS in 5m 55s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fd082bd63bc24e73b1085b15c98e7f70 : SUCCESS in 10m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/957c0b531a8a4e55b2a43d02813e78a0 : SUCCESS in 12m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ee072428bb9401d95de96dd548529ad : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eee3b5d0ea704f5d81c7c7ae6707a485 : SUCCESS in 7m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3290c43fa85e4db1a11b671a6d72895a : SUCCESS in 11m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e30d82bade4340d682758da2a69f26c7 : SUCCESS in 4m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a0d4811a742c48f187b3884f2e025493 : SUCCESS in 11m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7d02c4d14163493f9bddfd38f0fb4622 : SUCCESS in 7m 12s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/0eff5e7430eb4bb4a9dbfb929a551839 : SUCCESS in 10m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4e358b251ab0453b975724c573a6ced3 : SUCCESS in 23m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/400b0d702ed24d29830b5f2c1839a40f : SUCCESS in 13m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ae8efa31ffea45e5826281091cb9a255 : SUCCESS in 6m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2771178b2cb34eb9b2b76952352fd38b : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/285e8d78ba464661a44a6869d0c0bb0f : SUCCESS in 24m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/df875f3e851a4a628928430ee2a28eb6 : SUCCESS in 6m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/057ed1a7dd6344fea1c155713ff1eef0 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/90b148ff71934480a710807335fbbf87 : SUCCESS in 8m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dc1464acc4734b588f85894c13bc7aa3 : SUCCESS in 15m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3874683ab97f45089bafeb503dc61e48 : SUCCESS in 52m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac5746ead4694f40aa1b659d9dd71534 : SUCCESS in 1h 36m 34s","accounts_in_message":[],"_revision_number":1},{"id":"340d0abab4f60fa0c3e78f9798bf952659d55da0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-26 21:04:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8085705d1ec79db9176640a14900e030fddedb4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 21:05:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"433f08aae0f994df5a0449113211a22392f53451","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:56:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5350f85c325841b4a1cd41d438f82115\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab8be60997b4fcd971cbbb8535d8359 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fe3a2078ae64ab0855de975669744ab : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f82b35ede5d9469cb8ac93cdca850b71 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a29e3551d3024f5eaad490adda45af7b : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/50c57f52961b4921b8e31c33300737a6 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/46113dae2d6b41d8ad9e8a0651ce72ae : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/899d97b3b17348618609ab4b87112656 : SUCCESS in 8m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9684f4ff6afa4f939da0cc4dc8bc642c : SUCCESS in 14m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5e6f190a3db34be9bd523ba8ab907dde : SUCCESS in 9m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8aeec5bc854241a88b52e89bbe24e5ce : SUCCESS in 5m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f9712d0dc791400d8136a28e87f52267 : SUCCESS in 5m 34s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6aab578e2c134f0196ff6d19c416feb4 : SUCCESS in 10m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a9f3fe46438b47919b349db1524e8d8e : SUCCESS in 11m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb46753a6e1340a6b98d7532238e6e6a : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b8412a7327644c42aa4b1c194bce5207 : SUCCESS in 9m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4cafc08f08f54cd9b4e60a4ac1c11361 : SUCCESS in 14m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1fce0e3d7e734b19999b4876a91dcb05 : SUCCESS in 4m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9dc93be2022e4547b8dfb5c8398d7093 : SUCCESS in 9m 22s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1d374bc36ba949bd90e439cb38a29361 : SUCCESS in 5m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a91a968bcbc841a69bf60aa4d0b7d5f6 : SUCCESS in 26m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5079e63128a446d89c723db0625bd93b : SUCCESS in 12m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6f656641bf5f4e24b7f451b5b2a5370d : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c7e856e13b64bcc99b0320966750644 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb3e719af57c4981a805b3e4478d3251 : SUCCESS in 21m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c4436e808a344a0f8aefadb02ef5fe77 : SUCCESS in 6m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d8a7a2eb930940fd8aee6eb6be80e872 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f5c1906f0194711adfa6d52985d8819 : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dca2c8b3c2fb4c1aac2644a8d3158fb9 : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d695b12764c434ba6aa15e9ffd9e136 : SUCCESS in 41m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3a91b717b984ef1965658e80803de1b : SUCCESS in 1h 21m 53s","accounts_in_message":[],"_revision_number":1},{"id":"f1d57bc80a921471fcb2111bd5aa84fb5dfb4157","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:56:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96ec97784fc77623f21c81073f7e7456814425f0","revisions":{"96ec97784fc77623f21c81073f7e7456814425f0":{"kind":"REWORK","_number":1,"created":"2022-08-26 18:22:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/854829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/854829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/854829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/854829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/854829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/854829/1"}}},"commit":{"parents":[{"commit":"b2eceb9ba40a6b008b1331561cf9bd4267097343","subject":"Merge \"update constraint for oslo.vmware to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2eceb9ba40a6b008b1331561cf9bd4267097343"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-26 18:22:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-26 18:22:01.000000000","tz":0},"subject":"update constraint for kuryr-lib to new release 2.6.0","message":"update constraint for kuryr-lib to new release 2.6.0\n\nmeta: version: 2.6.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie6904c7d17c093766b05f03f33e84a779a5f05bd\nmeta: release:Code-Review+1: Maysa de Macedo Souza \u003cmaysa.macedo95@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I74e51c069c7625b856ad20c280d7f01a69512447\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96ec97784fc77623f21c81073f7e7456814425f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96ec97784fc77623f21c81073f7e7456814425f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
