)]}'
{"id":"openstack%2Frequirements~854832","triplet_id":"openstack%2Frequirements~master~I4d5054392a44839ee73f56aabf478da876ef8207","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4d5054392a44839ee73f56aabf478da876ef8207","subject":"update constraint for mistral-lib to new release 2.7.0","status":"MERGED","created":"2022-08-26 18:29:05.000000000","updated":"2022-08-26 22:56:55.000000000","submitted":"2022-08-26 22:56:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854832-new-release","meta_rev_id":"aee1cbc6b94ae548bf39baac8df176f7b00b6470","_number":854832,"virtual_id_number":854832,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 22:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-26 21:04:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-26 21:04:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 19:49:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-26 20:52:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 21:04:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d15df9cb828634bec2028d61dc913493d965826b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-26 18:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d96c4f5eab6eadd5ee4809957509c89b2f8c487","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:49:59.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851812bfbefb4e10bad013e86cc5bfef\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e27d8707ba6b41b98bd208deb96899ee : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"85de890c48a6889ec7c0f9b96ab7b7427a6083a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 20:52:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8512090769744225a787dc2ec9a54ac3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb70b2e2c1149fcb3a194399589417b : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de626399ab8d405eb05fb74c607b0042 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f7c34c1c11a4df0ae6bce50b00bb33e : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9f027a1127d4a4fbdee514b30c7d218 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/057bb60359fc4dcdaf50334d6466da9f : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b03fb719c154c99b463868028654a4b : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39885f232ef3497fa765d735aab5ad4c : SUCCESS in 4m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1f40c4bfb0f0454fa62ad3b4dec5643d : SUCCESS in 11m 01s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2b815c4378444660b4964fe439c5aacf : SUCCESS in 13m 44s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ba783f0b29f241f493a1a9413205b94f : SUCCESS in 11m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/08f92e232d994eceb8fffc250cbf8e72 : SUCCESS in 5m 58s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3c8d6f95edb5460082d4f7df4ba604b0 : SUCCESS in 6m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f4ba9c4237994cacbc41e453dd994851 : SUCCESS in 12m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1850ef140d1a49508a37cd079352dc1c : SUCCESS in 5m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b3d301947a964ec38448ce4a3841066a : SUCCESS in 6m 08s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/858c3b5dce5c4e73a95d1ce2dd5bb3e4 : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e103de074161413e9d6e72aece179386 : SUCCESS in 12m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ef12b8c24c94400eaeb8fd07853a8bd6 : SUCCESS in 9m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/396cba2eff0c4196bb88083ac218ac23 : SUCCESS in 9m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/52da0f4449c94d528065a402c58a75ac : SUCCESS in 18m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4c04c7b7d0dd4b71951f6269c1445bfb : SUCCESS in 5m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0179cdc1428e46c3b404ad5c2bc83484 : SUCCESS in 11m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/439332f5493e410bbb85cd3be79e9a87 : SUCCESS in 6m 00s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/34f20c4cc3244251aa110b21d4a18822 : SUCCESS in 9m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d7ca99cdafc14394a89d354fefa18737 : SUCCESS in 25m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/818b783cddb3410b90abf224cadac292 : SUCCESS in 14m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bb2e288b1971433ca36a83f30a9815dd : SUCCESS in 6m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/88632e88bd0643fabc35469d69a95da5 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e04ef9ac69d54d76a6d8db0ff1854472 : SUCCESS in 20m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/65c926d893cd40419f952d686afe43ec : SUCCESS in 5m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0bdc295c77a042d5a990941f554c622e : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f712623ee494392b5cc2cf04f2e46ce : SUCCESS in 6m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/842f662daee141b0b385e3ee5d58a855 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ba9492932c640c6986ccf41f5cfd29b : SUCCESS in 55m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02c74e88da2649d98a563f71763209c7 : SUCCESS in 1h 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"3c959d0eeb8530384c1f17ce1f6fdc5e1bf5c0d0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-26 21:04:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dac8498ef02a9de6dc79d7b78195ac852b7a4000","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 21:05:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b0f7316395fef5d383de5c04060a85665467c08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:56:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e6801fb09b042d0b96e66e065bc5fde\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e3f57a5ca9c4da09d14801cb4871f93 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b744c688264d3dbc7945e19f621fdf : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/103055b51ad0481498e033eff51d194e : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00ef1f983a604cb0a9debccefa7cdb78 : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0233a6ebea7478187755b6957da8037 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00307b6fa51a4eecaf3960ece01a1af1 : SUCCESS in 3m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a9c548bf1e29468fb95d118af1c5e4c6 : SUCCESS in 10m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/62f88393b52441379f44192ca850321c : SUCCESS in 13m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4a848cd6e1e64bc0b14ab633129795ff : SUCCESS in 10m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a1500c4c908d40dc82ed0a3a640dec99 : SUCCESS in 5m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b77a69e323964c0890ba6a2234dc47fd : SUCCESS in 5m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/4c767f6e978046e395d75444dc8c9ced : SUCCESS in 9m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b468a079ac3a4708b5d2a0dc05ce3dfe : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7bf3370a3c6a4e3cade25e37808d7340 : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1f5f47d1eec648c8bdf262d861ff7ce1 : SUCCESS in 9m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d7044379db74446e8669a125baf4b3d7 : SUCCESS in 11m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fed1b8a5f30c4bfd833cad9887330579 : SUCCESS in 3m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/04dcaffdc4304600bd948d38770072c0 : SUCCESS in 12m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/fa3f0d0ccb9f4e35ad0c499a3d4831e4 : SUCCESS in 8m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/66fe9337e0044dd7a120ce0221c22adb : SUCCESS in 28m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c9d4c9e67bd84942893b003775a6c20b : SUCCESS in 13m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cddb7429803c4620834adcf34661d0f2 : SUCCESS in 8m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/66a98fee97a849d8a990deaf1310a989 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78ea97e96f60493b826ac751d3aa5ded : SUCCESS in 19m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e4acda64780645c3a0a3ea440a75de51 : SUCCESS in 8m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/af8cff41a52940d7ba92eab1ba4e1832 : SUCCESS in 4m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4cf90e6bcafc404b9d518f354854d02a : SUCCESS in 6m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7631fd65af204c8bb4b66b861d489eb9 : SUCCESS in 13m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3039dfa4a314d9996156476bc542cec : SUCCESS in 55m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c4ffe1e2c24746f683000afcf424f2d3 : SUCCESS in 1h 30m 52s","accounts_in_message":[],"_revision_number":1},{"id":"aee1cbc6b94ae548bf39baac8df176f7b00b6470","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:56:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59797eda3ddddcf51624fb66f6317623b2118101","revisions":{"59797eda3ddddcf51624fb66f6317623b2118101":{"kind":"REWORK","_number":1,"created":"2022-08-26 18:29:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/854832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/854832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/854832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/854832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/854832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/854832/1"}}},"commit":{"parents":[{"commit":"b2eceb9ba40a6b008b1331561cf9bd4267097343","subject":"Merge \"update constraint for oslo.vmware to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2eceb9ba40a6b008b1331561cf9bd4267097343"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-26 18:29:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-26 18:29:04.000000000","tz":0},"subject":"update constraint for mistral-lib to new release 2.7.0","message":"update constraint for mistral-lib to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: If89a3958c25fc2eaca9bc977c96c32409221143e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I4d5054392a44839ee73f56aabf478da876ef8207\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59797eda3ddddcf51624fb66f6317623b2118101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/59797eda3ddddcf51624fb66f6317623b2118101"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
