)]}'
{"id":"openstack%2Frequirements~854835","triplet_id":"openstack%2Frequirements~master~I0db61058c2b97ebd320f7e479654830d105c85ee","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0db61058c2b97ebd320f7e479654830d105c85ee","subject":"update constraint for ironic-lib to new release 5.3.0","status":"MERGED","created":"2022-08-26 18:38:16.000000000","updated":"2022-08-26 22:58:23.000000000","submitted":"2022-08-26 22:56:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854835-new-release","meta_rev_id":"3f45340e1b7585315c00866803221616b569eee8","_number":854835,"virtual_id_number":854835,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 22:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-26 21:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-26 21:05:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 19:50:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-26 21:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 21:05:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"afd04b28d3cf6a5d630273caaaf1e49c6958b145","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-26 18:38:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bd5c3d835d8e48db87659344322b8445e3346ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 19:50:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93c0d96bf65b4784b5c97f013b4ab315\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/77fea1517bfb4fb5b1eaee545e5f5d25 : FAILURE in 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5fc020cb3911d22724562dbd47205dc541b2b5f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 21:04:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00c57af2deb94b5c8f23c5e010d52819\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980282dfa6c04c6eb933d89b77ca1584 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca2fd9c727941ab9d9ba8026acb3f35 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0233df7f46349169bcb1b2e303fb51b : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c22f659635240beafd115084c3f005d : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1388bdc85af544f18f69ac5403f5fb93 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ae1e785a5af942ffb649cc7c48817cd4 : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/759282cf99c84eb9974f9b2a9cb188c7 : SUCCESS in 3m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/04233abe4543403f9256a1f6a94eb82d : SUCCESS in 9m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b239ab0f0eaa47ed868e55f540754cb3 : SUCCESS in 12m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f40bef0426e0452aa9901275e24bc50e : SUCCESS in 8m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3808f0405473468eb1253a633c80a855 : SUCCESS in 5m 51s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/9423c92fb8bd48af97b7d765fac430c5 : SUCCESS in 9m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8e89b99f56314617ba256fb2b38ce5e6 : SUCCESS in 11m 51s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/94e71101778240558a01ae284bb368a4 : SUCCESS in 6m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/df4e417a124645e2a3ae3ca6fbffc5c7 : SUCCESS in 6m 18s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/66f7225bcbaa45a39c411458a30e58f9 : SUCCESS in 9m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/77ed092cfaa041e89e150cc418735623 : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1dcd9e228bf842938f1ab2cad114af69 : SUCCESS in 11m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/14a075cf107443388f084ef44a9b1566 : SUCCESS in 11m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/66f51b0041884c6fa75bd17c845a61b3 : SUCCESS in 11m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f41222ea7e2746ae85f355d77b10f8e0 : SUCCESS in 5m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/58be7d6b1fef401fb99c4336907f7d93 : SUCCESS in 11m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2484ab79a10a49a281606b12fe1465a4 : SUCCESS in 6m 04s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/ae3bb82329ff49c28cb8f9205386ffdb : SUCCESS in 10m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cacf916abfd54046a20e2655afbfa37e : SUCCESS in 24m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c68876e869804e2aa59be4bce072943f : SUCCESS in 13m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4c3b31b895fe410ab30de26e43ae67f3 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3a311d0ef6024ca8bad68128a4e3eb95 : SUCCESS in 5m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6164eb9cbbc34620840ba8f8bdd7b1e8 : SUCCESS in 25m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0c6912b88fe04f77b21d198c57840ee6 : SUCCESS in 6m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/074c33ed6f3740d49b85bc2fedc6c5f7 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0640f7ffdbc64b58ba89d4966e744bb4 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b03ec72a93af4441877e24b9680c8faa : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4da493406510427493796eea1edc54de : SUCCESS in 47m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/011533fdcb3343fba234e7a226f9297c : SUCCESS in 1h 43m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6a7e2117a277c02e838f781178ed811724429abd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-26 21:05:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86201f974e79cf2370c0f7351d2312e4a17e949c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 21:06:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba1cc640664330bd8f27a83419112662b2c09e16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:56:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/543417ca79c54fc1a634d7afb6909961\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf18fc22d313429894e2cc4a25fd8790 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f02ab2f14644031b57c9db3bded3720 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afd5b6ab0203427d82db66286ed2fda7 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44851306358a44218f8d4041866f6ccf : SUCCESS in 5m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fbcd4f916554ce997288d7a927ca04e : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2a386c8353e54fef809ec76ce00f4cc4 : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/59ff1740078b4bf499ee91635beeadb1 : SUCCESS in 8m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d6a80a8f5a9e467d9c1f52a7f255d433 : SUCCESS in 13m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ef3d5412106043b7a7751e1842554241 : SUCCESS in 10m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/cdd2e4a8a2f44968a269bd47acb09819 : SUCCESS in 5m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/15d58ad800344f9b9916065060148da8 : SUCCESS in 6m 08s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/69b607c363004e7c9eae29f1701f559f : SUCCESS in 8m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e7a455f2114e4653b48c5eb4859219ae : SUCCESS in 9m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26070dbc370041ef8af4071b859fa8d9 : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1680ecfb1f554c659fdd232885e2593f : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0cc2d487d0e8420596db8ec679ffa5fc : SUCCESS in 12m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4617d234d969476f92ba3fdcba99c240 : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a50d76671790438c9c43d6d0d17e9a9e : SUCCESS in 10m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d032ae10360e4b8683d699e1b917eb9e : SUCCESS in 5m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/49488479d6514857b872310aa50c4ee4 : SUCCESS in 24m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/09aa2a043dcd4551a79874a9999bf5ce : SUCCESS in 13m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/377f8c69c9c64f62bbadc6ee57978f9c : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0667cfa5aacc4da0bc0635678c9f19a2 : SUCCESS in 7m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0790719287b34a00a7e559adbd2e7cae : SUCCESS in 21m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/308743e6313340139ccaf58d02ff378e : SUCCESS in 6m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/797bbe770c7846e48931a07bda33912d : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a09eea6bd8b046508e55409292c8c3d4 : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e038e4aa8fab4b30a54ccf5375cb1ccf : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67269c9d63dd472e9ee11a2173e08968 : SUCCESS in 50m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b5df3f9091ca443a863681d16e35af8f : SUCCESS in 1h 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"334c90c06ad3ebac749f458d40f352ba662f3856","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:56:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3f45340e1b7585315c00866803221616b569eee8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 22:58:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53a63eb4235044a2a955d46a9983de28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9adebe4332294a87ac2363f683463de3 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ae6dcb0e53dbfa774d4ad30ce2e13ad5540c0cf","revisions":{"1ae6dcb0e53dbfa774d4ad30ce2e13ad5540c0cf":{"kind":"REWORK","_number":1,"created":"2022-08-26 18:38:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/35/854835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/35/854835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/854835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/854835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/35/854835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/35/854835/1"}}},"commit":{"parents":[{"commit":"b2eceb9ba40a6b008b1331561cf9bd4267097343","subject":"Merge \"update constraint for oslo.vmware to new release 4.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2eceb9ba40a6b008b1331561cf9bd4267097343"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-26 18:38:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-26 18:38:15.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 5.3.0","message":"update constraint for ironic-lib to new release 5.3.0\n\nmeta: version: 5.3.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I05f08dd52723ceb1f7ddb1d5e31b999e12603298\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0db61058c2b97ebd320f7e479654830d105c85ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ae6dcb0e53dbfa774d4ad30ce2e13ad5540c0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ae6dcb0e53dbfa774d4ad30ce2e13ad5540c0cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
