)]}'
{"id":"openstack%2Frequirements~855029","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I8ed2a3fb959244742af01146e2adb6fbbb1b7509","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-08-30 06:28:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8ed2a3fb959244742af01146e2adb6fbbb1b7509","subject":"update constraint for oslo.db to new release 8.5.2","status":"MERGED","created":"2022-08-29 13:10:03.000000000","updated":"2022-08-30 06:28:39.000000000","submitted":"2022-08-30 06:28:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855029-new-release","meta_rev_id":"93feb05b53a1eaccfc5552d910107a8e58a7a3c8","_number":855029,"virtual_id_number":855029,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-30 06:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":2,"date":"2022-08-29 18:34:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2022-08-30 04:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-08-30 04:20:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2022-08-30 04:38:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 13:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-29 15:16:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-29 18:34:39.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-08-30 04:20:39.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-08-30 04:20:39.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-08-30 04:38:25.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"b0c8957fc4f8bba52a2d4fbbbb59b161fd228c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-29 13:10:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ecd257ef552e2cfc6bafb15cf24dbb43ddca7b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 13:25:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77a1453675c7436caa62b5ebbe4c4422\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/49e50a6a7d3d41198f8b574cf190b35d : FAILURE in 13m 03s","accounts_in_message":[],"_revision_number":1},{"id":"311a23b5ba1b2081bf30ed129413d9c3b6a47d5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:16:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e9969b281c64245b8f2597357248fb8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527d5481718f4751b4e54478d3040213 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e774c1a57dd48f2aaab1fb543aed121 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fad67fd50054452f8795f23e17dc10be : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab902ffdf7a94b4490873d81cd8b7a5f : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b8121998c2040278b74f2999c8b3c5d : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a62b346ca54042799bd6745ffacb1df3 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/efd090d995964e7387b4cf95227e9b41 : SUCCESS in 10m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/071643d5366a4fc5b9e8bda12317d506 : SUCCESS in 9m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/756924fe47e24dd0b3bb235169a179e2 : SUCCESS in 9m 56s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/e985e8dba8b84f5e99578b529a052eb3 : SUCCESS in 5m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/81be3884b7874afa90af84fdec112f67 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e97cf85753224f649d5a70bf8557ea26 : SUCCESS in 8m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4d9af94960374167bbd678823d68bcae : SUCCESS in 9m 00s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/63783a75ef3248ce8a8804c36ffeeb21 : SUCCESS in 34m 21s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/263797fe57314b23b264b3942f7a97ba : SUCCESS in 5m 36s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8e24b8e3941a40b9834b5c5065d45a1b : SUCCESS in 29m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1f6e7786cc494b2a91f73dea0a918916 : SUCCESS in 12m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2262739354ad402ca58243aaae5013c5 : SUCCESS in 5m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d516ef4e46524ce9a30e280534dcc9ec : SUCCESS in 19m 33s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/43271f7a3faa4d70b201c854ef1d1443 : SUCCESS in 5m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c70229edcb824dbc9e9f56b6cede9dbd : SUCCESS in 6m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fc8051dce20c4956827c27962f3162c3 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f2bc58a1c5c470184553f85246a3355 : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cd65fe33e7904582a7ee99e8004bc98a : SUCCESS in 2h 04m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d9df36d8474a3d0112bf551c1479098f80514fcb","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-29 18:34:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3a4aefe63d4938140bd853af37dad26206294dfa","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-08-30 04:20:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"17c33e951b5af4df107c1a8d27a4d2574892e4bb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 04:38:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"359c9ec98e03adaf498fe47f6d852038ef6f4665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 04:39:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebf0a6b555d1079f45049d38b37222872360cd5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 06:28:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e400a59d592e419aa9fbaa5faca329c6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/296ca5ae52ef49f8aee536bc29189926 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/994375d4c50d4f75b9e9f89c3d2f4adb : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/924d2ea2695249668ed4b66bc41c4d62 : SUCCESS in 4m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd63e921331f42cd8459d13f3aa94018 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6c7915a604234b9692031f20af84415d : SUCCESS in 4m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d87e382705dd4c479fb9d6938ca428bc : SUCCESS in 8m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0eed2d9fa71b4f06aec8e34b0fe76231 : SUCCESS in 7m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0f4410f6aa26490ea43c628325909ed7 : SUCCESS in 11m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dabd8f09f71b4246a05cb904cfaed156 : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f7f448dd76d2423495ff3834efea00a2 : SUCCESS in 11m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/28398b9ae8dd4d98894d53af62926bc5 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/01fbf82f89414205acbcd670980d9be8 : SUCCESS in 8m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4153779856f04e01a36f46e7e822d54e : SUCCESS in 14m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b31236c420434cf5ba49cb42a15fbce2 : SUCCESS in 5m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f27143857bee403ebe25b2f3308bc199 : SUCCESS in 29m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/52615a7954994611b5bfd59258966f64 : SUCCESS in 11m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29d4e9f9cd5540569e98afc9193b19b2 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa11d6cd4357426cba79bf9ca86075ac : SUCCESS in 15m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c76f0ac608cd480485e0a6ae22bc9829 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fc5a009c4ecc41e98535ebdf91cb54f6 : SUCCESS in 6m 34s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c24bbc7e738b44c895e09e6c13571c36 : SUCCESS in 11m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddb950be60b74c2dafd92114f6f7082a : SUCCESS in 1h 02m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6194c5d99a5e4ecb98662c1198803b6e : SUCCESS in 1h 48m 45s","accounts_in_message":[],"_revision_number":1},{"id":"93feb05b53a1eaccfc5552d910107a8e58a7a3c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 06:28:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cf6844e004cd29d8214e93742303e39f04b6652","revisions":{"3cf6844e004cd29d8214e93742303e39f04b6652":{"kind":"REWORK","_number":1,"created":"2022-08-29 13:10:03.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/855029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/855029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/855029/1"}}},"commit":{"parents":[{"commit":"c6030e008c9b1639cfaee27f1f2bd321ada1f372","subject":"update constraint for metalsmith to new release 1.4.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6030e008c9b1639cfaee27f1f2bd321ada1f372"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-29 13:10:00.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-29 13:10:00.000000000","tz":0},"subject":"update constraint for oslo.db to new release 8.5.2","message":"update constraint for oslo.db to new release 8.5.2\n\nmeta: version: 8.5.2\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nmeta: release:Commit: Alfredo Moralejo \u003camoralej@redhat.com\u003e\nmeta: release:Change-Id: I34de5f9d5e0e0794d7790e18562bb545470beec0\nmeta: release:Code-Review+1: yatin \u003cykarel@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8ed2a3fb959244742af01146e2adb6fbbb1b7509\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3cf6844e004cd29d8214e93742303e39f04b6652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3cf6844e004cd29d8214e93742303e39f04b6652"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
