)]}'
{"id":"openstack%2Frequirements~855353","triplet_id":"openstack%2Frequirements~master~Ifc0ebe16c8d6e2d24729f3ec52fca8b90616c727","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifc0ebe16c8d6e2d24729f3ec52fca8b90616c727","subject":"update constraint for aodhclient to new release 3.0.0","status":"MERGED","created":"2022-08-31 10:32:39.000000000","updated":"2022-08-31 16:22:42.000000000","submitted":"2022-08-31 16:21:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855353-new-release","meta_rev_id":"58383386cad72d476d61c69f700f7fb09adc31e1","_number":855353,"virtual_id_number":855353,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 16:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-31 14:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-31 14:35:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-31 10:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-31 12:07:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-31 14:35:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f7625b41803dec14324b7a02941478960d61ddc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-31 10:32:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbe3248fa499272b70ca9e15bb4c0ae62e6106c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 10:43:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c978655beaa43428970bb9ea49427d8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/db50cd1c51ad48ef891a1db9390390e5 : FAILURE in 6m 22s","accounts_in_message":[],"_revision_number":1},{"id":"272ab99941bd60c42ca983669b242ef725842cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:07:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d282f91a387d4b1a96c255a78de0fd2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635543d33d11497cb2dcef1076eae9bf : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a901165a06143bd9910b6455bee7fb6 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d561b371e6a245598264292cf5f57008 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79c3ee130a8c4ef98969cbf7a237af0f : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69a5d8db915c42d698ce8cdf7d60cdd6 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/978ac739fba045a8ace37b954a886092 : SUCCESS in 3m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7e0cf0773fcc499f9bb8449efebfa537 : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b12164cd0ea24cc49f71ac3d727d4456 : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/38338f22a1df4af29e032e1b924be3d0 : SUCCESS in 13m 15s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e3990228bf614658845c5e2d1c6fa361 : SUCCESS in 7m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e8a34077dd4f4f08a78796890855b200 : SUCCESS in 4m 57s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/efa1e3df5012400a977ee421baaeb1aa : SUCCESS in 5m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ca772131f4434ef3bbb810eb1dac388d : SUCCESS in 10m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/038bbe3d4f86424a9912b3a69032e9e1 : SUCCESS in 5m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ffc8cd6c323a4008bb5c6458759be01e : SUCCESS in 5m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/36e9c46adf88401cae7222a55a21ae6e : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cc597034883246d29ca56ae4bd61eb76 : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d31fc98d97f40b39aea9893ca4ace2e : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5ecd8584f20d4862979698d8c4c23791 : SUCCESS in 7m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a59acfee649147f9a90cbbc32b780aac : SUCCESS in 13m 26s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/519e69e579d745979cc75c6dde4905fd : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1c7a1d21063f47d8b248edb775895ccd : SUCCESS in 10m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/07d7bfad9992419fa68b0d231bf51f09 : SUCCESS in 4m 25s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/66b93aaddb724c689809650fc6b98067 : SUCCESS in 9m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a03c1c82dea646c882e3a79e80153492 : SUCCESS in 24m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5c859f8d73b84d23b8f6bd8513510d21 : SUCCESS in 13m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c50167920ac24740a31370ba1e4366c0 : SUCCESS in 6m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c08eb65e44894a46a3f51ab1ac2cfddf : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0fb5a759ad3844d5b6d1c558c6525ddf : SUCCESS in 25m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1da9a73526ae47f49e5eeb1eb4aecf7c : SUCCESS in 5m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d5479dcf00a445b79e69e9253dbbf9b7 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89bbf4c3dd99467fa26d4c144404af1c : SUCCESS in 6m 59s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ee72d95f037748d99a21c2bf305b7c39 : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48305b36631d456bbce0455d18a1ffab : SUCCESS in 1h 11m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b66ab232914c42169c6d18354e31d7ff : SUCCESS in 1h 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"51c1bed6618acce3d68add02bae48d23b8a503dd","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-31 14:35:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b21db98a499b2b155788f62438a8334c4aac6fd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 14:36:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68dfe9357aabdfd0146f52ac595c304e1591dc48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:21:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d52361eae17e40e9bd50d7ab0a380b9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a341959746c44a3ab0e5f3ec4758d04 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5faa8d36020745d2a01ee8b1472757d7 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b279ae723ec94a13b72724ac86bb7cbc : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/149686ac6952477bb2e43691de5f5936 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/652fe722384d4a0d85fe7f04ed12bfe0 : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d8bf836932714cdebb5fe248ca3a68f5 : SUCCESS in 6m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c695eafffff94340885c35ca964318d5 : SUCCESS in 11m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0b742035bdc9479aa638323e7c838129 : SUCCESS in 13m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ad9559b14ca54e8f8a3ed4a685a98efb : SUCCESS in 12m 42s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9ec709d697d242e3a48e1e67349db1b8 : SUCCESS in 5m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c7473cff1520487786642741d9a95426 : SUCCESS in 6m 45s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5e899430b7af4089bbf3be36ab09dee6 : SUCCESS in 8m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/108a2c13e3eb4f1aa3e0ba17dbca77c5 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3ac4087620644f0ebe812cea0ed4095c : SUCCESS in 9m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bf7b90eaa3c04058a2fc3b9da3a52cc9 : SUCCESS in 12m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/087d26e6db9a4db1b34c74ec5337dcf2 : SUCCESS in 15m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/75cb973571034505a3adca17eaa5fba9 : SUCCESS in 4m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bb275a6bc6a44140b9db67a942be2944 : SUCCESS in 11m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/533aecc5a8384c04b22c84945e61a0c3 : SUCCESS in 7m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7b4e59623b614a9884b78db3d2d10539 : SUCCESS in 23m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/75fd02a571344593bdfd2980c726269d : SUCCESS in 16m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cfad317e1c0e4e4e84ca800d46ec9c6d : SUCCESS in 7m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/150015f3745a45ea90b802f5686daeb8 : SUCCESS in 5m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9fb299eaabde4234af0bde23e530e288 : SUCCESS in 21m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c17a7a7400ec4d358aa3331437dc6e36 : SUCCESS in 8m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f61bb803ee194e0c8becf08c317c4ffe : SUCCESS in 6m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c26f864da46c492c83b9454a753bee5d : SUCCESS in 8m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/81c1a44eed574aeaa92e29cfc8126c87 : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c49bcc6b7784b6988ce854d69a7ac31 : SUCCESS in 1h 11m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27850c6c5b95475589d5c9a58d8a9606 : SUCCESS in 1h 43m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8ab0c8925d0f6000be90214240e67cf5b0596e6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:21:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"58383386cad72d476d61c69f700f7fb09adc31e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 16:22:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43c490555b52466ba91b5d8697ccce1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2621012079c64ae193855ed1a89e6a57 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ce5beffc8f80ea5dfa94a6e7a300544240969ad","revisions":{"1ce5beffc8f80ea5dfa94a6e7a300544240969ad":{"kind":"REWORK","_number":1,"created":"2022-08-31 10:32:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/855353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/855353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/855353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/855353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/855353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/855353/1"}}},"commit":{"parents":[{"commit":"a0bedb39c9e6cc0b11be46468ded0a33ec46d291","subject":"Merge \"update constraint for python-vitrageclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0bedb39c9e6cc0b11be46468ded0a33ec46d291"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-31 10:32:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-31 10:32:37.000000000","tz":0},"subject":"update constraint for aodhclient to new release 3.0.0","message":"update constraint for aodhclient to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I404afc5442e0befacb701c43008052390a71be8f\nmeta: release:Code-Review+1: Matthias Runge \u003cmrunge@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ifc0ebe16c8d6e2d24729f3ec52fca8b90616c727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ce5beffc8f80ea5dfa94a6e7a300544240969ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1ce5beffc8f80ea5dfa94a6e7a300544240969ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
