)]}'
{"id":"openstack%2Frequirements~855354","triplet_id":"openstack%2Frequirements~master~I347e38fac5f1e591d96c0153a4144d4979e1128c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I347e38fac5f1e591d96c0153a4144d4979e1128c","subject":"update constraint for python-cinderclient to new release 9.1.0","status":"MERGED","created":"2022-08-31 10:33:10.000000000","updated":"2022-08-31 20:54:19.000000000","submitted":"2022-08-31 20:53:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855354-new-release","meta_rev_id":"a92ae00e8c67cf41592c3d1603b459e13ac2089a","_number":855354,"virtual_id_number":855354,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-31 20:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-31 19:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-31 19:21:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-31 10:46:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-31 12:03:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-31 15:35:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-08-31 19:21:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"55597afe4b8fc791e76272565004b5a9290a062e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-31 10:33:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a58fc8fbecbd303c6ef2052fae3ea83150389bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 10:46:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2e9f7cb80fb43bca6cb8c352ff56d07\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/33e30284a1bc4ccebdcb99e9aebd095d : FAILURE in 11m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ccd68951e0a23e1b6d4ef8bac14e795d298a3a34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 12:03:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2428d441ceb946dfbaf670f65a96d8ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13fe12163300472cbc08e07b03d72ea1 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42638de8d1564248924a379f788912cd : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf312d886ebb4635afa43fb9a127feb7 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f760f0fe21544f8b3e1d463dcb38b32 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c591fd6dbda04940aa3cddc4ecaefc0c : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ea511dc191af49f3a17388ada7b47eef : SUCCESS in 3m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/54bbb34abe09452c941889c847ab282f : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/510520440596429c95e810d0d5dc9292 : SUCCESS in 7m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a339bc68e9be41a19a1e2275d78d92d8 : SUCCESS in 12m 00s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/fd93022f0ff2412a8e520d546e580062 : SUCCESS in 9m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/104e398966614e11be344c631c885928 : SUCCESS in 4m 41s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/11684ca6dfa047f6b8f0e8dfce0ea560 : SUCCESS in 6m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3b1e396e0a7644a3b39393d4a6277152 : FAILURE in 10m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a7571d643f4f4be89d4ea2f68c02a2ce : SUCCESS in 4m 07s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8007e9033df24e68b11122e97b20f532 : SUCCESS in 6m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/265e605e442346cba1ab8c8835c3a375 : SUCCESS in 9m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9038f899461046bb8c351a99d5961d23 : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66fcc912b94346a189d9722cc68a01f1 : SUCCESS in 7m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/41bbcda0c5f54954bb1b00743068eae5 : SUCCESS in 10m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/81ee01a01aef4fd1a396eb30e73dc8b4 : SUCCESS in 10m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b5670271b10f4ccaa15dcd015a0ddcba : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/74c50b5023bd4c059983f8f8f7c8719f : SUCCESS in 9m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/604a694e6dee4e57be382822081f3896 : SUCCESS in 5m 34s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/7118a8ad463d444c85e003bfa758b988 : SUCCESS in 8m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1693b215a6934f1995a6d9a4c79d9a2f : SUCCESS in 24m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4d3a95c3a5934dd3a54b4a74b963f0ac : SUCCESS in 13m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a2c6070f5bab40629e70bfb8cc62728a : SUCCESS in 4m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e2936d2cf3ec4564bf26a02de553dbae : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e2950aeb696446999cca67f7b2ebb3df : SUCCESS in 22m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c5631c6b13cf45f2ba611a571bada60b : SUCCESS in 4m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/15932d8eba8747c5aef7187c820f6ffb : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e04ec2b3ebc47b9b9a431c39a05e983 : SUCCESS in 7m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0ea4d041665e40b29b9ba9f795722b7e : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/220f0c3bcb344a0c9e3bc01b6efc6cf8 : SUCCESS in 1h 06m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a534f9ad5e064b0abe370e6dcd51bb7b : SUCCESS in 1h 28m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a0374118118557d6c6607d49b8997ae3d87edbd8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-31 15:35:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b45cfa771effff90398c4d0b8e13caee6bc5fdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 15:54:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69fca1a7fc234514a182dce7d5435d26\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8775ad806dac41deb2c073696c76b07d : FAILURE in 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"f0d9c73192ac2d736675d1421b5626f99d3faf0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 17:34:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93d234ee7fba47a3990080bc29580685\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e0f53daa7f140688a1fd9e478be591b : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00e2b41e19c84c5cbcc442e452342f43 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02a0d407a5bb4f45bdffd02ea80de219 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55db69db76084e7681544790b4480c50 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/505682113aea43ed94766602e1ac8163 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ba5b1d07bca41a59bdd49c997071dde : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8b2e0301e19e4478992358727242342d : SUCCESS in 4m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35477494812f458aa29bdcde0af84f56 : SUCCESS in 10m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ed90428a336c4ba089b2a67b6baffdc1 : SUCCESS in 13m 53s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/9eff170b44384d19b6972d913e72135c : SUCCESS in 6m 31s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5cf1e42207e2483483057d1d98196e21 : SUCCESS in 6m 17s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/0bbff0c2d88d439e80004223e9f0ae31 : SUCCESS in 6m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0a136444af3b484780a2f23eb7b43cd1 : SUCCESS in 11m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a6a4ef1ce6d541ec92fb15e6178fb27e : SUCCESS in 5m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/296e3b07094f485babdf94719964551f : SUCCESS in 5m 43s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d05d175c10fe47e399b565c2308dad59 : SUCCESS in 8m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/850397e5d5f9468c826684627abb2a27 : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c49350e824494454878093558c745063 : SUCCESS in 8m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/507b0e3712eb4d1ea857b0b376f7af2c : SUCCESS in 7m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4f8a4e3950bf499f9cf94d7506abdc35 : SUCCESS in 16m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7ff50d403f384701a7ebbd40f3e1e027 : SUCCESS in 5m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a3001d9d67484feeae7553072d86f87a : SUCCESS in 14m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/11a16b685e0248e7a8b688caa38f358a : SUCCESS in 6m 15s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/884e59baec21434b9235671fbc99b930 : SUCCESS in 10m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6d8678ed9d824530895506f7a4bd04f0 : SUCCESS in 24m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e313bb28263747cbaf76411dee6c5bdc : SUCCESS in 11m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/587b1354b8a7450dbfb9deace378fa99 : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c59a5e5b8e3248aeac8c0755ce15461f : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d214e026574f4d0294b931cfe55e1734 : SUCCESS in 23m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d824ee39a9964061a99c32eedba960f7 : SUCCESS in 6m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1430e989456f43e9bab231bf783781fa : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d5f94467af74d26b4ef211cf2879025 : SUCCESS in 9m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e1bdee5902984242a1b1a3bd5b9959db : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5dbd1c234b246c1ac22746b17e9be5b : SUCCESS in 1h 14m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc6823ad23a44fada1c98b77c9430b35 : SUCCESS in 1h 37m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8444f50d1437b3aa8c0054b018e66fedcd060975","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-31 19:21:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"185272ce766aaf0528ac6fa9447f78314dd8eaf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 19:22:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"adf5643cd762091c5b859c700398ecffdeaf9488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 20:53:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b39b277b9ddf46ffb38b0863e04cdddb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d01ee049004993b9d518881ce67ba9 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d754847df47a470493e6f4d2a65eb171 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b32d9d854e24a7783ba49938cffe888 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef2733a7958c4821841d7edd16e1144a : SUCCESS in 3m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/29ab2428bdac4eacb2e401fd871788d9 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c74f7bb6b084c62964266819ebb9923 : SUCCESS in 3m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/91e5124e11ae4e11a41144242c8207a0 : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/224f2172f2ac46fca9e1e4ec14fafbdf : SUCCESS in 16m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6aa7d69d828547d5b71544fb53bc0e0b : SUCCESS in 11m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0a8f7122452d40cda83aa1cfdbc0b65c : SUCCESS in 5m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/decdefbdeec84f468acea6a9e072b1af : SUCCESS in 5m 04s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7440c059779d49d49916738f4a826451 : SUCCESS in 10m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f00ba48877d14c1885616efaa944e03c : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09f76e9aef2048e6a497f54b12c5207b : SUCCESS in 10m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/152eb10c36a146e191d3da37d8cc16b6 : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ba985c5533b24eabbf617d81cf3f82ac : SUCCESS in 13m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a13016d1175244fa9a9d100a1002ad5a : SUCCESS in 5m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/29a694c7c989427c9db26a8bc00f880f : SUCCESS in 17m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b2d95611765e4508812cb4a7f56116d8 : SUCCESS in 6m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f940473a0d634d579927fd390c1dc059 : SUCCESS in 27m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/eecad1f7861744eda6fe5faf9e08a013 : SUCCESS in 15m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/63051b2430474531a1d74c2ca82cfb35 : SUCCESS in 7m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/97788d8d572b49408fa7d9e28848bf7b : SUCCESS in 4m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e4b28c40e595406ea0f9320b3e942113 : SUCCESS in 21m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0fe7b4a4b848414f99d82e2bb29ad06e : SUCCESS in 8m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f787710d49704756acd84621ab8f1062 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/03b6a7bad94240db85b565abee573a03 : SUCCESS in 8m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/175e1bc4c1964499bb9f6c83046a987a : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1ffbd1315eb461b97978c56dff0241e : SUCCESS in 1h 02m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f6ed8872cbb847ac9723506f22322021 : SUCCESS in 1h 28m 43s","accounts_in_message":[],"_revision_number":1},{"id":"029ba7978a8bebb2e59ec5f78e6220386cba8246","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 20:53:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a92ae00e8c67cf41592c3d1603b459e13ac2089a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-31 20:54:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1788c7836b064064b7b0b3cc4134ee1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69b758bde80a44508d19628496b7f5a0 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c141ab05afdd52b5e74d19ddbe6ba130ce58660","revisions":{"5c141ab05afdd52b5e74d19ddbe6ba130ce58660":{"kind":"REWORK","_number":1,"created":"2022-08-31 10:33:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/54/855354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/54/855354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/855354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/855354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/54/855354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/54/855354/1"}}},"commit":{"parents":[{"commit":"a0bedb39c9e6cc0b11be46468ded0a33ec46d291","subject":"Merge \"update constraint for python-vitrageclient to new release 4.6.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0bedb39c9e6cc0b11be46468ded0a33ec46d291"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-31 10:33:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-31 10:33:08.000000000","tz":0},"subject":"update constraint for python-cinderclient to new release 9.1.0","message":"update constraint for python-cinderclient to new release 9.1.0\n\nmeta: version: 9.1.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: If8c58735b1f24ba6a77800ca5dfbc49a75e5a09a\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I347e38fac5f1e591d96c0153a4144d4979e1128c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c141ab05afdd52b5e74d19ddbe6ba130ce58660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5c141ab05afdd52b5e74d19ddbe6ba130ce58660"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
