)]}'
{"id":"openstack%2Frequirements~855629","triplet_id":"openstack%2Frequirements~master~Ibb37b85b1a51ca09bf5e78ac1fd523d17b0e0dd9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibb37b85b1a51ca09bf5e78ac1fd523d17b0e0dd9","subject":"update constraint for python-heatclient to new release 3.1.0","status":"MERGED","created":"2022-09-02 09:38:33.000000000","updated":"2022-09-03 18:21:36.000000000","submitted":"2022-09-03 18:20:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855629-new-release","meta_rev_id":"f1275c1093c3a8afd3c851aae4264dc314b53aa6","_number":855629,"virtual_id_number":855629,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-03 18:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-09-03 16:38:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-09-03 16:38:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-02 09:50:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-02 10:45:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-09-02 11:14:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-02 14:00:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1e8d9cd0a456d6b546e4d0e94a88e4c979a87e90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-02 09:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43bf438a90c444487b57aa7c7355b6f803132497","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 09:50:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0d83e85bd29411991cf611dd98f3f55\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9d860110892d4c0ab7f9cf138d3d6d71 : FAILURE in 9m 37s","accounts_in_message":[],"_revision_number":1},{"id":"86b46f418d4bacbc0b5bf4e7025441ab26d378c1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-02 10:45:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9a66ca03969f93f6d7de9823514af231953a472b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:14:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae0447f2986f4ad8874407e24d57207e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fcc6a1e9b944d03b295e3b67f8dc4ab : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d5f03adc1fe4cc4bb1f1bd383637de5 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f078e8ab5554efa995abc4d81ac7883 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f29bc3dce430468d890efdf7c035fad6 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45245421b65e4716acf30cef2f6401d3 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9f2fd2779ec54e51812cc1ca789b0453 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e285eea6530446e881f9a6bb24f6cc9 : SUCCESS in 6m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/18dde9205cad4c1483dfb50d22610087 : SUCCESS in 9m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/697edcb81fae42cda06966dc70f0ac69 : SUCCESS in 12m 52s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/8f0764f168264e92a2fc211452f5ee37 : SUCCESS in 10m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/758caf3a71984adb87460dad17808a55 : SUCCESS in 7m 30s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/cab0dc80dbe84d98bcae5ba89a27e0f5 : TIMED_OUT in 41m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/166e6b681b1b4d8781fbfb063873a44f : SUCCESS in 12m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e492541b04f64e6d838a98d78bec5eb0 : SUCCESS in 7m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/52ceb1d261d6415abd65bc4f31881a6e : SUCCESS in 6m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/33a946a251d54e6485c79d1d01d12778 : SUCCESS in 8m 58s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e68b1cba41ec4baf878ac8f185def1e2 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d3882f7ef85e4c7e8494b76cf5ed9785 : SUCCESS in 10m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b3b8f6dad82b4b20bc3bf93e37847176 : SUCCESS in 10m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1d77aadfb05b440a8852de0a2bb72280 : SUCCESS in 12m 54s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0f4d74cadef045bfa7f57c4168bccfaf : SUCCESS in 4m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7a52423c780c4e2ea1da8a993695a842 : SUCCESS in 11m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/89894525205a44e7a82e26a313be5ddc : SUCCESS in 5m 01s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/78dd377e14b8477fb54feced1fa7df5b : SUCCESS in 11m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/25ada1280e6a4e298b7078b56ff2c55e : SUCCESS in 28m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e231041bcb594c29bb52c2a3855d23df : SUCCESS in 12m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2a2b493d5d664835aa4af9651660c0a7 : SUCCESS in 6m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/018422e979a6454f9efba20a6d81f72c : SUCCESS in 7m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3fd5aaf9848047328a8ba975b36f14f0 : SUCCESS in 22m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c02091150a1042b2bc2e101df4b42db5 : SUCCESS in 6m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9cad6243cda24f34831d79c88fe98ae8 : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a219fb56ff9499b865837550ac742e1 : SUCCESS in 7m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1269496c40a749c3a3ac2e88d16ef740 : SUCCESS in 12m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f01d57b6ea74ead8345a57f0e48f9ad : SUCCESS in 1h 23m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60f658d6af0240d7a36596df60474118 : SUCCESS in 1h 33m 41s","accounts_in_message":[],"_revision_number":1},{"id":"63a7cbce80f2fa44ba7047c554fd477f49e40f82","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-02 14:00:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b88ca26ccbd511421de17a25b4d4bd37a529c934","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:20:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b59972d59bb54ad78e2efa2412fd798d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dec1a885ff2f43c1832e25d1b93799e2 : FAILURE in 10m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6d8a11600cd1beded1ffa4c72013d03b55ec3c43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 15:58:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c8224a00dce4a1ea0ab0d1f2e9d955e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1274427f17d49a08b593fa21ed696e6 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f721683906c485da3b6cdd4f6f3f238 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7e6ccd43284437c88b655244348a80f : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b01af4372d07482b99d9a348ca3c5659 : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8aa43943da5240e1a0d9d5bc20ec9b65 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b42c503582254f5ea5b6b4a2cca5ca62 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07866fd03d5d429891320b614a5a2f50 : SUCCESS in 3m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/25f833791441438ab25f2921263a0fee : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17a2053e3ea34c7f98e7002fe252195c : SUCCESS in 12m 47s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/3cca4ef673df42588c546d63b3a9ec90 : SUCCESS in 8m 34s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e20da26e8503485da27e39e876857c6e : SUCCESS in 6m 55s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/5b1198028a4a4336819a42ad91d20f9c : SUCCESS in 7m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3eef9c5bba344be6802eddeae5046cdd : SUCCESS in 10m 21s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/71ff3a4e8b4448189cad05da1a18ebc8 : SUCCESS in 5m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/75e4787b248e44b8afaa9ffd878cbc32 : SUCCESS in 6m 31s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/9574ddb914f34a2887c474029f878892 : SUCCESS in 9m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a933e11eddc4413f8e7b93ca1a77ce6c : SUCCESS in 11m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c58ba0d2c0e846b5bbcd070c34b6fe72 : SUCCESS in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dc1f8f6314474e148816555a99b7a99e : SUCCESS in 7m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c2f813fa8ff44efa8b86211520690817 : SUCCESS in 14m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e8b45f7d16c349cc8d2cdb37854a547f : SUCCESS in 3m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/533ce572ef9d480bb2129a331a6c24c4 : SUCCESS in 11m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dc0fba1e13c542b8b48edd2f29ad20d3 : SUCCESS in 5m 45s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/0cede2fb8f1642388efa84e90bb697c3 : SUCCESS in 9m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f9a1d226b7c34a3896dbfdb8a92d1b76 : SUCCESS in 27m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/88108be760a242f5b0d2176d97113a91 : SUCCESS in 14m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ddc56ae05d86456c80f0b90001e67209 : SUCCESS in 6m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8394fdc91e114694bd286b86612a4856 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c36a3b9aca2c44e4a56bd784fa165e91 : SUCCESS in 21m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/70317b22ede14e0bb5631f0d5f7a21af : SUCCESS in 8m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/659c9135b06846169e79e4b2297300dd : SUCCESS in 6m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b2325e2b6bb4b9ba29dd23e1160e1ed : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7b2c82c363024ace9180b3141af18d4b : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd923fac31384e1b8ef8e705d0483ff3 : SUCCESS in 1h 09m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/688534b55f1c4350ba0b1c3fa0b1a983 : SUCCESS in 1h 49m 35s","accounts_in_message":[],"_revision_number":1},{"id":"66fc4867c016c8bc2c6f3c7ff028507d764dd277","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 17:34:56.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 855629,1","accounts_in_message":[],"_revision_number":1},{"id":"effdb558a000d7708377ac86063d9fbdb47e1702","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-03 14:31:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0eeec9a77f5ada3f75f74016a486238ac27b46c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 14:42:01.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2361838e8c00415eb2236d99d0a82a62\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/068922ccd067447e9e3e101bd6ebe6e7 : FAILURE in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ec499c98bc4fb60b88b8cf14a5697b5df2e7b0a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 16:24:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6af27ce3fdf8479ebe954d7fe15c419c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8284f163b7345dd98b653d290ede082 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1125021baff4ac6aa5743f983e90702 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efe663115cdb4a79b7edd08f012b66b7 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14de3a8a6fcf4261ab3d0af13cea8b7c : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ca6096c51cb4f3286c82ec1b00c9979 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4433394e69e24ab58a1533f2c7f0ff4d : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3b59f3cc5c9d49798fa965c7197e63c7 : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/19d9d2e864ad4aae8de701039c366a8d : SUCCESS in 8m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7041c10166b1447cb53a7c6124d36552 : SUCCESS in 10m 57s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ba0f99cfb18c491dad12e3c691a06c43 : SUCCESS in 7m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/748d297d399d4243ad8ea889d566eab0 : SUCCESS in 5m 37s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c13f6fd11bb9401496dad8287b93a642 : SUCCESS in 5m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9a5143f1179e421b8bab0d935da294e6 : SUCCESS in 9m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5aeaa02326974a05bbfe480409f93736 : SUCCESS in 5m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a21ad9064d9b470ea626a0a166654094 : SUCCESS in 5m 52s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f41bd431a4ac46a08c293c3f7c5b8e10 : SUCCESS in 11m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3c26b626078447bb937274236d563ee2 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da5d631f0ec54d738cb3806b3f65554b : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23e6bb4babe64e1fa6eeff481a6446f3 : SUCCESS in 8m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bcfd656e7f3644f08657cb96134b7163 : SUCCESS in 15m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/68cdbd03d98646da87f25453aaa0d778 : SUCCESS in 4m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/199d1309a5df49df8b1076a306b1d2ca : SUCCESS in 10m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e2b5e97eb11f4731b7eb22457ae40911 : SUCCESS in 5m 45s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/05f93250cbd64b80b1ad2073d0ae92b0 : SUCCESS in 10m 00s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d0c8b5a56e9242fab969afe11013a6fe : SUCCESS in 26m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e433f1ef309149be936a88e4c8eea731 : SUCCESS in 15m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ce1f9707f6934c52826f4696572944e4 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1983db5f88de499f8ccd915888becfe4 : SUCCESS in 6m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fe73362cb8034b458b36197dffa1181f : SUCCESS in 23m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cc4fe1cea30a43dfaf577e4319e91c74 : SUCCESS in 8m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b1a23d0d036242be94c3576b7549d3e7 : SUCCESS in 8m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3acf3f0d11f94e9ab7a88e32c50f1ef1 : SUCCESS in 10m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f6fb3d05c71a49328172935f42f08920 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8793a63d58564081ab73698cdaf1f283 : SUCCESS in 1h 17m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a90dee4a5ae24014bba4da93f9572553 : SUCCESS in 1h 45m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d6e555c1c63d88984167cd76b4837d71f2b5095d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-03 16:38:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fc49db1cb1e6748bf9f48c4efe7b7b75ee1bb074","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 16:39:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a0befd0accf5fcb3ec19867bf21a8da231915cf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 18:20:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e1600f958214ccebe2323451f43743a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5422335cbf674edf84e2e07c5b7a85cd : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfc0da240e0749389f9d821977bd34ef : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8c3eb54ee204ddf914fd3bbcca34624 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bcc2be0b5f2483f97222aa334244436 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0a36a725cd87482f8b5f0c95aae23362 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6cfbf90b2d5846d49c095fe75ed1119c : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ba671cb2a33468ea29cde50a233dc57 : SUCCESS in 8m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/c0a1ed89d57449aabd944c092201c790 : SUCCESS in 12m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a9d5c88b4a1d40c98a509c60cec2e83a : SUCCESS in 11m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e899053b7f584f02986f7c8aa2d849fb : SUCCESS in 5m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/834589d96fab4bd7904c769bea66abc7 : SUCCESS in 5m 34s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a435afe687e646438ea630107ede7636 : SUCCESS in 8m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a7f52f9f36e44be089754b78af3f5542 : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/070d39a2195b4ddabf6daa4c97a54e91 : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b8acdc3940cf4438aae07a311d305d8b : SUCCESS in 12m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e4eff1f8dcc3468ebf2dffc15d0a1c4a : SUCCESS in 12m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b86c696b901d46d388d8c638ae1e54f9 : SUCCESS in 4m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2bd93dee09564f069da9dbd07d61726d : SUCCESS in 9m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/04269abd756341d0a3b7d4be62942588 : SUCCESS in 5m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f1ab57c5913546ec8a75f5b9b24a2d73 : SUCCESS in 39m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d2a2ea94dcca446390e263538cb443ce : SUCCESS in 12m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c576b5af60ed409baf43b64f953003a9 : SUCCESS in 5m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c75fc428a0a24fae9c9c075e1051f0f9 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b31b3ceea5b44ea9ebfd6de65ff7109 : SUCCESS in 22m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/88d78e3fe8fa4e28a587160fbc2ddcf5 : SUCCESS in 6m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f4f49d96564244ddaafb75c795f57052 : SUCCESS in 6m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/61fcde5f102242baa3496a3f1ee259bb : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/116094e5205b4fcdb7f0aba467b7b33c : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97ba8c6c93154807bc629c5923f719fc : SUCCESS in 1h 14m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06e9cb5010004f8ca4550a233151fbc4 : SUCCESS in 1h 37m 36s","accounts_in_message":[],"_revision_number":2},{"id":"277e78112624747a78631fd0d8f779682a79eded","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 18:20:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f1275c1093c3a8afd3c851aae4264dc314b53aa6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 18:21:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f05f8634dc144b995645ea3633222bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4f1cd6a2624a6da2dd278fe49284e2 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6f70b3c3d1d030863f65c96b7a281df7b31feab","revisions":{"f7dc861413dbab5f87f7997857373a023b97267c":{"kind":"REWORK","_number":1,"created":"2022-09-02 09:38:33.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/855629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/855629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/855629/1"}}},"commit":{"parents":[{"commit":"5b346cf74df6bbfc67f794c053d3c4210da4bacb","subject":"Merge \"update constraint for etcd3gw to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b346cf74df6bbfc67f794c053d3c4210da4bacb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-02 09:38:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-02 09:38:32.000000000","tz":0},"subject":"update constraint for python-heatclient to new release 3.1.0","message":"update constraint for python-heatclient to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0b75186270411911413ea9595d89949d15f5c81a\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ibb37b85b1a51ca09bf5e78ac1fd523d17b0e0dd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7dc861413dbab5f87f7997857373a023b97267c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7dc861413dbab5f87f7997857373a023b97267c"}]},"branch":"refs/heads/master"},"d6f70b3c3d1d030863f65c96b7a281df7b31feab":{"kind":"REWORK","_number":2,"created":"2022-09-03 14:31:04.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/29/855629/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/855629/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855629/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855629/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/855629/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/855629/2"}}},"commit":{"parents":[{"commit":"5a602fd50a317dfa314700bbd41ece7ae7469b4a","subject":"Merge \"Bump jsonschema to 4.14.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5a602fd50a317dfa314700bbd41ece7ae7469b4a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-02 09:38:32.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-09-03 14:30:19.000000000","tz":-300},"subject":"update constraint for python-heatclient to new release 3.1.0","message":"update constraint for python-heatclient to new release 3.1.0\n\nmeta: version: 3.1.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I0b75186270411911413ea9595d89949d15f5c81a\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ibb37b85b1a51ca09bf5e78ac1fd523d17b0e0dd9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6f70b3c3d1d030863f65c96b7a281df7b31feab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d6f70b3c3d1d030863f65c96b7a281df7b31feab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
