)]}'
{"id":"openstack%2Frequirements~855639","triplet_id":"openstack%2Frequirements~master~I22237a14cfbaba69aa476407f6984916d460ac3f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I22237a14cfbaba69aa476407f6984916d460ac3f","subject":"update constraint for python-keystoneclient to new release 5.0.1","status":"MERGED","created":"2022-09-02 09:52:59.000000000","updated":"2022-09-02 16:02:52.000000000","submitted":"2022-09-02 16:02:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"855639-new-release","meta_rev_id":"8c683da38983a319bacc63a3f1c9212b51b3d4ad","_number":855639,"virtual_id_number":855639,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 16:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-09-02 14:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-09-02 10:45:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-09-02 14:00:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-02 10:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-02 10:45:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-09-02 11:22:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-02 14:00:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"61c897a6df7267a5353d18dd9eaa43ac63cd712d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-02 09:52:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51e12b38afd8fdd8ec1fddf3056700fe16a8016e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 10:10:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/083398e933fb44dc9bd8fbc5d45c6e99\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/97d34c4e86a7441bbef833cc88524759 : FAILURE in 11m 07s","accounts_in_message":[],"_revision_number":1},{"id":"3949219ab4cd5eca6a203b2e586628269fef7748","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-02 10:45:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c918d36868b96ee2bab17b6eabe0d1e14bee662b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:22:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b24af7524a429db2a9dad5929af504\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe46930bed04ccbad945b95866ebf43 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec4aa5f96144ffeac9667747561808e : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b7b8f98cfc44774b02e17842cb7a86d : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffbd8f2b0b3144f89df89e4b4515473b : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/32f50ba5278c4fa596e60cd003b1a9de : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b1e3cf5863e497aba7cae605516d440 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb42bc8d79dc4d07b234dbe47df12096 : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9e4c5c093b194f138ea11517057b135e : SUCCESS in 9m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b7e33c736ac446d099a832e76c48f0d3 : SUCCESS in 12m 19s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/67d5ee4ee21940959e761f0efeabb597 : SUCCESS in 9m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b2b8b82303d44c13b4d265edc96679a8 : SUCCESS in 5m 30s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/01955d46ff264d99955b8721add33614 : SUCCESS in 7m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8e3ce31b4dd048c79e8b08015a9f1f6c : SUCCESS in 11m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d074937f69e74ce3a7311dd5184db5b8 : SUCCESS in 4m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e29949d228484e7ebaf1b9b94583705a : SUCCESS in 6m 37s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/93ce973134284fef959063a64525ae92 : SUCCESS in 9m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/29b206c33946475286b9fb644dbc90f8 : SUCCESS in 9m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c552952d80bd485694feb6e0c4695ec7 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6d872b9b96e4fda8ccc73f0f1d4e959 : SUCCESS in 8m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7fd0693b0b3641339d2e7a43534c07a2 : SUCCESS in 14m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/416d3c6f287e434bbfcd3e3ed26d463d : SUCCESS in 4m 46s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5b356ca98dd244f996aa05a4b6a30d82 : SUCCESS in 13m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7f6ea8742a3c43538523a4cf9c4b4df0 : SUCCESS in 5m 52s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/53436b6fc80e466d896ab57b87ec03ad : SUCCESS in 9m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ce6499ebf1994f05b5238209ddc54dc3 : SUCCESS in 22m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/402fb7c3a4f54841953505d28551fef9 : SUCCESS in 13m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/36004c6f8463420bbb1c61903c355683 : SUCCESS in 7m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/440abc7c5e2144e691569b6385a0bdc0 : SUCCESS in 4m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6020360463e849d99eaa9533c1f16888 : SUCCESS in 23m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8fce58fa77434cc78f9603d43f2aafe2 : SUCCESS in 7m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d092705f807343e09975aa77c7ced87a : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5fb4796189543a8b8f78cf9e4dd2d76 : SUCCESS in 7m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/32c89f2516ea4d96b1c272402cfeb8c8 : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/853451784acc496fbd10293b15060dfd : SUCCESS in 1h 09m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5620f626dfbb44f28e47a183cd72031a : SUCCESS in 1h 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1ab259d064486bbfb85db15e5558ed7fc7b6d02e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-02 14:00:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f470df62dc89e8dc59fc85687495d05ed5f3c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 14:01:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ec044b1279dde5280188dfb19bd5274a8eaac8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:02:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8f9855cde6d433685dff23f8b662e9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c756fb44b88b4861a7a817ac877340d5 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f4b17aac9c542d3a6c31965f60d737e : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b2c0a0d3a8449fa97b15ad9a4ca5e13 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baab3abcb7174b5ba7f1d0c4172953f6 : SUCCESS in 5m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/35d6eca971264b238264a41a4610d42a : SUCCESS in 6m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f5efb9807fb4fdd9a660c08390eb6ba : SUCCESS in 4m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1e647d61128c41bd81bfb1e34ae99136 : SUCCESS in 11m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/679ea3173fbc4976bd52f0fee026d016 : SUCCESS in 12m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/57a91f3986f1444f855b7d58d34fd35b : SUCCESS in 11m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9d270d58723a4df38d4ba049232f5dcc : SUCCESS in 5m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9572a73ba30243f4b3f53f7018ba1304 : SUCCESS in 10m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/85e6c1f0d5fe46f894a29db8a5d4f3fc : SUCCESS in 9m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3c7a3740dcb74e3ca1167d1f3fec21ee : SUCCESS in 10m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3605678256d40418d128dc9bad52a4b : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f0b040566d55401e9d1bee487abdab62 : SUCCESS in 8m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9e8085ca8faf4736bbadfaca1edbdfe2 : SUCCESS in 13m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dcfe23e92214480bbfa56dc8fdfdeb95 : SUCCESS in 4m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/102218f34fb642a1934607f478c39628 : SUCCESS in 12m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f3e9aa5920d9436abc4e76d2bf7b60de : SUCCESS in 6m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/941f82d91fe4491f8bf522a16298af64 : SUCCESS in 24m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6d6696723078412290d7151169f2a5a2 : SUCCESS in 12m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/93cd9eb7a0654cc4b70147d2da4fed0e : SUCCESS in 9m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/16bed2f02c6247e9adf61b3ca0341955 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f34d23346dcd4620888843bceac68fd2 : SUCCESS in 27m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0209956e95254723af3cccafd8eb6658 : SUCCESS in 5m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/31b424f13bde47f08b68b60c620d3cce : SUCCESS in 8m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b56f38d8f64c4b9b98fcd9d31bf46c7d : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2d1ee979d4054923a1214133a35a296b : SUCCESS in 13m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8abf361a86948bfa415e8ebe0f312f8 : SUCCESS in 1h 16m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b982ee02195845cf88275cb923c81d17 : SUCCESS in 1h 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8c683da38983a319bacc63a3f1c9212b51b3d4ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:02:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33f0cc22e90aa738d621e33be7b1eabd864f1554","revisions":{"33f0cc22e90aa738d621e33be7b1eabd864f1554":{"kind":"REWORK","_number":1,"created":"2022-09-02 09:52:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/855639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/855639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/855639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/855639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/855639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/855639/1"}}},"commit":{"parents":[{"commit":"5b346cf74df6bbfc67f794c053d3c4210da4bacb","subject":"Merge \"update constraint for etcd3gw to new release 2.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5b346cf74df6bbfc67f794c053d3c4210da4bacb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-02 09:52:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-02 09:52:58.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 5.0.1","message":"update constraint for python-keystoneclient to new release 5.0.1\n\nmeta: version: 5.0.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib490664c9db6edf9cd1b7596303ddb1fd9ea735f\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I22237a14cfbaba69aa476407f6984916d460ac3f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33f0cc22e90aa738d621e33be7b1eabd864f1554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33f0cc22e90aa738d621e33be7b1eabd864f1554"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
