)]}'
{"id":"openstack%2Frequirements~855973","triplet_id":"openstack%2Frequirements~master~If276deea824d2732cdd6c849aac3afdd544b7b8c","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-09-07 10:02:03.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If276deea824d2732cdd6c849aac3afdd544b7b8c","subject":"DNM: Validate fixes for oslo.db 12.1.0 support","status":"ABANDONED","created":"2022-09-05 17:01:34.000000000","updated":"2022-09-07 10:02:03.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2817ce388328115a6f05b39fe888c96064ff6534","_number":855973,"virtual_id_number":855973,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-09-06 12:14:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-05 19:01:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c0146d1f5f6394dea4dc444b2306f5c28b1f7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-05 17:01:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75d576e2f93af9deef0944c89480bfc894fded37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 19:01:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76cb760fe0264cb48aa22e220910efce\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/58f0678ce89c4e4ca84a3500509e091b : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/605f688043a947618fcb36cdb3d79ab5 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2ac64f878574cdaa9eeb7d6c7f9fe88 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b074798d9fb4ce6b5d7526be2d29566 : SUCCESS in 3m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dbd3f678c269485bafed96c63b305238 : SUCCESS in 1h 52m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6f66380b91acefdd864ed98b1d486c66d63e255e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-06 07:13:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5bdbdfc90500e451b405af8f2a37eae470028b82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 07:26:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c988d597810406299bb094a6a4a9e80\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/51c02d2549924f0b8c88bfccf8f38ea6 : FAILURE in 8m 22s","accounts_in_message":[],"_revision_number":2},{"id":"14f2c5c9d665255e87bbd0d558eec476c7e903b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 08:49:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eeed5c64e9804d26a4189aa84f80fe0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75074dba8a34809adf48d4e3662ad85 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/057a2293709847bb9bc38834f61ad5f0 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bb777f7ec0a40159bee4403a7741803 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cef5257114f46e99a3f14be0b027fc2 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e29f699a91224a47bf60d6dac2ca2ef5 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40bfaac32d5243b9a89043d352bd7e12 : FAILURE in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f3ac4450b2a74af094b2edeea08eb842 : SUCCESS in 3m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d5ee2c1b621442058d7e2ee86a9c79ea : SUCCESS in 9m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/98661989fadb4db1b163be35f26860c1 : SUCCESS in 15m 25s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e35426e9d13042a5b3aaff308a8bd73b : SUCCESS in 6m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/419524755e064b08ace8aac0c8f8830a : FAILURE in 4m 49s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/8b4505bb64994058a52310f306a6ea88 : SUCCESS in 5m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b28b134aeb9247c8a0aa522084bd9748 : SUCCESS in 11m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b7b2df5f026f44b2aad9edab7dc72da0 : SUCCESS in 4m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/12698ceeeb27401b83c0ca1ad4cdbcfe : SUCCESS in 4m 56s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ed57ecf3e9b4471ca1d3052b8f563ec5 : SUCCESS in 7m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c29dac7890db48fc951afff06c5e42cf : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fe874b79fb1f475bba50e7524f05913b : SUCCESS in 8m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aab6d7fc9cbe4867affff33c0e0327be : SUCCESS in 7m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6e6a2e939d074b41a0684c283e3a6350 : SUCCESS in 11m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d6c9168275b74328acbce995fa83af47 : SUCCESS in 3m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a7eb8674fd504698a47771a6100c216e : SUCCESS in 11m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2c884dcd0e57490088d311e815543aaa : SUCCESS in 4m 21s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/db8448bf678e42faa9c71fd684391ac8 : SUCCESS in 9m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dc2e9a1d075d4df4993a4bf6896bba34 : SUCCESS in 25m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4e5a8e7fbbd0459685970ec53d3a6a7e : SUCCESS in 11m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c1451fc7487c4d01bb3ca303e44ac7a1 : SUCCESS in 6m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/51af707d72ea45919a9400a5e6b28c99 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c8219de836554066a678466b930e5355 : SUCCESS in 20m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/40798ed345984697927398693b97fc92 : FAILURE in 6m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/88ec5e6e57a24949bb8ae53090dc2753 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9dcdca194dbb4f5483d1ba2e9c3c6fe8 : SUCCESS in 8m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/09cb5861941b4c7b824c6938baeb5c2f : SUCCESS in 10m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6140d755bb404727ad80709d1b478b08 : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45af5fc94f8b4e8da8823eafb6638f8c : SUCCESS in 1h 34m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4e4817c9c0d8bfddd50fd1f331a60dd0e3932d8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-06 10:35:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"01060dbdc9fb7e7cf23345903488267c43626995","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 10:47:42.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e54dc9a08384f749354faa5ac44d004\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1170789cd3954d27ba0f85f82a547480 : FAILURE in 8m 28s","accounts_in_message":[],"_revision_number":3},{"id":"70525c768849b13dc11f57fe046c41a26b08b359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 12:14:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc3623b0812545e183ab94f2d4c4185e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7224b7a4d84b819a1847ccb5de970c : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/965cdcbfa60f4f5fac68ecd7ccb8a48f : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2518583112674215be9bb57017df6d66 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d4f7470455e41fc8f35c69b07f447df : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf92d6427630436e946614af8712e3b1 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b541ab03856f457bb4e336f39daf7c3d : FAILURE in 9m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6d0dde7141e74a47921c82aafd885e58 : SUCCESS in 4m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fade6a1cb7da4b8588ba7e00bd1d400e : SUCCESS in 9m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e53e13d6bf7f4043884997b98b83818f : SUCCESS in 10m 43s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/88f01b351e2c4ed98c09da3448546bb1 : SUCCESS in 7m 42s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5d8b9674857a472d83edc11485ccffa5 : SUCCESS in 6m 23s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/a89b49ec621142a2ba707f76797f210b : SUCCESS in 8m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d6bf92b0b05847f087235b24198113c7 : FAILURE in 18m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/57f3fe1d3d664b158fb4ef4015a92bff : SUCCESS in 6m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e0fa5a50b30b4c69bc18dcc3d50f2817 : SUCCESS in 7m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6637a277378346feb933f2d12ff87e26 : SUCCESS in 8m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a41bb8b082143de9449d89f4e61a6aa : SUCCESS in 10m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4d1453bb4414bfc9ef64d06ddac3d66 : SUCCESS in 10m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c95e5d69f84b42a690d5db9bf7185648 : SUCCESS in 8m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2e7c65033e6c4b05bc5ebf612253fc96 : SUCCESS in 13m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e4b49c0dab9c4b138f99c189f554b7bc : SUCCESS in 5m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/05f01353b0f94a16b72c300136f9503c : SUCCESS in 8m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e9d2dedc078f4447aeb6806a217faaa3 : SUCCESS in 5m 46s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/2a505795b48e4bbdba2a031c2ec4a047 : SUCCESS in 11m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1e9a2f27dc74414c8c5cc923b8803cb2 : SUCCESS in 28m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d02f34450952494f886eea2fbdd61d9a : SUCCESS in 12m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8f4dc6c09a434fada6fe6167bf518d4e : SUCCESS in 7m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eb0a9876301f4ad08c32f3f8ef4792cc : SUCCESS in 6m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6c73fc0ea0334e1f96a47abe24db7b0c : SUCCESS in 23m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4a3a2dcb48664d80bfae3081444b671f : SUCCESS in 6m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/35b0d7a27bfa4147bfdf21877df2986c : SUCCESS in 7m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2a3b15a520048b4a84d5a9b276d4399 : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/84d1965d94d348128cd17d0d13edb998 : SUCCESS in 14m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b32cdb9b6d804fec9d36f3f307bd827f : SUCCESS in 1h 07m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1708f32d8164219a660b116a84454f0 : SUCCESS in 1h 37m 15s","accounts_in_message":[],"_revision_number":3},{"id":"12afbba8d0a7fddd78792f53e93a0b317ea6d10f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-06 15:00:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b5e7e60bea31c753a89f3147b348a6706a8e8b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:18:09.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ab3f7d140143f291573fafb697f7d4\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e8f3eadeb40543bd8ef4d0c4227e7ba9 : FAILURE in 15m 30s","accounts_in_message":[],"_revision_number":3},{"id":"8a3b0047281f5534002601805d346981f5b9db40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 16:33:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41bf037a3eb743aab092c37dc3d04012\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904396cd959d4bd1a17958ec11e8a350 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eedc85b14914d71aa62e10716e45af2 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5065d953c1e74be2b4aced40efbcb0fa : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0355c4c459b642c8b9f9e5efae1ceab2 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccfc22c1f390453eb32bf1b839f7eecb : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/53c3a778db2144c48d5c55422434fdda : FAILURE in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b45ce69862de49b1a883155af94a5b76 : SUCCESS in 4m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b4212291a41a4e148ab0bec83ae2f3dc : SUCCESS in 9m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e70f1360207e4cd5b883686aa67ba4bd : SUCCESS in 13m 24s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/5612771384bb40fa941343c37c5b2f5c : SUCCESS in 9m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/912425f0d01540c08aacae1a9e9e7d7f : SUCCESS in 5m 49s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/74875b13b42140769e2064dc780ccce3 : SUCCESS in 7m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d3c469e5e6e94547aff2b1dc5dd84370 : SUCCESS in 10m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3fbbfe02ebb6488e8a93855835ac301c : SUCCESS in 6m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/784fc3c9604441ce829587e94ac876fe : SUCCESS in 5m 46s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a11dfc3234c64154b58535fb0bb2af42 : SUCCESS in 9m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/530ed100132847a89c18ccd13344cf90 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/273b196859f54e3693e3fff45fbfcc7e : SUCCESS in 9m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/234e61fd28ba48af9b719b495c9464bc : SUCCESS in 8m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/263a29bb5ba9425bae86ee4cd5eb260b : SUCCESS in 12m 40s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fbe775d7f69040dab545d64558f75966 : SUCCESS in 4m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/33fb17ec070549adb7dbe37a945423f7 : SUCCESS in 9m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/24f13900562e4e7ba4ca5d6cc661ee3c : SUCCESS in 5m 08s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/7866eabad08347e890187bae9002fc69 : SUCCESS in 11m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d8f691ffde3c4d3092ff70c551eb3db3 : SUCCESS in 30m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dad2e6d0d17543f1976521399613da7b : SUCCESS in 13m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4d07136e340f440f9d725c439722566a : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2312ccaca6904477a9e3ba18d2fd1d67 : SUCCESS in 5m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/29c4d491d35d469a8c5c9abf497ee293 : SUCCESS in 21m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5f88db33e0864c0e87d30334af87772b : SUCCESS in 7m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8bb16ed008524514b6f96a87ba408c18 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d0db9f5531ca48ba83534a3f1d9f1ed8 : SUCCESS in 9m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/35bb6244a3474daebaf1a123804b23db : SUCCESS in 10m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fe86740558d448a89fbd072ab3e6223 : SUCCESS in 1h 15m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9528877f9574964a6069507f9ae85c1 : SUCCESS in 1h 27m 01s","accounts_in_message":[],"_revision_number":3},{"id":"2817ce388328115a6f05b39fe888c96064ff6534","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-09-07 10:02:03.000000000","message":"Abandoned\n\nThis has served its purpose","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5de07ec26763891fb11c7c6050012e5ea04c214b","revisions":{"e336e4b3c3df4be26c9da22cf8fab6a779375694":{"kind":"REWORK","_number":1,"created":"2022-09-05 17:01:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/855973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/855973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/855973/1"}}},"commit":{"parents":[{"commit":"6d3d20b7a7f391160eab3b22bbc6c8cf5f89d862","subject":"update constraint for oslo.db to new release 12.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d3d20b7a7f391160eab3b22bbc6c8cf5f89d862"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-05 17:00:22.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-05 17:00:22.000000000","tz":60},"subject":"DNM: Validate fixes for oslo.db 12.1.0 support","message":"DNM: Validate fixes for oslo.db 12.1.0 support\n\nLet\u0027s ensure these are all working.\n\nChange-Id: If276deea824d2732cdd6c849aac3afdd544b7b8c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/855962\nDepends-On: https://review.opendev.org/c/openstack/designate/+/855965\nDepends-On: https://review.opendev.org/c/openstack/manila/+/855967\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/855969\nDepends-On: https://review.opendev.org/c/openstack/heat/+/855970\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/855972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e336e4b3c3df4be26c9da22cf8fab6a779375694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e336e4b3c3df4be26c9da22cf8fab6a779375694"}]},"branch":"refs/heads/master"},"154eff62c560f70d60dcaa43432bb0797d4bada2":{"kind":"REWORK","_number":2,"created":"2022-09-06 07:13:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/855973/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/855973/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/855973/2"}}},"commit":{"parents":[{"commit":"6d3d20b7a7f391160eab3b22bbc6c8cf5f89d862","subject":"update constraint for oslo.db to new release 12.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d3d20b7a7f391160eab3b22bbc6c8cf5f89d862"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-05 17:00:22.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 07:13:17.000000000","tz":60},"subject":"DNM: Validate fixes for oslo.db 12.1.0 support","message":"DNM: Validate fixes for oslo.db 12.1.0 support\n\nLet\u0027s ensure these are all working.\n\nChange-Id: If276deea824d2732cdd6c849aac3afdd544b7b8c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/855962\nDepends-On: https://review.opendev.org/c/openstack/designate/+/855965\nDepends-On: https://review.opendev.org/c/openstack/manila/+/855967\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/855969\nDepends-On: https://review.opendev.org/c/openstack/heat/+/855970\nDepends-On: https://review.opendev.org/c/openstack/barbican/+/855972\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/154eff62c560f70d60dcaa43432bb0797d4bada2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/154eff62c560f70d60dcaa43432bb0797d4bada2"}]},"branch":"refs/heads/master"},"5de07ec26763891fb11c7c6050012e5ea04c214b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-09-06 10:35:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/73/855973/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/855973/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/855973/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/855973/3"}}},"commit":{"parents":[{"commit":"6d3d20b7a7f391160eab3b22bbc6c8cf5f89d862","subject":"update constraint for oslo.db to new release 12.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d3d20b7a7f391160eab3b22bbc6c8cf5f89d862"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-05 17:00:22.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-09-06 10:35:37.000000000","tz":0},"subject":"DNM: Validate fixes for oslo.db 12.1.0 support","message":"DNM: Validate fixes for oslo.db 12.1.0 support\n\nLet\u0027s ensure these are all working.\n\nChange-Id: If276deea824d2732cdd6c849aac3afdd544b7b8c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nDepends-On: https://review.opendev.org/c/openstack/aodh/+/855962\nDepends-On: https://review.opendev.org/c/openstack/designate/+/855965\nDepends-On: https://review.opendev.org/c/openstack/manila/+/855967\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/855969\nDepends-On: https://review.opendev.org/c/openstack/heat/+/855970\nDepends-On: https://review.opendev.org/c/openstack/placement/+/855862\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5de07ec26763891fb11c7c6050012e5ea04c214b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5de07ec26763891fb11c7c6050012e5ea04c214b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
