)]}'
{"id":"openstack%2Frequirements~856044","triplet_id":"openstack%2Frequirements~master~Ide8a99cc536f5f6a6554718b2c446e6b25c74d18","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ide8a99cc536f5f6a6554718b2c446e6b25c74d18","subject":"update constraint for oslo.concurrency to new release 5.0.1","status":"MERGED","created":"2022-09-06 11:57:42.000000000","updated":"2022-09-06 18:11:30.000000000","submitted":"2022-09-06 18:10:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856044-new-release","meta_rev_id":"d8c60cef6a13ae2ae3735000d89f8ac461b7b56c","_number":856044,"virtual_id_number":856044,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-06 18:10:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-09-06 13:22:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-09-06 12:08:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-09-06 13:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2022-09-06 16:08:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-09-06 13:22:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 12:08:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-09-06 12:22:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-06 13:00:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-09-06 13:22:18.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-09-06 13:45:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 16:08:38.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"a8b3a4f6dd42b6117d48d6f5dc6bbed71f0bfe6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-06 11:57:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f3c1f7e627497015741204fb0116615f7df73d3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-06 12:08:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe819bfb78c6a510f35eaea6c0811a75d43fe076","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 12:22:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a129eadb46914bfd92e5febdf92c82a8\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2a22545ae384415fab362c9b5c8d540f : FAILURE in 20m 39s","accounts_in_message":[],"_revision_number":1},{"id":"30c4636406f24b6824220a0f1d1f32a560e61dd4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-06 13:00:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"399e6e3bd791ccee51e963a9be727bca6bbfb21a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-06 13:22:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a9503c8458d8176d2f09012251fb381258cfed9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 13:45:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/712c25a31f614866b991206f50143493\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9478a11b48a6446ba10ecdf5bbbb8069 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b5883e36e94edd855be815d6fa015b : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7c99737e7d146d1b07c8cfb61f3ae9f : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/421bf43ecf0d4e1f99e3f4a81b1f243a : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6094c20225764ac8bb897d7683ad219e : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/236e2cb677b843f3934c40ac91bc9480 : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5730ffb3f3af47458c145bd63b10d7e7 : SUCCESS in 5m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f5bcab1e34684dc8903b143acca75ec5 : SUCCESS in 9m 26s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ef1cdce7bfc74ec785a8554de2350064 : SUCCESS in 14m 05s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/dbbfa882b47742fd91128489737d5900 : FAILURE in 4m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6544adca732e4c9bb7da3650057ca1b1 : SUCCESS in 5m 52s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/251944e420784cc785e6a7554e5d5b7d : SUCCESS in 6m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5dfc578c356446089ba886f03dc71acc : SUCCESS in 10m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7570eaf23b0b4440ad71661c3ca3cb22 : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e70670c4a78e4a87b145d5a3cbcdd120 : SUCCESS in 6m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/08d7afa6d4c24161a6c7e221965d35fe : SUCCESS in 8m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1aee565e8d874b6eb2761a58d37b1fe4 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f95e4014fa6544a29704597442907464 : SUCCESS in 10m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ad62051e26f94141af914c037b730127 : SUCCESS in 8m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1c2d3d49e2864e5f90375d2bd1d715e7 : SUCCESS in 13m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/82c52df977b04095b8184f747f60ab09 : SUCCESS in 5m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d2ae79b2bc43401caabecf26026ff62b : SUCCESS in 12m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9c734bde4dce41698cf326eca59e27b7 : SUCCESS in 5m 48s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/0f915309bbea48f0a8f67368d1fec6dc : SUCCESS in 10m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b16a12c753274b41a181b3eb1a407097 : SUCCESS in 22m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f33eb0310a334a5a94370540afcfe7fe : SUCCESS in 13m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f16e966f9e00464e97632f6b3ce239ce : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/447314397707453c8f37578344175497 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a4a72c19d5ca41c89d2ffd218ae3a6a8 : SUCCESS in 23m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c4ece78b2ae94eef88efafe20fc8b70d : SUCCESS in 6m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4991548c2fb042ce8f279a34ce527ab7 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71641b03ba9040899a5c555d4c9cde6c : SUCCESS in 6m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/41e2ed852a3447859ce5ef01aff79a4b : SUCCESS in 10m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8d6ceda83c947cc96ee57998e2bf4d9 : SUCCESS in 1h 04m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/496d45c9fb5a49f49746e57efa33df3d : SUCCESS in 1h 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0a5fde323043d8b22a51f2754d4efd447e7f3d7f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-06 14:01:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"516969a7e18402c21d526789d03a589cce43edcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:13:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c6334bc97cc49219df8bd525ec22130\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b69c67ee5e7a471f89d0265846a19b29 : FAILURE in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c94181a5b1fe8fd2037e0ad5c766e3c8521f1544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:47:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b661f822e546402d8d2afe568b9f9a94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de8e6a3c5a641fe950fb44d8badbc9a : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bc24ef9b5f34f90a98cbbc0359a4e55 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c24e11c84bd54ff2a95a665aeda78afb : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22ea8133dddf46afa6dc89fe2b7067b2 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fffef0f64eee4bc7bd26d196fbb8f30e : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7c85ba9f471642d196e5e73ace536a34 : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5ee9d17232b343e89a30178daa10859d : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/aa8efe5de73645eca7a2a3d8c8e69917 : SUCCESS in 14m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ce9d1cf439f7413ab73e41474f4fa4c5 : SUCCESS in 14m 18s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/0b1ad48ae0d348d691c92ae6db8d43fd : SUCCESS in 6m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2310930d81ac4a9f98b2fddd1622ea84 : SUCCESS in 5m 56s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/6372908da94842a6b6258c1e5d58d8ee : SUCCESS in 7m 43s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/558606144fe24b6ea61be802d0ec0abd : SUCCESS in 10m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/624a266ed82a43b4827120f3d8976023 : SUCCESS in 5m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ff2b9a05025f496bb0cc4bd83db06d96 : SUCCESS in 5m 56s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bf216daf87b943ddb18f03152f06673d : SUCCESS in 7m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/09d9ebf23b7842e4a69d373a9e1dfac1 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9629bc81765c47589a410ee76c7598bd : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/883df8f6a20845a9bd1f0c171797c196 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/299f2ad6b1c5480daefb34bcefae1467 : SUCCESS in 12m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f356f2cce5354e7e9368285266d87e96 : SUCCESS in 5m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d938be6c5e9d4c83a43b77e806591fed : SUCCESS in 9m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/125d4e9a97974463a3ae649c9e48d8e4 : SUCCESS in 5m 10s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/8e364ec9aa084512aefce8072e4ba148 : SUCCESS in 10m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c4d7277a891c4b32947c77f9a8cef07f : SUCCESS in 30m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3584c770763c4e7ca7ed24d8d1c3e056 : SUCCESS in 13m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5fac0a0a74e54c6285105bc0901880eb : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/da8b058318774e8baf71965ff9de8a87 : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c726156b996244d5bfe595d0107dcb27 : SUCCESS in 23m 10s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/489a5d4e112648588824065e56a27a34 : SUCCESS in 6m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c472b5995c9245d7b6f7887423beffbb : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e6ea56adf2340a795babdcad047102c : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/03aa2e1fced145e3907591e90bbfbd8e : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/318cc37ece9a449f9125dda1b90f048a : SUCCESS in 1h 20m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e80e0c288d54bafb7d9c27a7ab6e920 : SUCCESS in 1h 39m 06s","accounts_in_message":[],"_revision_number":1},{"id":"b07506a864a81d3d7752a596abae34a21e6d0774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 15:48:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b83e1aa2fbfc211b4f9ddd3fabf46738700cc6dd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-09-06 16:08:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"436c62bc0cf94e4049eb92ae618d08061200e47e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 18:10:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b35129bbc01e485bae8f3865081f9679\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebea37f7e989484ba0ef1d157384a621 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0d5a87f8a64a1699b81019291f2edb : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed68c37c42324c298a1e674cda05790e : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2407e3f81a7b43c98add12efb734bbc3 : SUCCESS in 5m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58bf3a506b284549aa1791aed2321c1f : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a33d181b3988465fbd61012d19f6df05 : SUCCESS in 4m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/52539c44c7ce42569d711685101fc0cb : SUCCESS in 14m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fb52d74e11704ab5981271520a26197c : SUCCESS in 13m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7ed560d30d43488a86653b9d552ad6d2 : SUCCESS in 9m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6bfb26a9c5ee4236afa4bed90f98b02a : SUCCESS in 5m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d1f0c82c5f494300a6e82a114d366711 : SUCCESS in 5m 48s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d652a255fa414f6fa37cea58719926c1 : SUCCESS in 8m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a2a316edf20462593334b30f12c742b : SUCCESS in 9m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0ee438a4e184075af89b6408b56ba29 : SUCCESS in 9m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3c6c850df8794ca88b2a360432afca61 : SUCCESS in 9m 58s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/549e2db65ff646a88852c18a251f74c0 : SUCCESS in 15m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e3ef216acb604f6eb1d404ec8fb261c5 : SUCCESS in 3m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b78493d3b6284049bcf8c73eacb2241d : SUCCESS in 14m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5b70d785a76244888111b2116baf872a : SUCCESS in 5m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f1dc58a88f0143dea9b8b511f3b92f70 : SUCCESS in 21m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fb4f827c10a54ea0b6f731a9ca868bb5 : SUCCESS in 20m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f828236c6797457081f5170db3d1c3a3 : SUCCESS in 6m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/79c55e53469a47fd8389d3ea77d2fd8f : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7214cee7b8664cad9de29c3b59168ec2 : SUCCESS in 25m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/120408fe6f744ae3b0bd564b1870c657 : SUCCESS in 6m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/707c7f9c70ab42a1b35b57c8dc6f2306 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/68a78eaf929c4d6aaf163fabd134fcab : SUCCESS in 7m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/669213fda4f9496c989c813f37f578fa : SUCCESS in 10m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b020fa33495f47438ccb37e98e006d3a : SUCCESS in 59m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90ca48f9d2404bff9c7dddac2f08e0e5 : SUCCESS in 1h 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6eb7163b9fc21450529c56e680541938dcc6296d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 18:10:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d8c60cef6a13ae2ae3735000d89f8ac461b7b56c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 18:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b9c5a1ecdc94f3d93e25ed4457113de\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218d089218ed4b01ba7efb01f7761519 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b66fbd0f98897c12f4b6e9361e44562894a1032","revisions":{"1b66fbd0f98897c12f4b6e9361e44562894a1032":{"kind":"REWORK","_number":1,"created":"2022-09-06 11:57:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/856044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/856044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/856044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/856044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/856044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/856044/1"}}},"commit":{"parents":[{"commit":"3c79926e7c1478d0f4798c413600152b1b17f23f","subject":"Merge \"update constraint for keystonemiddleware to new release 10.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c79926e7c1478d0f4798c413600152b1b17f23f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-06 11:57:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-06 11:57:41.000000000","tz":0},"subject":"update constraint for oslo.concurrency to new release 5.0.1","message":"update constraint for oslo.concurrency to new release 5.0.1\n\nmeta: version: 5.0.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iba860bc47cd40a4b61a8e77c6aaa361b9dff1299\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ide8a99cc536f5f6a6554718b2c446e6b25c74d18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b66fbd0f98897c12f4b6e9361e44562894a1032"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b66fbd0f98897c12f4b6e9361e44562894a1032"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
