)]}'
{"id":"openstack%2Frequirements~856097","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ib7689f7388a6e98c33913c119f092e6ed20371c5","project":"openstack/requirements","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"8161":{"account":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"last_update":"2022-09-06 13:45:24.000000000","reason":"\u003cGERRIT_ACCOUNT_8161\u003e replied on the change","reason_account":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"}},"30566":{"account":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"last_update":"2022-09-06 18:25:04.000000000","reason":"Change was submitted"},"33081":{"account":{"_account_id":33081,"name":"Jin Hase","email":"hase.jin@fujitsu.com","username":"hase1128"},"last_update":"2022-09-06 18:25:04.000000000","reason":"Change was submitted"},"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2022-09-06 18:25:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7689f7388a6e98c33913c119f092e6ed20371c5","subject":"Update constraint for defusedxml to new release 0.7.1","status":"MERGED","created":"2022-09-06 12:59:23.000000000","updated":"2022-09-06 18:25:04.000000000","submitted":"2022-09-06 18:25:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856097","meta_rev_id":"f1fed8b4c58d1f6cb52160fbc9a0ba8303829a7c","_number":856097,"virtual_id_number":856097,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-06 18:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-09-06 16:07:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-09-06 13:45:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-09-06 16:07:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},{"_account_id":33081,"name":"Jin Hase","email":"hase.jin@fujitsu.com","username":"hase1128"}],"reviewers":{"REVIEWER":[{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},{"_account_id":33081,"name":"Jin Hase","email":"hase.jin@fujitsu.com","username":"hase1128"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 13:00:56.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":33081,"name":"Jin Hase","email":"hase.jin@fujitsu.com","username":"hase1128"},"state":"REVIEWER"},{"updated":"2022-09-06 13:00:56.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"state":"REVIEWER"},{"updated":"2022-09-06 13:00:56.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"state":"REVIEWER"},{"updated":"2022-09-06 14:29:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 16:07:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2bdcffcb1ef2f4b34b9f54b614da54a9e5fd6bf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2022-09-06 12:59:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df8e12a64f3621047bf407ee6a604f3858a7d6a3","author":{"_account_id":8161,"name":"Lon Hohberger","email":"lhh@redhat.com","username":"lon"},"date":"2022-09-06 13:45:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"22224bb8895530e6eaf1f4da89ab1bddb51ce5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:29:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd0cc9a75f244985ac3a092bbd3bb4a4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7746edc468f044fabd6ae46d725d79b9 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f624c2590274d079db327eb08d8aa2c : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cf50adc8422483cb94637823e793dff : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a1f4bc5fcc7401db7c5225dfef7ad43 : SUCCESS in 7m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ceb63144522a462d831e1cc2b730e90e : SUCCESS in 6m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/08ceb0dfbaa14ccbbb769906739c94c3 : SUCCESS in 6m 03s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/95b114e2fcf94089a7ee32d43eea159a : SUCCESS in 16m 40s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f753bde1ba684058a17bc4bb997b35bd : SUCCESS in 19m 51s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6dc33dfb511e4d6fb1340da777c45deb : SUCCESS in 9m 52s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/fad838874ae2465c8693d4f9ff69543c : SUCCESS in 10m 50s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/150317a1c4604af0b6f015d77c4cd2c6 : SUCCESS in 5m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/306ef3d9f9244e7f892c87a268249c1a : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/1ccc47ea80864075b356a3377e27e64f : SUCCESS in 13m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7b2eda4397eb460b9432f7489cf293f3 : SUCCESS in 8m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d3203e95167c47c88b6f4aa92a59679d : SUCCESS in 7m 48s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/dc28b13c90824de8bb16274c9a6d3d35 : SUCCESS in 15m 38s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5e01448f95294553a9dc620f7caab17a : SUCCESS in 21m 34s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/952c9c646aa3469eac4ecad07fd278cd : SUCCESS in 23m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bbd769877c674405924662960c5982fc : SUCCESS in 26m 36s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/650f2fdc2538437b8e92d4f207496eca : SUCCESS in 11m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a673a39008f342e5881b5be2a0a9c48a : SUCCESS in 17m 51s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/67f45541a4e042239247b2a8be062da9 : SUCCESS in 17m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bd8d052a06cb4beca719b8b05c02a1d8 : SUCCESS in 25m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/587782c631f442a9aa469ea3c72485d6 : SUCCESS in 6m 12s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/5b916879d0dd470d91b825121d4ab973 : SUCCESS in 8m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/69c6689ed533431581d53c9c2189cc47 : SUCCESS in 8m 32s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ce8194608b54746b3f66d6f67988aae : SUCCESS in 53m 25s\n- tempest-full https://zuul.opendev.org/t/openstack/build/75c3784e049146c68ba014f8abf2242c : SUCCESS in 1h 28m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8adac580b27c4809a9e1197a3d581c2b : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":1},{"id":"21b4841f27f13cebb99f513c07c2afadab67d32a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-06 16:07:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2880e1f724bd8fffd03f9a2c752a049d5acabb9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 16:08:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1e43c455f341b883d8e532aa7b40c3fcefb7ec0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 18:25:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e873df8aedc4eb5baa7b7b8b93ca91d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/993597435fdc4de9a61d97d8459e9a47 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69dfd2d4769341ba800f4dbe0f94fb62 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32d1c2c31ae141bbb5449c477bb16eda : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0567555999946f0a7c094b7ebe1b9f4 : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c8888b942c44adb812b582c6151abb0 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/235d5002c54a4804b79308cf000b95ec : SUCCESS in 4m 08s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7e0f40a2a1c442838ab862a8e46b779a : SUCCESS in 9m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d6c8f52ffa3f4687931ee5dc4fab5e12 : SUCCESS in 10m 04s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/9c9f5bb5a2be4e9898edd8d4e0d627ef : SUCCESS in 10m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/898bb601b3a844b8b845313e7e9a65f8 : SUCCESS in 9m 43s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/d28e76761c704208b49cf12c32938416 : SUCCESS in 6m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/64d357e0d7b649b9949d732008c9a58e : SUCCESS in 5m 51s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/0db3467848664adc832db2167c6ecdba : SUCCESS in 10m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/96affe1999d04a2688c5b2876267c78d : SUCCESS in 9m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d412c6fe6e3a4fd1a0ee9cf3b6df177f : SUCCESS in 9m 09s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7d625c2d99ff433dbab7ed54d8d09c41 : SUCCESS in 15m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ba955977797c408fa113fc25fe647cc4 : SUCCESS in 19m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fcdb72f8be404ad0a43c14d9872c4883 : SUCCESS in 21m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/df672dd1098e48c0b647d7e0ef25e280 : SUCCESS in 19m 21s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/38890b791f2843d6a858ff0734326767 : SUCCESS in 15m 32s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5b5ccd0250b542e0a55d061549b58c5a : SUCCESS in 13m 53s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c5c556070fab4acca5187c4a1a012ab6 : SUCCESS in 15m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb30bc1565c049d4886a3af27c4d402d : SUCCESS in 18m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/23dda4009fc54b7a8c0b2b9910319c31 : SUCCESS in 6m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b25b699837764f2994f5e2f63e0e9f75 : SUCCESS in 10m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cfb06ccb86a54a2b807c8cb90d0e2622 : SUCCESS in 12m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e32d0614f60a4c7cb8f30dcd2830e920 : SUCCESS in 1h 01m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4f71a0bb3bd14a11b68bf3a56b5610bd : SUCCESS in 1h 30m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/85a61d6e23414ea2b4c3d3cb18374153 : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f1fed8b4c58d1f6cb52160fbc9a0ba8303829a7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 18:25:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d80d5d4befb1d36b5e08c7293bf9d5d5edeafa1","revisions":{"4d80d5d4befb1d36b5e08c7293bf9d5d5edeafa1":{"kind":"REWORK","_number":1,"created":"2022-09-06 12:59:23.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/97/856097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/856097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/856097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/856097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/856097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/856097/1"}}},"commit":{"parents":[{"commit":"09fcb76336303cd62e15e49eadc27ec32f3e9122","subject":"train-only: Pin setuptools in Python 3.6","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/09fcb76336303cd62e15e49eadc27ec32f3e9122"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-09-06 12:47:36.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2022-09-06 12:49:55.000000000","tz":120},"subject":"Update constraint for defusedxml to new release 0.7.1","message":"Update constraint for defusedxml to new release 0.7.1\n\nscciclient projects backport a patch [1] from master down to train,\nwhich requires defusedxml\u003e\u003d0.7.0. Bumping defusedxml should be\npretty safe and it still supports py2.7 until 0.8.x (deprecated\ncurrently) [2]\n\n[1] https://review.opendev.org/q/I8ff057ee64c04c4cd5c92abf3e31b52c6225ed76\n[2] https://pypi.org/project/defusedxml/0.7.1/#defusedxml-0-7-1\n\nChange-Id: Ib7689f7388a6e98c33913c119f092e6ed20371c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d80d5d4befb1d36b5e08c7293bf9d5d5edeafa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4d80d5d4befb1d36b5e08c7293bf9d5d5edeafa1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
