)]}'
{"id":"openstack%2Frequirements~857929","triplet_id":"openstack%2Frequirements~master~Id726558b62ca3cdfbf4282004ca9c54793265d0d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id726558b62ca3cdfbf4282004ca9c54793265d0d","subject":"update constraint for tap-as-a-service to new release 10.0.0.0rc1","status":"MERGED","created":"2022-09-15 16:02:18.000000000","updated":"2022-09-15 21:22:36.000000000","submitted":"2022-09-15 21:21:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857929-new-release","meta_rev_id":"a896de2ed8b9daec1b0d0c013b9879523ce05a38","_number":857929,"virtual_id_number":857929,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-15 21:21:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-09-15 19:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-09-15 19:25:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-15 16:14:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-15 17:38:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-15 19:25:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e3a011de23de53dcb74931ec2c8512e62fe6446d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-15 16:02:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8c72d609a82122c7765b2737e974b560928871a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 16:14:52.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49fd7224082443869ffb468432b20339\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1d50a23369c54493a5e49a093dbf88bc : FAILURE in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"98d0c6b5e1e4464fa55a54b6884fbf39ea8952db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 17:38:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489cbebd08624397b05ebca50790d908\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e564257ba27442618d3b7911a1660db6 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3da73d65c254d54be3931185144fdef : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdbfe44c03834d8886d871fb15635aa0 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cf0f2cf4569482588b68a1540ff79ee : SUCCESS in 8m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14fad6cc2043486889e36e1a2f5db2c1 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27db399d21eb48a38eb131281005dc2e : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1739995983244140b9c04dcca959b96e : SUCCESS in 5m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1d98c67d8d02430ebb70452255d66cf6 : SUCCESS in 9m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ce199d26c5964ca692a4a35636160953 : SUCCESS in 12m 28s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/541a2f08594341289338c8fee88ce673 : SUCCESS in 9m 10s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7ce4e639e20546e5ac0b418679f32b40 : SUCCESS in 5m 59s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/01558f379c294fc2aaf64524b78e51a3 : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3f0d4e2e59ac4fe99929af955daa8b6e : SUCCESS in 11m 53s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b990b1c197a848eb9df0e2bd25242370 : SUCCESS in 5m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d2ac035b0d2a4e0f8acdfdf2d53640c9 : SUCCESS in 6m 11s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c49378e2624148dd87746f68f0622d20 : SUCCESS in 9m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/70e3f4e523fb4729af2325f49798fd6d : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4e8f709f735e475f95187285353e2163 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3f5abcb36b234077a3da11ebb93da1ad : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7fb8deaca4464910b213a1c7ef5c861b : SUCCESS in 12m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/310c969b119447159dea11e1de142db3 : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/379399abf69c4597aaaa52517a70311c : SUCCESS in 11m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b0293743399f4472926d96ab4c04522e : SUCCESS in 5m 20s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/826ce6b9bed64bd78ca4734b0b1a7be4 : SUCCESS in 10m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8c7647e22b73497082ed8a955da93b73 : SUCCESS in 25m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e83a53d379314d738b850cc030aaef84 : SUCCESS in 13m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3c5562d998fd4eec9c47776326848f13 : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1e924d88806e47848d08111d9234b54d : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0ab7de9ed04405484792e5345cee801 : SUCCESS in 24m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7f1b6790e13648cf9d2806185a26fd53 : SUCCESS in 5m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bafa51ee43ed4aadaacab8206c879a8c : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/447e0f6ed62743049253bdae6859264a : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3f6aa59249274713adfe53927e128862 : SUCCESS in 12m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/012bbf6d67ae4eceb4250ed4258005c6 : SUCCESS in 1h 18m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50070402503948f6979e952651a3fdf5 : SUCCESS in 1h 28m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ca00e44940835aa07c49d998ab41deb646459a91","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-15 19:25:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baf49b2d17c2d2596e59f1303ab26e5be944fc72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 19:26:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c779e09645e873a62025763f433fd132c1ce63aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 21:21:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e625a199c8544eb6856c928f93d3bde8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2be534ad3b446208177a633dd019f44 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e37f925cca4a25b990b06afd8938ab : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/175a3a0d8d8c45ceb1054aa9bdd2f463 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db1e1bd01f3418495d51f035b0f029a : SUCCESS in 5m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb9fe1a80ebc438cb9edae3e4851660f : SUCCESS in 5m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1c92be72b43d40e8a28bf7f046d4a405 : SUCCESS in 5m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb62ccc0bfc94a88b16f1edc55784d2d : SUCCESS in 9m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4720d6b457cc48669086f0807f687ba8 : SUCCESS in 12m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/73a9324e9bc24010a7cd54227903060e : SUCCESS in 11m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3f68fd459e444d6cbe663dc0d6bf623d : SUCCESS in 5m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/818bb003857149e786880f37ab40f814 : SUCCESS in 5m 55s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ebab7c217bc74de68cdc5319fd6c9a3f : SUCCESS in 8m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fd02663b555a4f0abfc692bf955884d9 : SUCCESS in 9m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/219268420cfe405dbfd7a033448f182f : SUCCESS in 9m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4d188cc5341b4ab1a3bad3b17968a66e : SUCCESS in 9m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4161e96b4c9c42f6ac8b91f6086be634 : SUCCESS in 13m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/00d376097b834c1aa877bafe4ea9e55d : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b8a2ff08e15f4eaab813f4ed48264997 : SUCCESS in 10m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/030b98a18b7d487095abc31ccf6abe29 : SUCCESS in 5m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/30b13c49fdc644a4acabb04b1112c90a : SUCCESS in 23m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2550dc92ebd54e689c2c43366a9a6df9 : SUCCESS in 13m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6e4c61ece9a2473fbabdfd667ac47dbc : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a76d2ff92c54e298ac6df1a051fa6ad : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9730e0ee5a724365b49468ad5319e32c : SUCCESS in 22m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6b2ec2e92b544b9fae079b96541e5794 : SUCCESS in 5m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/97e6525e84ec4c79ab6ed963246457af : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/91c0243359cf4a2eb14e4d0fd3976139 : SUCCESS in 9m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5f830bbcb30f466ea5945f1a674252ac : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67edd4d82e7e413d8fa3c24eb98ec848 : SUCCESS in 1h 29m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78782fd122a145de869e596bf8e062c2 : SUCCESS in 1h 50m 47s","accounts_in_message":[],"_revision_number":1},{"id":"560e20689a113e042ffc81d1b354e8354a2cf0ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 21:21:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a896de2ed8b9daec1b0d0c013b9879523ce05a38","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 21:22:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa5ecfea25f341568b28330b286354a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfef66dbe204416696830be3eaf23686 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0e7c6b8e5379d9f17541426e9df449d56bbcf4f","revisions":{"b0e7c6b8e5379d9f17541426e9df449d56bbcf4f":{"kind":"REWORK","_number":1,"created":"2022-09-15 16:02:18.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/857929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/857929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/857929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/857929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/857929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/857929/1"}}},"commit":{"parents":[{"commit":"023a6b7f0796a9910ce93aff88fc4ffaec14fd49","subject":"Merge \"update constraint for ceilometer to new release 19.0.0.0rc1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/023a6b7f0796a9910ce93aff88fc4ffaec14fd49"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-15 16:02:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-15 16:02:17.000000000","tz":0},"subject":"update constraint for tap-as-a-service to new release 10.0.0.0rc1","message":"update constraint for tap-as-a-service to new release 10.0.0.0rc1\n\nmeta: version: 10.0.0.0rc1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release candidate\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: elajkat \u003clajos.katona@est.tech\u003e\nmeta: release:Change-Id: Iac1d6c8ec6b2e08d932854999d3818163ad41337\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Id726558b62ca3cdfbf4282004ca9c54793265d0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0e7c6b8e5379d9f17541426e9df449d56bbcf4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0e7c6b8e5379d9f17541426e9df449d56bbcf4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
