)]}'
{"id":"openstack%2Frequirements~858088","triplet_id":"openstack%2Frequirements~master~I27f036a260fe569d16f6769f58874ccdcf56d420","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I27f036a260fe569d16f6769f58874ccdcf56d420","subject":"update constraint for python-saharaclient to new release 4.0.2","status":"MERGED","created":"2022-09-16 10:07:27.000000000","updated":"2022-09-16 17:16:30.000000000","submitted":"2022-09-16 17:14:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858088-new-release","meta_rev_id":"ae010872aebd2eb33fc635b4df16ed442a9fbf59","_number":858088,"virtual_id_number":858088,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-16 17:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-09-16 15:17:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-09-16 11:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-09-16 15:17:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-16 10:23:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-16 11:13:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-09-16 11:47:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-16 15:17:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7d092ef9320983b4b7f6c6b99ac5376b7c94fd3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-16 10:07:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2b8a7c663e0524802327d92dd761d15488cc020","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 10:23:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d09c726c3349748985c41676b94329\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/157f56eac8cd4b678a1dc326162cce30 : FAILURE in 7m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d34215171376be297afea3d731234414eff8c9ca","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-16 11:13:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"14757105271b95945aa506a32c539a261d61a6d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 11:47:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5db8699e45cd448e9195da8119ff3774\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7eae37ff32475d889f13d16e36ef74 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2632930c0a7c439f9bc8374edb5866da : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b1532aaaf534357ad16534975088055 : SUCCESS in 7m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/402c3090454349a491ec71df6e4544a0 : SUCCESS in 17m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78f1f37030f14c59b7f3f9465ab0d6b8 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9fef4d24990e489cad507669953aa137 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c448478cd08f4f509f549020bedf2f9b : SUCCESS in 4m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/efb97724d6914b47a46efb2331244c56 : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/eeb64710108641a5b7429678a3975433 : SUCCESS in 14m 48s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/38950f427838479cb7a1b4456fee68fb : SUCCESS in 7m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8facc69bb31141049afc1d79ac812780 : SUCCESS in 5m 44s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/52dffb9c518140e3b29b251657876203 : SUCCESS in 7m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/27d51f8af67f437a8522acd17e3ec8ac : SUCCESS in 13m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b5080997868a4c1ea13ac2c34cac9426 : SUCCESS in 5m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dad3d226064e4145a7c921fc157c09dc : SUCCESS in 6m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c0e1cdad8dee48248e189df231e834e3 : SUCCESS in 11m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bcfb84efddb24246a23a40a19daf1716 : SUCCESS in 10m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff31139512ae497e85700497a9d237e7 : SUCCESS in 9m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c2eb63caefff475e8524495baf061379 : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8a456bbb603c43bf8d543a97d9e367b8 : SUCCESS in 11m 54s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b7fbccf41f44466e96eb13350485b152 : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9df04d2389884cae994fa4fb048f1d7d : SUCCESS in 12m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/aead20845a004c9e808d34f4cbfbd2d6 : SUCCESS in 6m 40s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/562b0d52876940a784def80918b0b4ca : SUCCESS in 10m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2d01168c04a14f15adec97ca98c56f14 : SUCCESS in 24m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ae5506ad028042dd9f4fc98f5315b81b : SUCCESS in 11m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/efa2847b09334e089292677cb4dd46f4 : SUCCESS in 5m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1408eed33b6a4f5cafdc0287db50d252 : SUCCESS in 2m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/66e9573687b04883be4e85f6750069da : SUCCESS in 24m 42s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fa1a74a292a6470795d9e324ca23a9ca : SUCCESS in 7m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/cb5287bb74324edd92ae39cdf62f330f : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4d8c8983046b4a6e9b398ad9a3257560 : SUCCESS in 7m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d2d14587483e4f6a8bea0e12efa978c5 : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3acf72d6f66047638021c6932d3e4ef2 : SUCCESS in 1h 22m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/301046761e714a98ab666a85ab14b458 : SUCCESS in 1h 28m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2ca869c7981b12add70e3762345582d77c895b5b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-16 15:17:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dba46fe5a534d3f11026a8c50b13d77e0e6404b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 15:18:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7521d1c351378b028488f474c1986a4faaf7aa75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 17:14:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2434726802fe4e89bdc111dfe44afd20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a2a11881a74666a1f592c75a5a5b8e : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b435e47dbf4316a68ec7aa81b502cd : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5163aa1d11e49dfbfd42c5b76606567 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/261589ab5f774d5e95aef741c8ed3b04 : SUCCESS in 7m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fd05f6d9005e4832bef3cda6637af290 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72166dc3a38541efbeecbee7e312794d : SUCCESS in 4m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/525be734930c4be5959a95caea34ea87 : SUCCESS in 9m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fafeecd21c3f4c56b83be2b6f533ba2c : SUCCESS in 14m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b0c2900d84d24d1fb28c6d787158042d : SUCCESS in 10m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/484e027186e24ba78e77b5a46703b6ff : SUCCESS in 5m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/226e57a26fac4af68cdf2c92732f08ba : SUCCESS in 6m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/49a1f97f1c904d74bd3a8fef6a5ea6b1 : SUCCESS in 9m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1493d5b7ce284e11b7c665fa662c2592 : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/baa3288f78d24fe69f95aa593441d610 : SUCCESS in 9m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/27447b6a28214e9d81dd962e8dfb03ab : SUCCESS in 8m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5136aa97721d4004aa8aa86eff930f4c : SUCCESS in 12m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d5ac4b4e0e5d4143a1b2276771e8fbbf : SUCCESS in 5m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0f41e5a9c61d4afb848ff36f4663534e : SUCCESS in 9m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b7dff2b93aff48978143c5b481660909 : SUCCESS in 5m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d87a68791dde449fb5058a492c6b8746 : SUCCESS in 25m 14s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e61a97d59cb94baf9b9afed80520d075 : SUCCESS in 12m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ccaf47a88bc34aa8884c4ea35c952a97 : SUCCESS in 6m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e4b06dc269646c68dddea76759af6e0 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd916c1b714d485f9212aff14c22ca26 : SUCCESS in 21m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5204f6f0164443e99f83d5d3360d9c80 : SUCCESS in 6m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/806386856f03470192ed7d110924fa1f : SUCCESS in 8m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ccf29369bf0543bb95175a877c31b34b : SUCCESS in 7m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7bbf5f5134e3474395604ae46dd8ee36 : SUCCESS in 11m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95ba8b297edc47ad97fbed2f14951a53 : SUCCESS in 1h 06m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ade8cf87a78443c88eb4ad3a40ef1be1 : SUCCESS in 1h 50m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c86c9dae40a1f8c900cc051c0ccdb16cfc313462","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 17:14:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae010872aebd2eb33fc635b4df16ed442a9fbf59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 17:16:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e67244f0dbef4cb49f0ae3569a4062f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515ded668813429f995bed6602091fdd : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"939121e011518809468e31dbbf3aa37e5160560b","revisions":{"939121e011518809468e31dbbf3aa37e5160560b":{"kind":"REWORK","_number":1,"created":"2022-09-16 10:07:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/88/858088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/88/858088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/858088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/858088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/88/858088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/88/858088/1"}}},"commit":{"parents":[{"commit":"b0e7c6b8e5379d9f17541426e9df449d56bbcf4f","subject":"update constraint for tap-as-a-service to new release 10.0.0.0rc1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0e7c6b8e5379d9f17541426e9df449d56bbcf4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-16 10:07:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-16 10:07:26.000000000","tz":0},"subject":"update constraint for python-saharaclient to new release 4.0.2","message":"update constraint for python-saharaclient to new release 4.0.2\n\nmeta: version: 4.0.2\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I659662ae1c0891e81ea8189504cc82f526a6845d\nmeta: release:Code-Review+1: Qiu Fossen \u003cqiujunting@inspur.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I27f036a260fe569d16f6769f58874ccdcf56d420\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/939121e011518809468e31dbbf3aa37e5160560b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/939121e011518809468e31dbbf3aa37e5160560b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
