)]}'
{"id":"openstack%2Frequirements~858089","triplet_id":"openstack%2Frequirements~master~I8d78c177d79a45ce54d3c76f8fba4a9412e227b4","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-09-23 02:02:05.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}},"hashtags":[],"change_id":"I8d78c177d79a45ce54d3c76f8fba4a9412e227b4","subject":"update constraint for python-magnumclient to new release 4.0.0","status":"MERGED","created":"2022-09-16 10:10:06.000000000","updated":"2022-09-23 03:42:59.000000000","submitted":"2022-09-23 03:40:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858089-new-release","meta_rev_id":"8fa870b809c44a63767e232ef1c366bdb00577fe","_number":858089,"virtual_id_number":858089,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-23 03:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-09-23 02:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-09-23 02:02:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}],"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-16 10:27:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-16 11:13:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-09-16 12:13:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-16 15:17:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-09-21 05:14:23.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-09-21 07:52:14.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2022-09-21 14:55:21.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"80639f5a15d0647f5b123294edf9ddd7eb4482a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-16 10:10:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60a063da38f9791b8d209061bc4fb2c3d4eb4ad3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 10:27:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/987fc0ff02734b8999c8b1009aa8422b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/88050b9566d343988a35ffb4fd308b3f : FAILURE in 15m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8dd393f93e900e382b8f29f096fed2a26081171a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-16 11:13:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a085af881d15082d85ba42181083476beb947157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 12:13:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/846f9c30f74d406ea89b88da0041f480\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc45e26131c142889eca6ab9113f1539 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25f9ad0c16f4811906f8c861656f1d6 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84e9fce09f18400393bfe71d1ed22c71 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a33e98e968c42d89a769080256a8d54 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11ba376476ed4bd6bf4c72091f26c3c1 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10f16c544fb14ea5834452c1315ff160 : SUCCESS in 3m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/35d74fb1c9874eb598bae1e72254ab8a : SUCCESS in 4m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a3624a83dd9e4bb4af06b6534939f358 : SUCCESS in 9m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/69540e090ea04efdafc3e5094fdac309 : SUCCESS in 13m 05s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/9d85134b62a4495c87d7d06a9398e0e1 : SUCCESS in 9m 45s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3472517528ea40548c1e11b99b0400b5 : SUCCESS in 5m 20s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/584ebb1416ba4d3e94a37ddb05f1e979 : TIMED_OUT in 40m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/355197426fe543b093572a2944274694 : SUCCESS in 9m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/180e354e8fa646c48e49d0605bd4b033 : SUCCESS in 5m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/07d8ff8c2e8a468ab2ac1471f89a26ae : SUCCESS in 5m 58s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c1832b1a569149cb86ceb5e34e056ea4 : FAILURE in 8m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c05400863a274512afe1f1d24c7d6408 : SUCCESS in 11m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf2264331742490db375f82155dc22cf : SUCCESS in 8m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1c606ba6f5204e6aa65e452a2943b3fb : SUCCESS in 9m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1782309bc8ae4507ab72b078af8cb4f7 : SUCCESS in 12m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/572f6709566d423babcc8004538b1054 : SUCCESS in 4m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d2de6fec16d84b4186c32a1334bece8a : SUCCESS in 9m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f3f6123babbb492285d77355a7913fcc : SUCCESS in 4m 57s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/a59de4b3fcbc4fcbb6bf3a9502ca3794 : SUCCESS in 11m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/59faf9b2242646b6b577a87af0648605 : SUCCESS in 25m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cb16f7c78b7440d686b237878a907c13 : SUCCESS in 13m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ed8ef0e362c2450cb9385a8369d209fd : SUCCESS in 6m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d517aeb8488841c1a73974a1255267e3 : SUCCESS in 3m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fb0c5f4accf6469084e1eb88a79cfcd4 : SUCCESS in 20m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a41348812ae34a8dafb79102d9f4bf8c : SUCCESS in 6m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a7ba78ed5a9349fcab1a271364ce8504 : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2aa31dc6d1ef49998bf4433bb80eb12e : SUCCESS in 6m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/48f0e7b547b64701a50107204b0ec376 : SUCCESS in 10m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba5f3b8d9b9542f789eab972da002137 : SUCCESS in 1h 24m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5f039de7f004d2a9d0962da3c462f99 : SUCCESS in 1h 48m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c1fea5e7e3975afd55ceccce8c756d3514659ebc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-16 15:17:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f67c1d01e81fe04068abe77dfdcfc97b1af95e37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 15:29:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92b4921f6c6547eb87b3f3fa4d5ab7c9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f06107f265124ba2825bbda59636119d : FAILURE in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2ad91c780f86acf529fc9804a599bf5d48a452d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 16:58:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e5db2c4886d04315aa8c551501773aba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66d3734174034f3c8296933d5e6f48a8 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbb3a02e46c5467096ed97d9c55be4fc : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2032aa8c759145e0b07aa94db106a2b2 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/887d5a29991e4d49aee32e62c0fb44ef : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5abf03c0b0a414ca300d777ea777a3d : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9966d0dcc984dfa8887bb881fb026f8 : SUCCESS in 5m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1f5be27c9254d78a6ab32613570e524 : SUCCESS in 4m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b2d5546b7a704425a7385d9bcce0bedc : SUCCESS in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e6356e37be90483981b29678c5790b5e : SUCCESS in 14m 06s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d501e4bda1f1431f8c9c2587c3d6ff8e : SUCCESS in 8m 22s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e839f8bb8453411fac4125b0c84e8c0b : SUCCESS in 6m 51s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e806795a67b842ee8ca8397cc85adb3b : SUCCESS in 7m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c52f478458734b0f955f241a58dede96 : SUCCESS in 11m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f51b15caebe442fe8befcc462d1edbe9 : SUCCESS in 6m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bbae3d57bd8b4fdab7dc2adb1ee47b7b : SUCCESS in 6m 13s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fcb2b82b08a54aa9bb1f3c40e598773b : FAILURE in 8m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f0f84a39c17042c2bd764b1fd0b3fdb6 : SUCCESS in 12m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f27e2b28001c44c585db51869fff7e92 : SUCCESS in 8m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/13e10d26a09d401795a25677e7150571 : SUCCESS in 8m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b8d00162414349418a8be38781adbe9a : SUCCESS in 13m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f2bb1362d3614cefbc539229e8fb069f : SUCCESS in 4m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/478325827eae4af2826a7500bec822d1 : SUCCESS in 9m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d706bba0094c4f688701e2b25f6c0394 : SUCCESS in 6m 06s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/628a1a9c71bc4ad99ac53e6ea4dcd74c : SUCCESS in 10m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b628808d379e41ee829cd414d177ca94 : SUCCESS in 27m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d87fed99f19b41c2b24f7760837bcd54 : SUCCESS in 16m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/26f76775d4104cafb0ea32e8ccc5c72c : SUCCESS in 7m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3a5b530d2844c45b74e8196f183b138 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b488f33ec9ac48289d38434bd2bb905c : SUCCESS in 22m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6639228f030b46ebb4c3b58e742c1935 : SUCCESS in 6m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c75fe6b6a2be45df8c513f5263d3e1d2 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/28d92ae98fe64f2cb9eda846592e0f4a : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cb707681f0e34176aeb2843ef3c87039 : SUCCESS in 14m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0aef9d53a23b4ed0a49b19b5318f93bd : SUCCESS in 1h 19m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95fa84f8ddb04d6a8bddb1aa95d7fb88 : SUCCESS in 1h 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3b4a5a55c189dee877e6a4fe97e696817db6bb5b","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-09-21 05:14:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"584cf6199b3856beb3dbc0473ee421bc8efeec04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-09-21 06:21:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"89b4ad3632581a04efe430f8bfe1035339b54f38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 06:32:51.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52a477b8141e4ef29dbcd081727d323b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9f869e6f88c34637a86003785e48d81f : FAILURE in 6m 45s","accounts_in_message":[],"_revision_number":2},{"id":"7efcbf4d41ffd386db0daa180d4c3282b367f7dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 07:40:45.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/858609 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"64616db85b06a454e39fc54773d9dd44f81551d9","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-09-21 07:52:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"847c9f4ecb031b1380cb01d29b3147d5c4112e6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 08:07:13.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6317bd9b5594467c870fb73a55ca296c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1a12b73c4771446da5b2ec1625eabe3d : FAILURE in 12m 09s","accounts_in_message":[],"_revision_number":2},{"id":"10da0c05a1fdbbb31d97706f4c9f8f5edaebaff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 09:56:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ea5c10742684f4eb56fbfc24418aa3e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71048d9cc3234952a4e739b04e583974 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83b931497cdf4b1a9795decf36d76952 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4af364b987534228947e8f39bdb691f2 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39adfd768b6a4f9bb052f559a198d45d : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b213e63a44947ceb93367dc2f87fc66 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc20089063a34f93ae8ede36df10eb52 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8df3da5538d648368c7b427fa6c5a5ab : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/153b3155dc6349879b277d9a3c9f59d8 : SUCCESS in 11m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8ecbf0e2392846f5b57e6bdd4022874e : SUCCESS in 16m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/1640ea47813449de80a70a40e72e71c2 : SUCCESS in 7m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6b6dc0fb50a34e59ad13dca5ba16a8cd : SUCCESS in 5m 25s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2e095a4be87d432c8caa1e1c87824211 : SUCCESS in 6m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/928bad225ff241bc861c0de73cd30b55 : SUCCESS in 10m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8a48f6913db24ef79a35ae454cc5f1b7 : SUCCESS in 5m 08s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/32f7fcaa14494c0c980f0d45235ffc8f : SUCCESS in 5m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a42e7aed5953453abfd980947cf9f023 : SUCCESS in 7m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9c308b86018047b2933dcb46df4a57ea : SUCCESS in 9m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9afedee8789c453d99da0b9d45ae21a7 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b37b4f5901f345a4af383988fcfb1973 : SUCCESS in 8m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a7aa3cc3fc4e4d4c80e46bbbf1a82af5 : SUCCESS in 12m 07s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dc930f37d51e4728b6c9eff1fc6e46ca : SUCCESS in 3m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9d78a35c32f94f10a921ee7534434870 : SUCCESS in 8m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/85a47875913c466c9407ce526fc51ba1 : SUCCESS in 5m 10s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/dee3584b20c84d1d8fcbb8ea547833f9 : SUCCESS in 9m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a92611dfd57d4494ab34e367d124eadf : SUCCESS in 23m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0b3384508679483b8bc31cbd0beabd79 : SUCCESS in 15m 06s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c6c05a51f9194d6db2b02588216f4847 : SUCCESS in 5m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a0824bef838419498be9d43053007cf : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95f10f1c04d64d28a883af60379cb5f0 : SUCCESS in 21m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/108dbb835f3f4f6980f63e62ea1da5ec : SUCCESS in 5m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/96101111cf3f4f668dc8fb23fe2ab322 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7444bc711082468c8f2c74f4eb373b06 : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/376cc34dde294393aa885cf992c97379 : SUCCESS in 10m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10a012a8841f4b4089887b2e134392fe : SUCCESS in 1h 08m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3a1d446f2074f3fb71c78334aa7824d : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f2f1f0173a2e490c2deb90f28fe6d63c4ac8d5cb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-21 14:55:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e73f56507d9ed757dd72f9cb38065a4b8f6c01c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 14:55:33.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23d40d5fff744d16a69f582945c8bfdf\n\n\nWarning:\n  Change 858609 in project openstack/heat does not share a change queue with 858089 in project openstack/requirements","accounts_in_message":[],"_revision_number":2},{"id":"a5d45a20ae4ced1a2603897b6cf2086f77e4aa80","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-21 14:56:32.000000000","message":"Patch Set 2: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"ae2907c86b4bc8bff166eb07d02e5f00a7c1414d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-09-23 00:09:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f0b630f1ccef6b5e866d074b39f563833e26cd25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 00:32:11.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b98981967ef84e47b40353193f152b9c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/135a6489fa304c66920391a3d073e439 : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":3},{"id":"8c0d00c0debe7c056329889ec7ee63f5acfc4d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 01:39:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87b1148d05754f0988f1d4a02cddec0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2e513ae9f1408194d84791484a50dd : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d533f4ed9a34415dbcb72f7342d3d387 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93775603958f4ae4a72a08a229285a55 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/302d02c0984a414984bff532dd94e855 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50967356f96948079ae4be5c30db6f2b : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52f7bc78ce8a45ab80434c04c4cb21fa : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/15c9393958434bdfbb01ef43193282e4 : SUCCESS in 4m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/df36e158086640f5a5c6fff853c403e7 : SUCCESS in 7m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0600d462e3c34f7595e79881bbb91e20 : SUCCESS in 10m 59s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/949740e3f45b4a6a8d9dec98ca950e80 : SUCCESS in 7m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3fd38940ec044fc78f0b73fa9b3a9039 : SUCCESS in 7m 46s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/09cce260d2654ce38ea9ca7f051ea031 : SUCCESS in 5m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/00bbaf47d95742ff95832eb6648819f6 : SUCCESS in 9m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/2fae74d45f5c47d09af40c7350f95e4e : SUCCESS in 4m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9ca3549fc0084e76b415537278638e90 : SUCCESS in 5m 21s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/17fc6ff692b947e98fcb0d6367b48a95 : SUCCESS in 10m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3e4f24518f09445c8b38d6d7cb26e9ca : SUCCESS in 8m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/993cbe00b29246edb4360a344b56569f : SUCCESS in 7m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5fd4976f58eb4e74ac90f0ac3facdaaa : SUCCESS in 7m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f1d14f34204b403fa4434c6b39257c02 : SUCCESS in 11m 06s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2586af1d62f045a89ce40db82d81d9d8 : SUCCESS in 6m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9235161924de4f46a6cadfe854f12af7 : SUCCESS in 12m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/525bfba4436041749c5bc7bc01948035 : SUCCESS in 7m 47s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/3520310cbad34d969f9d0635c6b8da93 : SUCCESS in 12m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/aa858b001b864d8280027f23c3a25214 : SUCCESS in 27m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/166fbb38487848deaaa5bba67c0180b1 : SUCCESS in 15m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dd6063beee6d40b79671988c32af3001 : SUCCESS in 8m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/34da62e1be734796891e392e9d472c85 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f49473cc0fc4f5fabd4e9287c0f8b07 : SUCCESS in 23m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0032c77e759c4b16a59505e5f2100a5d : SUCCESS in 7m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3ead1975670c4081bebeeec0ea9fda72 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e1d7824a16f2460b8881c6fabe05600c : SUCCESS in 7m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/540a11984745432e857736f7848eb680 : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f91e65aacfd042cfa3aae07490fbb962 : SUCCESS in 57m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d7cec7b6e73416aa3a3bc375f5e7703 : SUCCESS in 1h 28m 45s","accounts_in_message":[],"_revision_number":3},{"id":"02562558efab61259e9b2b62a73fafd51fe940ed","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-09-23 02:02:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cbe0659fe7a1946e3a1923c47e6100505af78355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 02:03:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"233843df2c88e2e28b8fb730779d65f2fa60e0e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 03:40:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da6365e25fbd43afbcecb41263122727\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d6e93cc56a4f3284f1e4126d3b849e : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd61c956db04eaa8c08d2b991aafa72 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ccbaba8cd0e54883854e250bd90eae89 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ed7dc3ae7024d59b30a428c53d2ade6 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/24e9bc300db04e0aa2806cb45547ee8a : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c997fb6ab0a94133a902e5e8c9fc24a4 : SUCCESS in 4m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/73696b2eb2d44fc287705fdb08a62edd : SUCCESS in 10m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/806a6bdbb3f04f4c96e4a3050bbf4915 : SUCCESS in 11m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b2b3a80e23b040de97c048fbc3734029 : SUCCESS in 11m 14s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d6ec028135434ee1b46959bc5b1953c6 : SUCCESS in 4m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/187f2d00909348029a902a2e6fdbd8d6 : SUCCESS in 7m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/348f6c8e7dbd41fabb16ef2b2d0f4644 : SUCCESS in 8m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/787b4c826ca54a2391d41517487200a2 : SUCCESS in 10m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/64247655a9d74f938d624a15beaeb1e7 : SUCCESS in 8m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4caae12d99164977a601941b85a34794 : SUCCESS in 10m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/74ca8a2bbdc74c44bd8f562ec8041cbf : SUCCESS in 14m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/95e831a6c35f437f92cf9113ace7af2e : SUCCESS in 4m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0b4483fea7be4686896759aafc3ad358 : SUCCESS in 12m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/baf82bb7c16242b4acdd0543e76ca4e4 : SUCCESS in 5m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f281194f32094c34b2067c170bd7921e : SUCCESS in 24m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cbfbf3b3c1664befaaaa4de03aa8d168 : SUCCESS in 13m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dfe5d38461564808b106ecaa3da068b9 : SUCCESS in 7m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b5a496ae0f04661aabc5bfa45446ca9 : SUCCESS in 8m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11bd2dd0b9304a56a259373e810fc713 : SUCCESS in 21m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fc40d490f60d40cab1b4b849981bc71d : SUCCESS in 8m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2fc7916bf8684b4b82d2e71b7dd7ea8c : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4b19632122ab44b482af0ece566f6220 : SUCCESS in 8m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ac937dd74e3a4381862e183412381acc : SUCCESS in 15m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d8f89ae330e45b38dd72ce87fdbd857 : SUCCESS in 56m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c1b2c5c9caf4cfd9cef8b4ed55dccf0 : SUCCESS in 1h 35m 52s","accounts_in_message":[],"_revision_number":3},{"id":"628bd8f7e3c5e09864eb36c19e34b966d8d88e54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 03:40:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"8fa870b809c44a63767e232ef1c366bdb00577fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-23 03:42:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/233584d729a94221a635dd3e8461f7a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc94b7f17de4928bb0e8e3f4d023d1a : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ff21b2d880552acbec1c3e8267364563b1ddc52a","revisions":{"330bc82a341836bd15bc20c3a3e3dd14c4dbd910":{"kind":"REWORK","_number":1,"created":"2022-09-16 10:10:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/858089/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/858089/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/858089/1"}}},"commit":{"parents":[{"commit":"b0e7c6b8e5379d9f17541426e9df449d56bbcf4f","subject":"update constraint for tap-as-a-service to new release 10.0.0.0rc1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0e7c6b8e5379d9f17541426e9df449d56bbcf4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-16 10:10:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-16 10:10:03.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 4.0.0","message":"update constraint for python-magnumclient to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5453c6f86f2265a4014b44fdaabc99997a82b426\nmeta: release:Code-Review+1: Jake Yip \u003cjake.yip@unimelb.edu.au\u003e\nmeta: release:Code-Review+1: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nmeta: release:Code-Review+1: Spyros Trigazis \u003cstrigazi@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I8d78c177d79a45ce54d3c76f8fba4a9412e227b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330bc82a341836bd15bc20c3a3e3dd14c4dbd910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/330bc82a341836bd15bc20c3a3e3dd14c4dbd910"}]},"branch":"refs/heads/master"},"0074f2b006557fde0ea97b06823e666aa61f8fd0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-21 06:21:48.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/89/858089/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/858089/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/858089/2"}}},"commit":{"parents":[{"commit":"b0e7c6b8e5379d9f17541426e9df449d56bbcf4f","subject":"update constraint for tap-as-a-service to new release 10.0.0.0rc1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0e7c6b8e5379d9f17541426e9df449d56bbcf4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-16 10:10:03.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-09-21 06:21:48.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 4.0.0","message":"update constraint for python-magnumclient to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5453c6f86f2265a4014b44fdaabc99997a82b426\nmeta: release:Code-Review+1: Jake Yip \u003cjake.yip@unimelb.edu.au\u003e\nmeta: release:Code-Review+1: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nmeta: release:Code-Review+1: Spyros Trigazis \u003cstrigazi@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n\nDepends-On: https://review.opendev.org/c/openstack/heat/+/858609\nChange-Id: I8d78c177d79a45ce54d3c76f8fba4a9412e227b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0074f2b006557fde0ea97b06823e666aa61f8fd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0074f2b006557fde0ea97b06823e666aa61f8fd0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ff21b2d880552acbec1c3e8267364563b1ddc52a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-09-23 00:09:14.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/89/858089/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/858089/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/858089/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/858089/3"}}},"commit":{"parents":[{"commit":"b0e7c6b8e5379d9f17541426e9df449d56bbcf4f","subject":"update constraint for tap-as-a-service to new release 10.0.0.0rc1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b0e7c6b8e5379d9f17541426e9df449d56bbcf4f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-16 10:10:03.000000000","tz":0},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-09-23 00:09:14.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 4.0.0","message":"update constraint for python-magnumclient to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I5453c6f86f2265a4014b44fdaabc99997a82b426\nmeta: release:Code-Review+1: Jake Yip \u003cjake.yip@unimelb.edu.au\u003e\nmeta: release:Code-Review+1: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nmeta: release:Code-Review+1: Spyros Trigazis \u003cstrigazi@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\n\nChange-Id: I8d78c177d79a45ce54d3c76f8fba4a9412e227b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff21b2d880552acbec1c3e8267364563b1ddc52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ff21b2d880552acbec1c3e8267364563b1ddc52a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
