)]}'
{"id":"openstack%2Frequirements~858299","triplet_id":"openstack%2Frequirements~master~I80a24e382c972f9af2dcad002cf09e5ebf0b137f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I80a24e382c972f9af2dcad002cf09e5ebf0b137f","subject":"update constraint for python-watcherclient to new release 4.0.0","status":"MERGED","created":"2022-09-19 11:28:10.000000000","updated":"2022-09-21 05:43:03.000000000","submitted":"2022-09-21 05:42:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858299-new-release","meta_rev_id":"e5c208df855a35e2807ae8104e0204ee8438093e","_number":858299,"virtual_id_number":858299,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-21 05:42:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-09-21 04:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-09-21 05:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2022-09-19 12:34:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-09-21 04:27:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-09-21 05:01:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-19 11:40:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-19 12:34:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-09-19 13:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-21 04:27:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-09-21 05:01:05.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"aa27ecc361e3f7be76ee2342f613dcbce694dac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-09-19 11:28:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9117590d259693198be5e48b55a469335be333a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 11:40:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8429c4968b7e4d00a845225677dba885\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a00c4206bcc34d8881115c4c9fae1cac : FAILURE in 10m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2e42e00b560ddab7bca40b435fdadc439b7fed6e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-09-19 12:34:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5d9f4e8eaf08d49a128d81d8eb6c75cf569632a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-19 13:11:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e6c6c2911c447a4b5a78d42f435dd9a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d27d45ad3554b45b5c8b0427c4d76d6 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95a0d027fe474985a05b6b3373abd838 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3444dbf2f4e4d2599799f29604748b6 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a4900e2617b4a7594005ee92d6dfc90 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8b495280e134bf9a8cf56cf19576712 : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b366f4b4987e46eaaab764c6d0f26023 : SUCCESS in 3m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ff552c7b0d6841da8271c1c42cf8928a : SUCCESS in 9m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b5e02361c5ec4797953295ded5946bb7 : SUCCESS in 12m 58s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/bcf519aa9eb14d17a7461f559bccab05 : SUCCESS in 7m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ba2630e6fe7b4972978a47cecddea4a7 : SUCCESS in 4m 33s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d16ba91a2b05430aa15ac6cc03f92572 : SUCCESS in 7m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/322b25bbdf43407da497344160e34968 : SUCCESS in 9m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ba601759de7a4b088aa3873de7d8384c : SUCCESS in 5m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e68eef0c1e2847ae890b88ba0d8882a5 : SUCCESS in 6m 03s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/4d3ba2035b0d4b6ea62137d0ae48345e : SUCCESS in 8m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7f63530b633142dd9487a98350f3166a : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ad6a1157f8748a6a7b06c4007033a30 : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/70b8c2f988f1456d88e0e2a6c64f3eef : SUCCESS in 8m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1b6cca983ee043d1971c6823f4085c4b : SUCCESS in 11m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0b4619b9cf604e03bd67b1dedd5d6c00 : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ce7e23146dbf401bb212374acad3c2db : SUCCESS in 9m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ce14353b56c14b5e94ecbb59d79329b5 : SUCCESS in 4m 20s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/9a1a260ef94342efb36d1aeaa1bb46cd : SUCCESS in 10m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/99db861457234c02b89d564a34613fdb : SUCCESS in 24m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/81323445b91a410f844085ed7e6c69ce : SUCCESS in 12m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cca6f7f7f6a34018a3a76cb0424fdbff : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ae50deedbc2a420f85bf6a897a61f686 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e6169c8b786846e88305eb3912152f9a : SUCCESS in 20m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4610532450db41afa177828525404d7c : SUCCESS in 6m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c41c04f30afe4514a8fe16a077478ad8 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d892aa99a73245668ee91f2d875e5022 : SUCCESS in 7m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d9327af6abb44e4e8d9b6fcc88b9bd19 : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ae2e16c61064d909376e89b024ff82c : SUCCESS in 1h 20m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0c2853c5974420d97ad22f8df1a1886 : SUCCESS in 1h 37m 56s","accounts_in_message":[],"_revision_number":1},{"id":"61562ac3a88eeb54433962d4ce83b79c3a2b7972","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-21 04:27:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e3c0bc219b784d63682f01855391983be28bc05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 04:28:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b39af6347b9fca71a3979ec3a82adff5ce3054ea","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-09-21 05:01:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e71bd15840b8bbc5ee1c1d42d46f03bcb90b1942","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 05:42:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bad6dd4571864ad583f5a9ada19a9c96\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0d3fee30c0d43a3ba0821ae988f58d0 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a45233f6c89a45cd91d48cd0ac8e1b40 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/987d0c29d0744f8e8123d33c6b2cd561 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24dfc5cfdd7f469a803d3153a07c449c : SUCCESS in 5m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14904c22e2394240ae3cd8421c72d56b : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9498742d04eb4cd7a67ba5459823aa79 : SUCCESS in 4m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ccf34172e594683b7595546aea1fe7e : SUCCESS in 7m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/43a3723bbfb0403cad3cc56d8d76f6aa : SUCCESS in 14m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0a6424178cf74ba5b9ba785e953f7639 : SUCCESS in 10m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/44feab6725804cc38cea0b442e5d3cf8 : SUCCESS in 5m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1fc5cfe678e14a78a59e46c28fd43850 : SUCCESS in 5m 23s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f0f9ada90438417ababe3df07e5d3ee7 : SUCCESS in 8m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/becd6eb82d604dedacb2f6b89df71e24 : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40fed576e9314b3890569b0b770b5d28 : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f2e1d699d8c94e469faadec89988704a : SUCCESS in 7m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/753ce27813394d27a53134fb0cbcf6f4 : SUCCESS in 13m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ce6522a81e2c4e9492ef2dbaa7be9f40 : SUCCESS in 4m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b2767fc15ac549379cb8b6328e2033bd : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a7ed2180936d4c3ab03f0538e124adb1 : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ca0a442a44fd4ef4a7e39c5b7e9c6676 : SUCCESS in 27m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/97dc8e68a619470db307791bed918871 : SUCCESS in 15m 03s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4ec6ea64d873475092003922fcecd5ae : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6fa6d461619c4148a932f1cbe71ee54f : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d10cee0bd6f2407aba6aa283e7f09aa6 : SUCCESS in 23m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/955d6bd5e5fd4f6d80edd6a6a518c1cb : SUCCESS in 6m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6ed89e1a1282473f806fafe7acc86c29 : SUCCESS in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e9dd785f7c743f5bf9f7aa03ea3c182 : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d5f6bd03b9a7430180baaffcd9dbdae7 : SUCCESS in 11m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74a71bb3fc944e26b61737f26560e7ac : SUCCESS in 1h 10m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0508beadaca243cdb3615e926a855a56 : SUCCESS in 1h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"fdff73d91208b608252e63a5c0153b5f065f1903","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 05:42:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5c208df855a35e2807ae8104e0204ee8438093e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-21 05:43:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dacf76c6316644e5968a07e3261493d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f6ec35a5ef4484b53ea222309e8b65 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4aca3134b2679a8114b38807ac80afdc6875cd45","revisions":{"4aca3134b2679a8114b38807ac80afdc6875cd45":{"kind":"REWORK","_number":1,"created":"2022-09-19 11:28:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/858299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/858299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/858299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/858299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/858299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/858299/1"}}},"commit":{"parents":[{"commit":"939121e011518809468e31dbbf3aa37e5160560b","subject":"update constraint for python-saharaclient to new release 4.0.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/939121e011518809468e31dbbf3aa37e5160560b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-19 11:28:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-09-19 11:28:08.000000000","tz":0},"subject":"update constraint for python-watcherclient to new release 4.0.0","message":"update constraint for python-watcherclient to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ied3910931cba712a5abc79d475a26b19ad2cb9db\nmeta: release:Code-Review+1: chenker \u003cchen.ke14@zte.com.cn\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I80a24e382c972f9af2dcad002cf09e5ebf0b137f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4aca3134b2679a8114b38807ac80afdc6875cd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4aca3134b2679a8114b38807ac80afdc6875cd45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
