)]}'
{"id":"openstack%2Frequirements~859291","triplet_id":"openstack%2Frequirements~master~I8b22e921785abb2c144bc0a8c446e3485ce739a7","project":"openstack/requirements","branch":"master","topic":"story/2010320","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-09-26 22:17:27.000000000","reason":"Change was submitted"},"32927":{"account":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"last_update":"2022-09-26 22:17:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8b22e921785abb2c144bc0a8c446e3485ce739a7","subject":"[CVE-2022-40023] Bump Mako version to 1.2.2","status":"MERGED","created":"2022-09-26 13:04:47.000000000","updated":"2022-09-26 22:18:23.000000000","submitted":"2022-09-26 22:17:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859291-story/2010320","meta_rev_id":"1a703a81da7bafb29a3089b6a400b86640ce2f58","_number":859291,"virtual_id_number":859291,"owner":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-26 22:17:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-09-26 20:41:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-09-26 20:41:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-26 13:20:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-09-26 14:39:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-26 14:55:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-09-26 14:57:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-09-26 20:41:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"afcd186df9a543165edb520cdcb327b83a7d1583","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-09-26 13:04:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4626d4d6636c773b199841e9549a1211f0008c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 13:20:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d67a935dbb1c42f7875faf928b544486\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/56127683624f4253973e0f5d65520384 : FAILURE in 9m 52s","accounts_in_message":[],"_revision_number":1},{"id":"934054054e9d0e9ac35c78512eff8aadd626181f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 14:39:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66bf96efdfff448e9280b7f62e91dc7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4383e21b8a64203bff8b57514401615 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63632a2eebe84eb0b6a9259db70ea874 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/855606a86c874850ad8adfe7c81f7059 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f72c3838a9c84cbd9d00bb6c3abd2c27 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c1b20afe02b40dfa603e9b7ce86e0e0 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b3cb6b53a834f96918d204ca64d559f : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1d1320e2c1cb483784730db7bad26a13 : SUCCESS in 4m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e8d9f02f8b5d41c2b4443e8a549f83da : SUCCESS in 8m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4c37b71d592043c9814986b5c36c0436 : SUCCESS in 15m 05s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/09dcbb80b74448b8905a92e4da3f8bc5 : SUCCESS in 10m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3ec7f64d1e3c442ab4483b62836f97bb : SUCCESS in 5m 34s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/6fe40b02e98849a480faa3a91ab5e4f2 : SUCCESS in 6m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9569bd978c404912b255d67e7f183e61 : SUCCESS in 12m 22s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b8417d9b12934071a836842faff80609 : SUCCESS in 4m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ecae08da4ad34b0289ad234e3bb8954a : SUCCESS in 6m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/4c26ad9b9fe04d7485b66544714d7e3c : SUCCESS in 10m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e4bd747afdd3419eb1d511917506f9a7 : SUCCESS in 9m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/832c115b30114848aa6342b464563327 : SUCCESS in 10m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ad1a2c287877409d91532a7f5628c699 : SUCCESS in 9m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/32dcfaaf9c41446a82172ce45bb6ee2e : SUCCESS in 10m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/eb6547fa4f184a838acc18f77cd12462 : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0579a0bcee454f998d18700a5fb9d3e1 : SUCCESS in 13m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8d649c7a3f3d417a8143fafa2cbab517 : SUCCESS in 5m 36s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/351c1e3a8dcd40f1aab5fa129dad676c : SUCCESS in 9m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/55acb35c9cd24408b2fbd6c40d514b5f : SUCCESS in 23m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a3992e38917c42ab8eb6faaa164ade3e : SUCCESS in 12m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8af3956a477e4eb7812f918a1003b528 : SUCCESS in 6m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/21571fdc4ac14e18839dd51e229bb4d6 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa7eaa18e7ad4081bdf11b03a099efb8 : SUCCESS in 20m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/50cb285b084a4526bfba101abf89dd06 : SUCCESS in 6m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/49098ac8ff8d49e1b105c630fe5c1ea9 : SUCCESS in 8m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0313922f8cd84c05ad4a36c4d342e231 : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a57c1993e9bd490d8801c0c5ff2d789b : SUCCESS in 9m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d39e2b548a9348e4a02a8518276aab03 : SUCCESS in 52m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/32c8325cee4646a58009c172ac43a9a3 : SUCCESS in 1h 31m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3f38760b4f238e10e79d1551a05b0a62b6d7a070","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-26 14:55:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49eff7bab7733f722fdc2a7c1254fded1c9e539e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-26 14:57:30.000000000","message":"Patch Set 1:\n\ndo we want to pull in 1.2.3 which has another related fix to the lexer\n\nhttps://github.com/sqlalchemy/mako/compare/rel_1_2_2...rel_1_2_3","accounts_in_message":[],"_revision_number":1},{"id":"dd7bbb8fd4acd4e7dd5bd0e4a43be750a0b03412","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-09-26 15:32:21.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5bfc1dd724a98772f9e7d5c9c322474f3eaaded6","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-09-26 15:35:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f90cf9babe2b8c781acbd360927fd0017f943730","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-09-26 15:36:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9fd36a0f9f1fd7f6f0a149b4376a91aa37911cb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 15:45:08.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25c82917596042269059c51948d3bb17\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/338b932e05844061b581e8d1e41b8c04 : FAILURE in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"47e1fbc9f7240933024bec93770ff60f50937b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 17:10:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78ae3084f5b846d5815142fed8f26652\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8474bdb2c855429dbbceb092a6ccecfb : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/450961020776404ca31c95f3f93dbd7b : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5168a5583df14b82b62aea68d6db98fa : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fb0bce79ef44bb78063eb908166a3fe : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac49f9c229964cca93efc9cfdc0890a6 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/808a920f3061469188034a75a03a8c6b : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fae8056c74444553b78dddd99ec8e3ac : SUCCESS in 4m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4f78370ad6b749bd906ea0c58148ac68 : SUCCESS in 7m 27s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/096e636ccc3b4a929f29131f037aee0e : SUCCESS in 12m 41s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/642553e2b355488ea10f69a8e57cc629 : SUCCESS in 9m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bfd93e80c9dd499c8e80e1be95dbb512 : SUCCESS in 5m 42s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/37c996abc99b40f4912128083fbb882f : SUCCESS in 6m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1e1ba1ef35d3488da28ee4f2e08e8976 : SUCCESS in 11m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0609586058344fc6b130de82dd04db46 : SUCCESS in 5m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4fbf9b8745ae4a3784379da0c6af94aa : SUCCESS in 6m 26s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6e737de8be134e3dae1180f506973eac : SUCCESS in 8m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2ab0b33aa80648eea18661839374063b : SUCCESS in 10m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b6603572299a4707acec9294726eb92a : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0147eea369a846ab807a8fcafbfccf88 : SUCCESS in 9m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f6de7fa0b1d244199c56cc3b758e2b60 : SUCCESS in 12m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bfa40e0c519f4bf2a8fcda1b3cf38b53 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/593b07b0f10a47a2a1225ef26481aa12 : SUCCESS in 11m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1d3aece3220049f3ac7e92c996310833 : SUCCESS in 5m 39s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/98837ad8e9594bba8afa6d3d5ff804ad : SUCCESS in 9m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a5cc41ab7ddc46ee94adeea8e10cfc8c : SUCCESS in 30m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a912ece51f1844eba335de406f4320e4 : SUCCESS in 12m 58s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f0fdeb34947e46e1b698aa71c8712916 : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c075fb881e5c44eaa155bba9e7c3155f : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a60095481ad04b8d90611f3e0594704a : SUCCESS in 25m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1b711b3211c9411ab727536a350c072e : SUCCESS in 7m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1c4b0b1e9d4a4751b188d1a247a8a395 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c6d8f1c688f3442db516c7a63f67941e : SUCCESS in 7m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/91e7d6be1cab4ce0b5bd6e2ce18ca1b3 : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50e7d7ebd54f4af6a3c33a8a1525b93d : SUCCESS in 1h 05m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8c1d3d5a3a8486d8eec55261feb8261 : SUCCESS in 1h 36m 17s","accounts_in_message":[],"_revision_number":2},{"id":"70904249ccce126afc5f91e66d71d21b1c0fc454","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-09-26 20:41:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1db497ba720072e4a0e76539ad0b3ec5616c7631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 20:42:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eea4485ee5cc09e6af09d3aa387546f67f9c11c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 22:17:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/479f858390fd42f1ba438e9a6b32649f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f2ab372a2848e0990d4adfc3df84c2 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98471f9f527d447b93ba1b122b85f979 : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49e6b2fd28e84664b47e307d647cb6ac : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c55244daf253444fadd5f09cefd39202 : SUCCESS in 5m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/abcabf87604642dd961239b2b1350b68 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a4fb54df50f489e9e150afa3c999715 : SUCCESS in 7m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c7a8c7be48d41928810ca1c51e5fca3 : SUCCESS in 8m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7fda1df041994402a83441fc71059a9b : SUCCESS in 12m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e4905dd61831441b965b76bf55bb4ef4 : SUCCESS in 11m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d9aeea3140de41da81a6a955b6101c71 : SUCCESS in 4m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f6a5878ad3f346cca102f155178d29fb : SUCCESS in 9m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/49ded09afe604a50bcc7941da36449b7 : SUCCESS in 8m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f146306ac96b453b9ae1bc8ab0707fab : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63b12ad56b3b4f5d966ed347ab8fd609 : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6d62471d37c4426d9dc1dc81103096c1 : SUCCESS in 8m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/14c2abab1e3b49609fcd4947b02a2ea3 : SUCCESS in 13m 33s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c8931d0aadb94acbb81497a9137d29f5 : SUCCESS in 9m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/34aa0d3c7a1c4fb88307bb9a5d5c3f59 : SUCCESS in 8m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d9001196899945d09eedd52191d039b9 : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/86e042e89f204bb6bcd6a9bd6b30fb36 : SUCCESS in 21m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b7a8b2f2846f4876b295b8c954b33d55 : SUCCESS in 14m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fe0451e1c74c455aaa4bd45cd523ff00 : SUCCESS in 4m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/69e45a878c024375a78ebcf7531a103f : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8480ce081a344f189dcb8f079c731b54 : SUCCESS in 23m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/786679db9ad44d09ba1a709fe3886056 : SUCCESS in 6m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/aea6b858253b4015bf2fdd6b8d4a2039 : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2a0f8012023844cab69bb3387ea3c5cd : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ddbb6c48fe1e4c95a063a6cc102006fb : SUCCESS in 9m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3839292d2aec448ca2918159135eaa87 : SUCCESS in 58m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a29d8e047484041a766b531f5087352 : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":2},{"id":"8730664f4355652248ae3678e356011a82c75db4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 22:17:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1a703a81da7bafb29a3089b6a400b86640ce2f58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-26 22:18:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bb42672f4eb433cb1180635870e445f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da71986ecf7c4c2f8a6ca5a5973da1a3 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"759b34c0a2d538464c8dc3e8ebd96f0f1bb2c090","revisions":{"888d37cdf534f0ca6ab5811931838dda19337862":{"kind":"REWORK","_number":1,"created":"2022-09-26 13:04:47.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/91/859291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/859291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/859291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/859291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/859291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/859291/1"}}},"commit":{"parents":[{"commit":"6ddb298796defca0dc0bae96a1b081cf5c7106fc","subject":"Merge \"update constraint for python-magnumclient to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ddb298796defca0dc0bae96a1b081cf5c7106fc"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-09-26 13:03:59.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-09-26 13:03:59.000000000","tz":180},"subject":"[CVE-2022-40023] Bump Mako version to 1.2.2","message":"[CVE-2022-40023] Bump Mako version to 1.2.2\n\nCloses-Bug: 2010320\nChange-Id: I8b22e921785abb2c144bc0a8c446e3485ce739a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/888d37cdf534f0ca6ab5811931838dda19337862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/888d37cdf534f0ca6ab5811931838dda19337862"}]},"branch":"refs/heads/master"},"759b34c0a2d538464c8dc3e8ebd96f0f1bb2c090":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-26 15:32:21.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/91/859291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/859291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/859291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/859291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/859291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/859291/2"}}},"commit":{"parents":[{"commit":"6ddb298796defca0dc0bae96a1b081cf5c7106fc","subject":"Merge \"update constraint for python-magnumclient to new release 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ddb298796defca0dc0bae96a1b081cf5c7106fc"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-09-26 13:03:59.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-09-26 15:32:21.000000000","tz":0},"subject":"[CVE-2022-40023] Bump Mako version to 1.2.2","message":"[CVE-2022-40023] Bump Mako version to 1.2.2\n\nStory: 2010320\nChange-Id: I8b22e921785abb2c144bc0a8c446e3485ce739a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/759b34c0a2d538464c8dc3e8ebd96f0f1bb2c090"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/759b34c0a2d538464c8dc3e8ebd96f0f1bb2c090"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
