)]}'
{"id":"openstack%2Frequirements~860091","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Iaae5a5e3b038ff17b785f59c6e83ea5e1825894e","project":"openstack/requirements","branch":"stable/yoga","topic":"story/2010350","attention_set":{},"removed_from_attention_set":{"32927":{"account":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"last_update":"2022-10-03 16:28:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaae5a5e3b038ff17b785f59c6e83ea5e1825894e","subject":"[CVE-2022-1941] Bump protobuf version to 3.19.5","status":"MERGED","created":"2022-10-03 12:50:37.000000000","updated":"2022-10-03 16:28:47.000000000","submitted":"2022-10-03 16:28:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860091-story/2010350","meta_rev_id":"61dfad9093ce8ce0b6a605a0e92e6cff92043d00","_number":860091,"virtual_id_number":860091,"owner":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-03 16:28:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-10-03 14:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-10-03 14:49:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-03 13:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-03 14:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-03 14:49:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d9f0fb7429c26556ef9b5afde81410584fdba4a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-10-03 12:50:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72941616ec5afb951b53b39b8a845a3181567670","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 13:03:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/827902293c15411e97ff8ed47337cd12\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/55be8fccf0a2495b89bca46a7d74a320 : FAILURE in 8m 30s","accounts_in_message":[],"_revision_number":1},{"id":"77ecbb76410fdfce18af690a8a37523d343f8bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:48:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27ac364123134dae9e1665a6641f946b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68387dc428794933b444b8063db58158 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08fd3524539b401e8e39104458010927 : SUCCESS in 8m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b89040d91e640c9ad0c374704994b3a : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76391394e3d7454cb676fff942e8548e : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0de06077fce04b658c844da507667725 : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/536cace8ce5b49b79baf51ab8ff2278f : SUCCESS in 8m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/230cd0dcf7814d1e931c66d197514b6a : SUCCESS in 10m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cfb7e13a426146a899fe7bbd97abb2b9 : SUCCESS in 4m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4f7a70331fba47f082945aea907ac996 : SUCCESS in 11m 31s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4581640692af48a38ff03c9d8773edf8 : SUCCESS in 7m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/614a4260134a436c995f5cc602bebd8f : SUCCESS in 8m 07s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a44333de40324600b5338d1365027644 : SUCCESS in 12m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/81400d54a3b84682b8de81533405e443 : SUCCESS in 15m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/452057fb965d4e3ab2e539c9a408e5be : SUCCESS in 12m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9dbc7b985a874989accd60ae17093875 : SUCCESS in 13m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c5c773b2c4cc4b5c80054fb4710b7f19 : SUCCESS in 5m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d6a19e3e08814666b73f944e1fa84111 : SUCCESS in 10m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/75fa2064f738480b9a8bdb88d3972dbb : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f70d3b86c9b7483985313393ef1a497e : SUCCESS in 22m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/05f8362cdd4d49c896b4919169dca07a : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/26443850ecb44eeabc2b8159b9552e07 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/317d218735f2413eb4ed32d883a0d045 : SUCCESS in 4m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/848455ee55ad4956858e93af47aa3550 : SUCCESS in 20m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/05a33bf26319409b8d2965a1d35ed569 : SUCCESS in 4m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d7ed19e4487a4fd39ee42d635b2c3802 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0471cf99b8d4a37a50d3abf874f6d24 : SUCCESS in 7m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f811a374c5224426bee8ae98af755904 : SUCCESS in 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15f32d8c21814495b9dbb89f1b848605 : SUCCESS in 51m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16b201a0b7c94ad3b2d414d0dd9a03cb : SUCCESS in 1h 52m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d7c04770cf416d30f0bef840db81ef2216b7e112","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-03 14:49:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df4e984f5b14b42f5eb13544a7f4b8ed40720de1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:50:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0346566758fc0195d9fcdbe3eb6211d47fbdfec6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 16:28:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8866cd5b3174dbdbec263591ac87d55\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6c00125fdd7428b938eb59cf4198ca5 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bb8d831474b4359906299de52fd701b : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5334972812944af3b9ffed0f5433cb9c : SUCCESS in 4m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e92c3df2a967475ead55aafbd28e6b93 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72209243929f42aaa8c55e04fad4b642 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/25123191b1f44eb98702baf28ecd75a4 : SUCCESS in 11m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c784b3cac316405bad286d1dd199d0d1 : SUCCESS in 7m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/63c6485a729740369dbcb50e36445b8a : SUCCESS in 10m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3594bc8eae2d4497ad39022a73258bbc : SUCCESS in 7m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/69b397ff02544805bd7e195ada7f47be : SUCCESS in 6m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/38063bdf29df416f89fde0ac82c512a3 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ce26833a976448a80cc37af4958bdd6 : SUCCESS in 11m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f317e284fa5a44d6bb5043f911fc697c : SUCCESS in 8m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/47142b42dc194b4da27412507651cb11 : SUCCESS in 11m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e0fed6805150499caa7da66a1ba85dc3 : SUCCESS in 4m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e800170004c74da0a1a1a763084f02bb : SUCCESS in 12m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2f4a03486e5e4c25a4618d9660769699 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f1c93e5471e04c189f83189d3541ec78 : SUCCESS in 26m 04s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a0aeb71a52574ba9965da1938c940cc6 : SUCCESS in 12m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a8e308489ea8482ab95eb6549d73a11f : SUCCESS in 6m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f03518ba98f4a5ba2e6c3dd2837918c : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0152cb5da8b94ae1b55240b4e6d556dc : SUCCESS in 21m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/01d506b3926645488320e9ee60a961ee : SUCCESS in 5m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a2d115380eca4b6da17e0e77a0bf053e : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3e228c9c14524a2cad676af6218594dc : SUCCESS in 9m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c1fe023d8b69479f8a598da2b6828052 : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c596dfbd4dc04fbf96452f9e605a82ae : SUCCESS in 45m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1861002ee3749578ee2ba701aa796b8 : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"61dfad9093ce8ce0b6a605a0e92e6cff92043d00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 16:28:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a43ba3f7f8ef605536621983c46afc791152778","revisions":{"6a43ba3f7f8ef605536621983c46afc791152778":{"kind":"REWORK","_number":1,"created":"2022-10-03 12:50:37.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/91/860091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/860091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/860091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/860091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/860091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/860091/1"}}},"commit":{"parents":[{"commit":"e64217c9f9316675f57e14ff559bb4de90bc6637","subject":"Merge \"[stable-only]Pin PrettyTable in py3.9 in stable/yoga\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e64217c9f9316675f57e14ff559bb4de90bc6637"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-03 12:48:35.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-03 12:48:35.000000000","tz":180},"subject":"[CVE-2022-1941] Bump protobuf version to 3.19.5","message":"[CVE-2022-1941] Bump protobuf version to 3.19.5\n\nStory: 2010350\nChange-Id: Iaae5a5e3b038ff17b785f59c6e83ea5e1825894e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a43ba3f7f8ef605536621983c46afc791152778"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6a43ba3f7f8ef605536621983c46afc791152778"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
