)]}'
{"id":"openstack%2Frequirements~860792","triplet_id":"openstack%2Frequirements~master~I039171301da04f5005eee8d6c4ff8d94bec6b165","project":"openstack/requirements","branch":"master","topic":"add-antelope-python-jobtemplates","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-13 06:02:38.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I039171301da04f5005eee8d6c4ff8d94bec6b165","subject":"Switch cross testing to python3.10","status":"MERGED","created":"2022-10-09 07:17:15.000000000","updated":"2022-10-19 17:46:43.000000000","submitted":"2022-10-19 17:45:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860792-add-antelope-python-jobtemplates","meta_rev_id":"a33c52d62ed85d1296eba5ecd3857185a67ea0fd","_number":860792,"virtual_id_number":860792,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 17:44:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-10-12 19:17:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-10-19 16:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-10-19 16:17:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-09 08:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 19:17:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-10-19 16:17:46.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"eef1c9a48318b13677f96799ef84c4e43ab3be4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-09 07:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51e5cfa6535eb463b719a1169c71f8be4aa77a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-09 08:12:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d1aff0391383832f9293fc793124b65f2e8748e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 08:13:24.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch master.  The error was:\n\n  Job cross-swift-py310 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              voting: false\n          - requirements-tox-py38-check-uc\n          - requirements-tox-py39-check-uc\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-aodh-py310\n          - cross-barbican-py310\n          - cross-ceilometer-py310\n          - cross-cinder-py310\n          - cross-designate-py310\n          - cross-glance-py310\n          - cross-heat-py310\n          - cross-horizon-py310\n          - cross-horizon-npm\n          - cross-ironic-py310\n          - cross-keystone-py310\n          - cross-kuryr-py310\n          - cross-manila-py310\n          - cross-masakari-py310\n          - cross-mistral-py310\n          - cross-neutron-py310\n          - cross-nova-py310\n          - cross-placement-py310\n          - cross-osvif-py3\n          - cross-nova-functional\n          - cross-placement-functional-py310\n          - cross-octavia-py310\n          - cross-osc-tox-docs\n          - cross-swift-py310\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      check-arm64:\n        jobs:\n          - requirements-tox-py38-check-uc-aarch64:\n              files:\n                - bindep.txt\n                - upper-constraints.txt\n          - build-wheel-cache-centos-8-stream-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy-arm64:\n              files:\n                - bindep.txt\n      gate:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py38-check-uc\n          - requirements-tox-py39-check-uc\n          - cross-cinder-py310\n          - cross-designate-py310\n          - cross-glance-py310\n          - cross-heat-py310\n          - cross-horizon-py310\n          - cross-horizon-npm\n          - cross-ironic-py310\n          - cross-keystone-py310\n          - cross-kuryr-py310\n          - cross-manila-py310\n          - cross-masakari-py310\n          - cross-neutron-py310\n          - cross-nova-py310:\n              files: ^upper-constraints.txt$\n          - cross-placement-py310:\n              files: ^upper-constraints.txt$\n          - cross-osvif-py3\n          - cross-nova-functional\n          - cross-placement-functional-py310\n          - cross-octavia-py310\n          - cross-osc-tox-docs\n          - cross-swift-py310\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - tempest-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"f12505fb63d206a8193d76cd30078bb84e24fa94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 08:13:26.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch master.  The error was:\n\n  Job cross-swift-py310 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - openstack-python3-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              voting: false\n          - requirements-tox-py38-check-uc\n          - requirements-tox-py39-check-uc\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-aodh-py310\n          - cross-barbican-py310\n          - cross-ceilometer-py310\n          - cross-cinder-py310\n          - cross-designate-py310\n          - cross-glance-py310\n          - cross-heat-py310\n          - cross-horizon-py310\n          - cross-horizon-npm\n          - cross-ironic-py310\n          - cross-keystone-py310\n          - cross-kuryr-py310\n          - cross-manila-py310\n          - cross-masakari-py310\n          - cross-mistral-py310\n          - cross-neutron-py310\n          - cross-nova-py310\n          - cross-placement-py310\n          - cross-osvif-py3\n          - cross-nova-functional\n          - cross-placement-functional-py310\n          - cross-octavia-py310\n          - cross-osc-tox-docs\n          - cross-swift-py310\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      check-arm64:\n        jobs:\n          - requirements-tox-py38-check-uc-aarch64:\n              files:\n                - bindep.txt\n                - upper-constraints.txt\n          - build-wheel-cache-centos-8-stream-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy-arm64:\n              files:\n                - bindep.txt\n      gate:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py38-check-uc\n          - requirements-tox-py39-check-uc\n          - cross-cinder-py310\n          - cross-designate-py310\n          - cross-glance-py310\n          - cross-heat-py310\n          - cross-horizon-py310\n          - cross-horizon-npm\n          - cross-ironic-py310\n          - cross-keystone-py310\n          - cross-kuryr-py310\n          - cross-manila-py310\n          - cross-masakari-py310\n          - cross-neutron-py310\n          - cross-nova-py310:\n              files: ^upper-constraints.txt$\n          - cross-placement-py310:\n              files: ^upper-constraints.txt$\n          - cross-osvif-py3\n          - cross-nova-functional\n          - cross-placement-functional-py310\n          - cross-octavia-py310\n          - cross-osc-tox-docs\n          - cross-swift-py310\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - tempest-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":2},{"id":"d6be11ce5952297e941dfa25a258063f188209d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-09 08:24:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ee27fba70b27e858fc6571dd008d0604c831cc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-09 10:12:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"302a514f0a36a3944c7e3d1855cd54cf4f7db5fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-09 11:47:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e04b93fe0b04b4090ec891e528b7896\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6d2b63159c54d44b22b64296f70c948 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a090ed9f5f64ade9e6cfda0fdd11270 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4374ab7a55744fc2a0e769c468241d3d : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47d116d58f4b4e12825fbfa474587ad3 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2283046b9672402887534c1cb6dea097 : SUCCESS in 3m 12s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/65d6102a7bb7447481b7fa0a5a0a5d89 : SUCCESS in 6m 41s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/abcc1b7a18b24c9baffb446653f4871d : SUCCESS in 5m 34s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/b216e7bc598342d6992765af08ff5a7c : SUCCESS in 5m 05s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6a21cefdbf7a4770b42be895e035fb8a : SUCCESS in 11m 04s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/ff5793694bfc4fd4999aed964c1461a9 : SUCCESS in 3m 50s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/94dfd549aff541c1b5ca14d6b1901a00 : SUCCESS in 5m 41s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/3cda1b5c91974a098b293991dc1e6198 : SUCCESS in 8m 03s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9fd0a36c7fa449108a61027c15f197e0 : SUCCESS in 7m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fc764aa8ad8f42f697b7f11c66d0feb1 : SUCCESS in 6m 03s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/6d305bf349544278b388c406da5ca314 : SUCCESS in 8m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/af2b049ef62f4fd492d0f93226dc1a11 : SUCCESS in 8m 34s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/199f7e494cc44f1ca6b248de04768699 : SUCCESS in 3m 16s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/7f65ea52ecda4a5698b6f75f39a8f4a7 : SUCCESS in 7m 11s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/fb15504d34754c2faa9b6f2cffc011cf : SUCCESS in 5m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/2d265be2e5af447aaf2c0dac782918c2 : SUCCESS in 7m 55s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5490811989bf4246b63c7ac0ddd25750 : SUCCESS in 19m 26s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/dcf441ddf4ef4c41bd041147420335e2 : SUCCESS in 13m 10s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/13e85e0b24e04345a856813b31374b09 : SUCCESS in 4m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/962f95b8f20a41e2b20b023ccf6dd3bb : SUCCESS in 3m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e19888b99c094aefa08d9c19bc9cb883 : SUCCESS in 19m 30s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/57840f4096f54c1887b8b758e76b095e : SUCCESS in 5m 15s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/bf6627d7140349898d4c0058a69339ff : SUCCESS in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/63820ba422b644c295326948eaa92e61 : SUCCESS in 5m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3c9b5c310def4bafaf8f4e6c834973b0 : SUCCESS in 9m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8259321c95904efcba089a4d64bd538a : SUCCESS in 35m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/621bc66587a24f7caca6bc39682dfda9 : SUCCESS in 1h 32m 39s","accounts_in_message":[],"_revision_number":4},{"id":"e4860ed4b512c23131f16903515eb2617f736eab","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-12 19:17:52.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2a206a683c1253a5f494d65535e797b22d3b94fe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-13 06:02:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d747975c3b8f02f5e8746a224f236eaf361b982d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-10-19 16:17:46.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"558855412bf7f8e62fc633103cb8a720acad818f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 16:18:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"144ef8cb7563ac96fe41c21b3da707e988016849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:44:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee069e250fca476fbde18d514b84ca29\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfe5df76707549f8ad0646f2f9f7f276 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf3c1f7b12fb4b5bb9f5c10bfc9450ed : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d172a1e766049d08675870796db28b1 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c245ea13bcc74db8ac25dfbd50528a82 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10a26b63b3d54cb4b4fc3329aaa584b2 : SUCCESS in 2m 59s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/387d9c960f69494ca08f51aca6b58b47 : SUCCESS in 10m 49s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/605ab1f290f54bd7b61a0876b2f4f610 : SUCCESS in 4m 38s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/63fb0c95c44943628f1d1dc8d0b34247 : SUCCESS in 5m 16s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/5879a31c09ca45299215c2aac53c16fe : SUCCESS in 8m 15s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/231d039e9f194a16ac22dff899333dae : SUCCESS in 11m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c09be45bac8543ac828618797a249c32 : SUCCESS in 7m 57s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/b667ed4f52ce429eaa5b58c964639d6f : SUCCESS in 9m 12s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8498eb79b575407690d9f2918f864802 : SUCCESS in 11m 02s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/f8f04a45c138472ba6e6f248d2d8b92a : SUCCESS in 4m 05s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/bb41acf14f314ab6b999823e83e56c94 : SUCCESS in 9m 52s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/479399b61ae045af8f79e0be6d8e1fe4 : SUCCESS in 5m 00s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/b53772ef2c254dc7a97822121e4d0b44 : SUCCESS in 27m 51s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/bdd5758130b24a5288c5facb6e94f29f : SUCCESS in 12m 59s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/a7bd3aa31888461994b276e6fb6e8dfe : SUCCESS in 4m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6915c7d844e7400289a562807aac8266 : SUCCESS in 4m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ef4795746064567be60d1503ead6169 : SUCCESS in 23m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/d3fec45d28cb4ce18d7434a54926c0ae : SUCCESS in 4m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/74ff2ef320a0441a9d3d6ca5c47b8ce6 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0aff9af19927408996cabdeb6581ec12 : SUCCESS in 7m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2022f2a706794daba09e1fd476ac0634 : SUCCESS in 10m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d52c774c6af0446aa8e77fb8e9bcb6c9 : SUCCESS in 54m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f91c57cba93c440e9bee739511207140 : SUCCESS in 1h 25m 01s","accounts_in_message":[],"_revision_number":4},{"id":"ffc06e20cd927c3fafe8ef2e5388fa90665c1f59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:45:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a33c52d62ed85d1296eba5ecd3857185a67ea0fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 17:46:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44657e2996774e2ba63c26aeb226d34f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34d96e4dff4f47b7bde1d7693ab5f615 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ba0b95bec368cdc182129a9ee9385d2f23b93189","revisions":{"dbf95e942b6880d4f2d5ae9a7ecc16fd662b82bc":{"kind":"REWORK","_number":1,"created":"2022-10-09 07:17:15.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/92/860792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/860792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/860792/1"}}},"commit":{"parents":[{"commit":"ed729a3fdc3e61696b29ed913c46340dee40be1e","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed729a3fdc3e61696b29ed913c46340dee40be1e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 07:16:02.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 07:16:02.000000000","tz":120},"subject":"Switch cross testing to python3.10","message":"Switch cross testing to python3.10\n\nThe PTI has dropped python3.8 as a platform for the 2023.1 cycle, switch\nall cross testing jobs to use python3.10 instead.\n\nChange-Id: I039171301da04f5005eee8d6c4ff8d94bec6b165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dbf95e942b6880d4f2d5ae9a7ecc16fd662b82bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dbf95e942b6880d4f2d5ae9a7ecc16fd662b82bc"}]},"branch":"refs/heads/master"},"2e412980d67aca4145f831fb29d585d7f6d95b34":{"kind":"REWORK","_number":2,"created":"2022-10-09 08:12:53.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/92/860792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/860792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/860792/2"}}},"commit":{"parents":[{"commit":"ed729a3fdc3e61696b29ed913c46340dee40be1e","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed729a3fdc3e61696b29ed913c46340dee40be1e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 07:16:02.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 08:12:44.000000000","tz":120},"subject":"Switch cross testing to python3.10","message":"Switch cross testing to python3.10\n\nThe PTI has dropped python3.8 as a platform for the 2023.1 cycle, switch\nall cross testing jobs to use python3.10 instead.\n\nSwift doesn\u0027t support Ubuntu 22.04 yet so keep them running on py39\ninstead until they are compliant with 2023.1 testing requirements.\n\nChange-Id: I039171301da04f5005eee8d6c4ff8d94bec6b165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e412980d67aca4145f831fb29d585d7f6d95b34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2e412980d67aca4145f831fb29d585d7f6d95b34"}]},"branch":"refs/heads/master"},"8e349450291df3acee0e31e0b80926a035b23dd6":{"kind":"REWORK","_number":3,"created":"2022-10-09 08:24:13.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/92/860792/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/860792/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/860792/3"}}},"commit":{"parents":[{"commit":"ed729a3fdc3e61696b29ed913c46340dee40be1e","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed729a3fdc3e61696b29ed913c46340dee40be1e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 07:16:02.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 08:23:56.000000000","tz":120},"subject":"Switch cross testing to python3.10","message":"Switch cross testing to python3.10\n\nThe PTI has dropped python3.8 as a platform for the 2023.1 cycle, switch\nall cross testing jobs to use python3.10 instead.\n\nSwift doesn\u0027t support Ubuntu 22.04 yet so keep them running on py39\ninstead until they are compliant with 2023.1 testing requirements.\n\nChange-Id: I039171301da04f5005eee8d6c4ff8d94bec6b165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e349450291df3acee0e31e0b80926a035b23dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8e349450291df3acee0e31e0b80926a035b23dd6"}]},"branch":"refs/heads/master"},"ba0b95bec368cdc182129a9ee9385d2f23b93189":{"kind":"REWORK","_number":4,"created":"2022-10-09 10:12:52.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/92/860792/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/92/860792/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/92/860792/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/92/860792/4"}}},"commit":{"parents":[{"commit":"ed729a3fdc3e61696b29ed913c46340dee40be1e","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed729a3fdc3e61696b29ed913c46340dee40be1e"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 07:16:02.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-09 10:12:27.000000000","tz":120},"subject":"Switch cross testing to python3.10","message":"Switch cross testing to python3.10\n\nThe PTI has dropped python3.8 as a platform for the 2023.1 cycle, switch\nall cross testing jobs to use python3.10 instead.\n\nSwift doesn\u0027t support Ubuntu 22.04 yet so keep them running on py38\ninstead until they are compliant with 2023.1 testing requirements.\n\nChange-Id: I039171301da04f5005eee8d6c4ff8d94bec6b165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba0b95bec368cdc182129a9ee9385d2f23b93189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ba0b95bec368cdc182129a9ee9385d2f23b93189"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}}]}],"submit_requirements":[]}
