)]}'
{"id":"openstack%2Frequirements~861193","triplet_id":"openstack%2Frequirements~master~I0fe79d4eb3d2b8f3f053a0afe7ad24224972e52a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I0fe79d4eb3d2b8f3f053a0afe7ad24224972e52a","subject":"update constraint for openstacksdk to new release 0.102.0","status":"MERGED","created":"2022-10-13 14:56:11.000000000","updated":"2022-10-17 18:25:46.000000000","submitted":"2022-10-17 18:24:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861193-new-release","meta_rev_id":"54a08e2e7a7f8f981520c5267534c90ebac41f05","_number":861193,"virtual_id_number":861193,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-17 18:24:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-10-17 17:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-10-13 15:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2022-10-17 16:44:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-10-17 17:05:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-13 15:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-13 15:10:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-13 16:47:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-17 14:58:29.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2022-10-17 16:44:57.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2022-10-17 17:05:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"657b4a6b0ca9aab63dfb8c6ab59180744cb50d57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-13 14:56:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0643fb91912f86590c15be2746c36deed64386f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 15:08:05.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fc2c292d038471c99a0d52b8d04e97e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/364da9950f14475ab5034afa4e80ba89 : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"5ed17072293f0175bfa7462512d0bdbd0692e36a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-13 15:10:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ddf69ee18cfcd372982697309a86ab605315d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-13 16:47:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3f209ffb07e412ca8ed8db15092e300\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78845859a924881b36693817ea69555 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2256c254b1aa4f13812427f92f41c536 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bdd0b1a9baa47929e50f12e23f996ba : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/984d6b78dc20413698efbf17ead90f88 : SUCCESS in 3m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d71cfad1f874840a7b25a68ff82e4dd : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ecd0508634454f2cb737528dba003a4b : SUCCESS in 3m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c7814e6657334be498b70cd28cb25406 : SUCCESS in 7m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e7131a76145a4dd690b414c53236201c : SUCCESS in 9m 51s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/71b72e36aebe406db085b680804966e0 : SUCCESS in 8m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/eb0ffc0f64f443479b391e4a6e8186d2 : SUCCESS in 5m 03s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/fcab847783f24b2ca344363afb7d76d7 : SUCCESS in 5m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c1861ee3273c4a4a93f345655af04900 : SUCCESS in 9m 35s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/98278076b3a84accbf8fff72f7c322d5 : FAILURE in 4m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0a2d04aee1d44eb2bf9b578af860d2ed : SUCCESS in 4m 56s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/44e0de4c6d7a4fac94f4a306bd3ec5cc : SUCCESS in 7m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3271d2b53b2343c98407f5110095e185 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2489d2d0c85a4aa5b714cca9bbe15b6a : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f6f7d39ef54044329485b099c506a850 : SUCCESS in 8m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c0f3be84d04d4502a12d2849a4eed622 : SUCCESS in 13m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/daa62dfecb9b4a04b49efd4f7dd41c6b : SUCCESS in 3m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/16e8306a6f554477835aed9ac826842d : SUCCESS in 9m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/aa1e351cc0a541ca83a7fb5ab803509e : SUCCESS in 4m 57s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/d3eb5abbd05242de8a1b41e36500ade0 : SUCCESS in 10m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d746800c5114450caabf65c58070dff9 : SUCCESS in 26m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/231b2ab3a0d54918912e3b82a6089bd3 : SUCCESS in 12m 21s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d1ce15ed3341463a8850475612480346 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f52eaa7bfdb649319456ecfaa0d1185d : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb3dc01700af49d69dfa88d5ef96dde3 : SUCCESS in 28m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1ec0b6c5521c442bbe3c9eaa97053300 : SUCCESS in 5m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3fd7e273b7084124ac184ba2e48b68e4 : SUCCESS in 4m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/abdb32e6d88f41d497d4987c293b6c40 : SUCCESS in 6m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d126d715bb974cc4bdb13cc4c9931469 : SUCCESS in 11m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ecae28886a94e22aaf28f0e607273f2 : SUCCESS in 3m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db1647ee70dc4f45a3fda9b39cdc4352 : SUCCESS in 55m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb4849b7bfe94daf9e63b4e8b8699c10 : SUCCESS in 1h 44m 37s","accounts_in_message":[],"_revision_number":1},{"id":"4bf5610eebebb56066f9aae88a3b89d0eefe86d7","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-17 14:58:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"84eb9621bee8331bd6b2fc62b28357d66fb3e35e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 15:11:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7ed2430f0040a488668ee836d0f461\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a80a049215df417cb92d607699252a5e : FAILURE in 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"71ad3cce67fc9f20725908bbb1f07cb6173b6974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 16:44:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/627f1f38196845f8ba353c50a1ed6232\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cb4ae9919c647e8b2a1c9fc1b24435d : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe7f312f554648dca3a2e5b8961f1100 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca6592f63f63412badb3a45c010120b3 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b84b66751f234590ad12361bbd117942 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/26187039ae6541009f69ddce79719966 : SUCCESS in 3m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ec1cbc60eb59417d9df8f6be26b6459f : SUCCESS in 3m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0988b316c95d4e2081a1bc74b406ad7e : SUCCESS in 9m 00s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e0e013fa90c74ff19f0963ac35194303 : SUCCESS in 12m 51s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/4e2079bc9b0743ff98851538827e554f : SUCCESS in 8m 31s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/209d329b5f42481b9f50730654875dd8 : SUCCESS in 5m 43s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/1ccdd1966202496b9279ad120b4df32c : SUCCESS in 6m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/95fd76edab66434283746edaf1788cd0 : SUCCESS in 10m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8c52b8beb53b4a548238f580f9412010 : SUCCESS in 4m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bc511706f3494224a086adb0f907738b : SUCCESS in 5m 08s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/aacbb0792082498a95c9bc17d6b9fb28 : SUCCESS in 8m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fa224c03647f4a468c571470c6cb561d : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3e37a76df9d746a687ce77ded0d8ab9a : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/eda0aedff6ff4b1584729fb7ef63ede5 : SUCCESS in 9m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5a4d36b9db4141039f66a47d75c10509 : SUCCESS in 14m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2c37988d12d5413fb1d3e3c90320e136 : SUCCESS in 4m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/93f41d84d50e478290f3994e39700162 : SUCCESS in 8m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0e48dd4bef7241559d1adabfad578dac : SUCCESS in 4m 09s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/7d787f87c07b4b37a69f547ddde89ee3 : SUCCESS in 9m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e5ca31e5b45c48768fb24b523cf6cda6 : SUCCESS in 29m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20aeae8719b5477bb5be47ced2f2e284 : SUCCESS in 11m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9e17f7bb172a4bec864a03b03228ddb7 : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e3717297e5ec44539b24240629f79d7c : SUCCESS in 2m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b127fcada8648efa12f834b85d6bedf : SUCCESS in 25m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/99b3569d2b14423ab5092d3eb93e1014 : SUCCESS in 5m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/aef96b15c3b543c9ab48dcba516b40e2 : SUCCESS in 4m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/30564aaedea84db6afe80abd61a6f88d : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/92a9e350187b400992ade7fc49cf3084 : SUCCESS in 10m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f615392186d246b8a6dffd00e1b20667 : SUCCESS in 3m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b6f3937312b462aa33fae4239f3e9e6 : SUCCESS in 56m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5397649c2b2f4275b9d347dda59e7e5f : SUCCESS in 1h 40m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6fef9b02ca27bf70e0184d2fc81e659e5579e4a5","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-17 16:44:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"faf7c849f14076d74a1f922ff971617f3d0469bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-17 17:05:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"048feaa665fb771722bc9bec4931e086733a0532","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 17:06:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e87239e2d5c4204e2618ec0038d4185ba5fbe432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 18:24:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fc210fc82cf4478a61f29a1ac867120\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39d65e35b4c4a0f9c37cb9014effddf : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1655a2d2fa254caf8162d48ac2b48da1 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36a28ef630a34a38b0a7178ae2ccca87 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/300e21a48dbd41388337a2ab8b32b3de : SUCCESS in 3m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/de570b6a99b149b5b3d0a30062c3a2a0 : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3706f879c3a54b4aacc2ad2ab30c45b6 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/702a2653228c4c85b1ccb6460d845dcf : SUCCESS in 7m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/723f1bc773b84deeb29cb546d7786fab : SUCCESS in 9m 20s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/240d1bb07f8b40009137cbfc02f67880 : SUCCESS in 10m 44s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/60f6fb0d576c4f97928c9de5bfb2f382 : SUCCESS in 4m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/da02f2ad5b5341879117c21a5add9a8f : SUCCESS in 5m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fd0c318ac5d54c2aa1495fbcbdf643d2 : SUCCESS in 7m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f02de78d9e5143dca4be54b3a5e11396 : SUCCESS in 8m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0bc2cee693da43159a6ea1b0291cd122 : SUCCESS in 8m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e0acec2d7c454d0ab0745c23bfac7346 : SUCCESS in 7m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e96451a23ea54ed3b42cfc9c27cbf809 : SUCCESS in 10m 23s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/df601af010ce4e2c88b2b5ad051153b6 : SUCCESS in 4m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1d7e5c0bc9e544dc8772ff820ab9733f : SUCCESS in 9m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8b1d084720714279ac6aee215583c15c : SUCCESS in 5m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/661a92ed35c742309b0f7215cc5b7950 : SUCCESS in 24m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e27126d7b0bc4b278fb1dd7edacd759e : SUCCESS in 13m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/495ef7c872564551a53233c4cccf7208 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b58468c24bbc44f78544d7a2c1351b45 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/34cd8beb1b7d43b39e21a997d59a8ee0 : SUCCESS in 22m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7e380f287314427ebc2927b3f37ac855 : SUCCESS in 5m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/78be18fb59a54973b025f926c002fd42 : SUCCESS in 3m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72156376087e4a9dbcf3556642371333 : SUCCESS in 5m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/eb62138f056d4e19a9419a1a5987db6a : SUCCESS in 11m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c79d4ee2cbfd41adb4a5138c3c3796d0 : SUCCESS in 3m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46f21999fa9f4dd8871d52a25b85861a : SUCCESS in 54m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/344b26741ec64f15863a5434b0e3f1a5 : SUCCESS in 1h 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c328627cc2b27dfa7c8050f9391a004aa1a70c49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 18:24:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"54a08e2e7a7f8f981520c5267534c90ebac41f05","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 18:25:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50d069a71f2e41c7bb8c515788e718cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad0a583b6424f8e84a7c86d90469347 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ba08b3f3c926bb2515eea2fe8216ec88041fbf5","revisions":{"4ba08b3f3c926bb2515eea2fe8216ec88041fbf5":{"kind":"REWORK","_number":1,"created":"2022-10-13 14:56:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/861193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/861193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/861193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/861193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/861193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/861193/1"}}},"commit":{"parents":[{"commit":"ed729a3fdc3e61696b29ed913c46340dee40be1e","subject":"Switch to 2023.1 Python3 unit tests and generic template name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ed729a3fdc3e61696b29ed913c46340dee40be1e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-13 14:56:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-13 14:56:08.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.102.0","message":"update constraint for openstacksdk to new release 0.102.0\n\nmeta: version: 0.102.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: I6284595825e2f4dfddd5bac81357c82b50aea8fb\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I0fe79d4eb3d2b8f3f053a0afe7ad24224972e52a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ba08b3f3c926bb2515eea2fe8216ec88041fbf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ba08b3f3c926bb2515eea2fe8216ec88041fbf5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
