)]}'
{"id":"openstack%2Frequirements~861771","triplet_id":"openstack%2Frequirements~master~Ia161fbc8d46efa148aef24c5699a298492d00059","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-25 06:53:53.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-08 17:11:13.000000000","reason":"Change was abandoned"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-10-19 08:17:51.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2023-11-08 17:11:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia161fbc8d46efa148aef24c5699a298492d00059","subject":"Add new dependancy: raven","status":"ABANDONED","created":"2022-10-18 15:36:23.000000000","updated":"2023-11-08 17:11:13.000000000","total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"7ae0438c4872624a78ec00431c19364c66534e47","_number":861771,"virtual_id_number":861771,"owner":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-19 09:16:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":-1,"date":"2022-10-19 17:40:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-18 15:37:51.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-10-18 15:37:51.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-18 15:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-18 15:55:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-10-18 16:51:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-19 17:40:52.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"93985fc118518798110e78d86a411945f2ae66d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-10-18 15:36:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8e08dc5758f8f453d7a49252c2a7c53903cad7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 15:53:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8c464484dac44c6b6ef659c78cc5c67\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3d180a11eca142a6bed87ebd370a17ab : FAILURE in 10m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ec1d81d7f30500b0fe0868488561de895aac0828","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-18 15:55:15.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d8b9ddb698d85b9a00758860fde026f23b7e2783","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-10-18 16:03:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"114b35e38d0e8447417a7509107d0783a3a3b0d7","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-10-18 16:04:51.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1e85fdbdd9ec65fc35e28a2456dda13796b583b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:51:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/131e8859c5244b4ba146d6dbab2d8aa4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3420cc6584d4419b89fb34c46546d72 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6f76a72d7ae40b18ed953381ae88211 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82107a417ad043fdb03a33d8a4f71319 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eed2cee163324fbf94bc7d017b1da0e9 : SUCCESS in 3m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a20a3d48931b4d7989525b98a7ef2802 : FAILURE in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3b13efbe383454c872a9b039bbea353 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/563fff3915ba4ae6be7f90e07caa5460 : SUCCESS in 8m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cd4fef770ef34c23a4a4f9afe672720f : SUCCESS in 12m 23s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/28773a99284d48e588a904b3ad275f34 : SUCCESS in 7m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/abd1d3d349d445e08ea6fcd20f7465dd : SUCCESS in 4m 35s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e07fe9053cf84bd9bf226851587b32bb : SUCCESS in 5m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5071fbc85f0545abb8a123019668a1c5 : SUCCESS in 10m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4e4f256eddd24287a4f449fd9359134b : SUCCESS in 4m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1c3b1ee7d09a493a844e78a870c54d3e : SUCCESS in 4m 42s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d0454f21e61447e0ab35dcf68b5abbbc : SUCCESS in 7m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c6c153d92a2744869815030da6babc78 : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8bfa9c94097a4960969d44e6f2200585 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/21e90692c2574d7dba0d22ae347266fd : SUCCESS in 8m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e191374d8eae477786b003e2d02265f8 : SUCCESS in 10m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/56d95830f68d43d3b92d2fc26c678cef : SUCCESS in 3m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/58ddf0f4b76748439a5cea92a0148c05 : SUCCESS in 9m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/60ec1069fb99433cb5f07d7a2dadca80 : SUCCESS in 4m 23s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/880a6251d14a4f04b5eff21a13110388 : SUCCESS in 7m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/59b64f1f596f489cbe33bb3b55c0e71e : SUCCESS in 22m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f7e387428c424709bafbc3cf0607ef61 : SUCCESS in 12m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/377090a1d19d433abaaf6c0c06590a72 : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0d940c903b924138a7b84977edc5dce5 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad0bb71c83ea4e49a5b2b42a67eb9ab4 : SUCCESS in 25m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1b66b249de9a47128b71783f43340238 : SUCCESS in 5m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f557e7b2c57d4d03b37793ca41f68fe0 : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f2a3c901dc3434baaea4320c1f15fa9 : SUCCESS in 6m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2af1777b596c441b8aefa4fa14f6ca94 : SUCCESS in 13m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12b55a463afa46368725a0ae44b2575d : SUCCESS in 3m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbd0c733b22c4c81a987066cc911d348 : SUCCESS in 54m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5e2086ce3af4a5baed23ab6b3fff0f2 : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c88b72c5ee80869038c2dcde06f570227201c8eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-10-19 07:25:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f102fdd6012d880b4ad8ae11c718b4918199e8c","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-10-19 07:26:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a822d94af9a54fecd06659a349495a3023056918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-10-19 07:28:09.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"179b4806ad7a66e35763e57b8a9e03869f1fab2a","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2022-10-19 07:28:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"015fbff8d31ffa39f7397a0039e47561536ba2ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 07:42:07.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17114ab2b075438ebe35821c7609723f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f9d2fbf8f37e4ccd88c4e1374b40117d : FAILURE in 10m 16s","accounts_in_message":[],"_revision_number":3},{"id":"3bb8b0c8aeb989e4fab350f7cd1b409c68f655bd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-19 08:17:51.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"205c9b344cd6b31c979b534e3c4a19efc62ab362","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 09:16:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99b60e6d2dac4b45a92a1204feb573b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac45c095a5e1492a8f30c0f589559336 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4941a58f53f04116a46567391538bd99 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6333e224565b4de690b7fbb5b7aae5e7 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e59694cd7694ae38bef6cabb856d3bf : SUCCESS in 3m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/18f167f3281640939ffd377f5f1b0337 : FAILURE in 2m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f379cdf166214513aa2ed0a901ca0125 : SUCCESS in 2m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83a9e223dad04eb8b75dcf97a77467c5 : SUCCESS in 8m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9cf5b48f1e5a4b158bdc81f029b53ce0 : SUCCESS in 12m 50s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/c1a88592efaa471cab65ced3659eacff : SUCCESS in 6m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e69e33b2fdf1447b8afa98dbe6468aad : SUCCESS in 5m 06s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d5277477b5a1405cace5d600109a1391 : SUCCESS in 5m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1d576b662475452c9468bac6ea536dc4 : SUCCESS in 9m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7fd4fe9a87ab484ba301b02ed49bc7cb : SUCCESS in 4m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fd596210b0e24208a780eb84ed3eeb12 : SUCCESS in 4m 30s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/299c5e31351b4da0b2406b11a026f824 : SUCCESS in 6m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/11d69ec6179f493181b1fe19707dd13f : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/370f163a18dc453380ab79daf90b6f5c : SUCCESS in 7m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/271d79a2dbd54a04a5a07f35665f5f4e : SUCCESS in 8m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b47b93518ebb48f4864c0c38c7fd9ba1 : SUCCESS in 10m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2dc5735932404991bdac91a680e77ebb : SUCCESS in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0e02b6888e5241619ba3c2c91d720038 : SUCCESS in 9m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3d74eccae4324968b0ea7b8b1d8a946c : SUCCESS in 5m 12s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/43cfb090fa78489fac26da5875a5e0df : SUCCESS in 10m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/839a7e706fea407db24a419cd2724b84 : SUCCESS in 22m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ff1f9c1da736437a94484df896fe1d94 : SUCCESS in 11m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ddd057af60074b88a0a5c6dd53b7eafb : SUCCESS in 5m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e177c1975892426693528ed1cc7f7248 : SUCCESS in 2m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0ca561ee638548a0bc6bda641e7aae25 : SUCCESS in 20m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d805af6077894de18496dc9385e45139 : SUCCESS in 4m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1f8964c47fa3490e91589a951ec0d044 : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4057558bad894056804b43e173fe5589 : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7a23328f7d254293b7cf21d674009457 : SUCCESS in 11m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00301a3d92844a7db3be57147e696b05 : SUCCESS in 3m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0341f8f033949938bb5f91e5c764be0 : SUCCESS in 45m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d0058ef86824bffa41d805340d0f2d0 : SUCCESS in 1h 40m 31s","accounts_in_message":[],"_revision_number":3},{"id":"6dc0b20719d0b81e3b0aa3a06cd5d596e7d5d614","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-10-19 17:40:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ae0438c4872624a78ec00431c19364c66534e47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-11-08 17:11:13.000000000","message":"Abandoned\n\nThis has change has recieved no respose and is more than a year old.\n\nPlease restore and update if still needed","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"22ceadb3aed308974f6f1ee0d2b03fb967264d09","revisions":{"0160db44a1702b0c4e91811b7039caae34751158":{"kind":"REWORK","_number":1,"created":"2022-10-18 15:36:23.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/71/861771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/861771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/861771/1"}}},"commit":{"parents":[{"commit":"4ba08b3f3c926bb2515eea2fe8216ec88041fbf5","subject":"update constraint for openstacksdk to new release 0.102.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ba08b3f3c926bb2515eea2fe8216ec88041fbf5"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2022-10-18 15:23:21.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2022-10-18 15:36:20.000000000","tz":120},"subject":"Add new dependancy.","message":"Add new dependancy.\n\nRaven has been introduced in oslo.log to add handler for sentry.\n\nChange-Id: Ia161fbc8d46efa148aef24c5699a298492d00059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0160db44a1702b0c4e91811b7039caae34751158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0160db44a1702b0c4e91811b7039caae34751158"}]},"branch":"refs/heads/master"},"202482e3f38ed6146bbccbe9628f3cd0824d7e43":{"kind":"REWORK","_number":2,"created":"2022-10-19 07:25:20.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/71/861771/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/861771/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/861771/2"}}},"commit":{"parents":[{"commit":"4ba08b3f3c926bb2515eea2fe8216ec88041fbf5","subject":"update constraint for openstacksdk to new release 0.102.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ba08b3f3c926bb2515eea2fe8216ec88041fbf5"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2022-10-18 15:23:21.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2022-10-19 07:24:51.000000000","tz":120},"subject":"Add new dependancy","message":"Add new dependancy\n\nRaven has been introduced in oslo.log to add handler for sentry.\n\nChange-Id: Ia161fbc8d46efa148aef24c5699a298492d00059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/202482e3f38ed6146bbccbe9628f3cd0824d7e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/202482e3f38ed6146bbccbe9628f3cd0824d7e43"}]},"branch":"refs/heads/master"},"22ceadb3aed308974f6f1ee0d2b03fb967264d09":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-10-19 07:28:09.000000000","uploader":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"ref":"refs/changes/71/861771/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/861771/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861771/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/861771/3"}}},"commit":{"parents":[{"commit":"4ba08b3f3c926bb2515eea2fe8216ec88041fbf5","subject":"update constraint for openstacksdk to new release 0.102.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4ba08b3f3c926bb2515eea2fe8216ec88041fbf5"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2022-10-18 15:23:21.000000000","tz":120},"committer":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2022-10-19 07:26:28.000000000","tz":120},"subject":"Add new dependancy: raven","message":"Add new dependancy: raven\n\nRaven has been introduced in oslo.log to add handler for sentry.\n\nChange-Id: Ia161fbc8d46efa148aef24c5699a298492d00059\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22ceadb3aed308974f6f1ee0d2b03fb967264d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/22ceadb3aed308974f6f1ee0d2b03fb967264d09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
