)]}'
{"id":"openstack%2Frequirements~861967","triplet_id":"openstack%2Frequirements~master~Idd9f48fcb393e1eb7ca66803f9addcc2b1e19b89","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idd9f48fcb393e1eb7ca66803f9addcc2b1e19b89","subject":"update constraint for stevedore to new release 4.1.0","status":"MERGED","created":"2022-10-20 10:45:42.000000000","updated":"2022-10-20 17:17:41.000000000","submitted":"2022-10-20 17:16:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861967-new-release","meta_rev_id":"bdd2a7c539bedb18c36bfddb8af451ddc2bf5b1b","_number":861967,"virtual_id_number":861967,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-20 17:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-10-20 15:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-10-20 11:14:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-10-20 15:39:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 11:04:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-20 11:14:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-20 12:17:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 15:39:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"69bf41aa46a3b19ab3da32e1dc8ef8e3daff3e47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-20 10:45:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06cf50c63fda35854c51227e0598132e0344170","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:04:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3283a3d53b64e1f8731948181f7cbd9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a44f381bd3914ab4b4633ba7a286ed03 : FAILURE in 10m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9ec06416633cb8b5bd5db8c7012fa7e22e4f4c00","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-20 11:14:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1b86f20c0d1ee1dc01a40f9f46c4eaeb8344604a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:17:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8169a247d51742968a4fe0b156cb0236\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a7c23a4825443948b3723b786b3ad55 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c67d8f4e787410fadefafaf4c1b1089 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d947d162a58a46229571e2a1d4820fcd : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ed765245fe14c94bece84000f1e36d7 : SUCCESS in 3m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ec2f28d88d8b493b84838ccbeda8665b : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a683b7868254ac09f1948ac5218b57e : SUCCESS in 3m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd514b6f21ff45829602a7358f306adb : SUCCESS in 8m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1d8f4597f04d4b45961e4fc3b3459f0b : SUCCESS in 12m 27s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/057a3b2b220747b3af67d9cdbfa6e155 : SUCCESS in 8m 38s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/71a7b4f18d6642ddb03c844c478a85d9 : SUCCESS in 5m 06s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/41a049267ee342ddb57cdfcfb649069c : SUCCESS in 5m 40s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/1bfcc592dd484392b24ade0aa66b1535 : SUCCESS in 10m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/fe52eb275db54510a9cbb7f8d57f19e2 : SUCCESS in 3m 55s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/8b4c34dd958341918324ab5fa982803b : SUCCESS in 4m 31s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/08c3d9ea08b8402fa6d2b7c5284e8ded : SUCCESS in 8m 17s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/f4c349b341284075a6eb876587ea706e : SUCCESS in 11m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7209d78681b4cd1a90285f8fd7f8f6b : SUCCESS in 7m 11s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/ffcdbb1326a346f6b9d37bfb5d31c62e : SUCCESS in 8m 49s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/2fddb510939b40cb84e8ef90cba8af39 : SUCCESS in 10m 50s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/a1a817514ee24e9b87313880d61ad02c : SUCCESS in 3m 41s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/eb695ac865ae4079a28319ba9f13849c : SUCCESS in 8m 10s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/400ac01141d34a73aa7163174296e8da : SUCCESS in 4m 22s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/d87a9c72e19f4ac586ec6fc8f65b9d31 : SUCCESS in 8m 48s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/86f3b5db6dfc474c8880a76bbd7b8c7b : SUCCESS in 24m 48s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/21ae0f4b3b344424a652e6b900dc7529 : SUCCESS in 13m 05s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/10b90994c299485eb3a1ded88e515575 : SUCCESS in 4m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c165d6d5e624f6e84100161221cbef5 : SUCCESS in 3m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35ac53469acf4dd48f09475798ca6e53 : SUCCESS in 20m 57s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/94f129bafdd24f768bbcd0aaec0996ea : SUCCESS in 4m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/9ba1dd87696d4a9b8e76a6e451c06bc7 : SUCCESS in 4m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3fb73497677e4ec8947bb6dc89fe5155 : SUCCESS in 5m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/37b03589ffe844778090f14642fe2d8d : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1db04e0dcf69460ea034841dc3de528d : SUCCESS in 49m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2dc5bd494d764c7f9013bded22b79ed8 : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":1},{"id":"cbbaf185e92730c596d308b78b847ae1e7d2dfa8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-20 15:39:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2755ce0a3f03541402a17f542f51f877a8a7d7a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 15:40:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"753d4da93d5e33b9d6e88c4d57da97be940fb81f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:16:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1015a93f87934502990825b572ac23ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8128f33df44bcba2b5bab251e28ab2 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a08de431b4743ef89b30fc51e7f94cc : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e0f9df44b8a421f88945bf8ddd83768 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0569bb25ea44e8a8f90f3744519b6af : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/477c413aa7954fc0a9303c347c97b043 : SUCCESS in 5m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b34f5ab268ea418c8b6cf7626b62afa8 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/711a755dff9c4998b5378995d07cac42 : SUCCESS in 9m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f3419e67c98b40199164d71fbe28634f : SUCCESS in 12m 41s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/af88d84ae13b4d71a10af2832231b014 : SUCCESS in 10m 47s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/7a657c6fbada4c5b90b1cdd0049536a9 : SUCCESS in 4m 59s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/087e2a59d7634cfea90ede9b62817988 : SUCCESS in 6m 46s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/71169b9a98f34e939067df54df46a096 : SUCCESS in 7m 49s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/82a1f2534e6b4fb588674c8bdbbe2a9f : SUCCESS in 9m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/26a7a158cd0c46778978dbc6aa151c5a : SUCCESS in 8m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/71f450cda0cb4244bb27294dba3ccdb3 : SUCCESS in 8m 42s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/4424f9fb48934fde9d54b872dadb7260 : SUCCESS in 11m 35s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/77b4467c314a4518862335e459d3d4f8 : SUCCESS in 4m 01s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0450df87801146a2ba227c2019a99663 : SUCCESS in 9m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3f53fdfb9ad24413a1e7ba65fab73243 : SUCCESS in 6m 40s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5beeafa590e94acb932f164bcbcae5fb : SUCCESS in 26m 21s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ea6fa459f6d44dd7a4c321339f55b871 : SUCCESS in 13m 36s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/fd4200ff0cdb462cb1f716d70d919b9d : SUCCESS in 7m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5133b127025441c682eed0afd93b04b2 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/46a19d19d77043488db49137a74754cc : SUCCESS in 20m 34s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/063719a494fa447b94a1f7363806389a : SUCCESS in 4m 56s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/abc5bb81b6384d09a0e634919d95d270 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0f2a09cbe704c7eb5dc945e70c9f045 : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6363862ec76d409386f12528aaf0e83d : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77f2b6730a914a8b803b959a1dc9fe55 : SUCCESS in 42m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7797b828b0da4785bc541d56bd6bf61a : SUCCESS in 1h 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"62e741d8462d7e5afcbd0c515e3b0c931ff5516b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:16:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bdd2a7c539bedb18c36bfddb8af451ddc2bf5b1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:17:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7385658fc79a49c288004d31c7d01c6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15d4168dbbcb4fddac61b23607a0af4a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9017475d6ef717777beb1f0ca5b10361bfd3a249","revisions":{"9017475d6ef717777beb1f0ca5b10361bfd3a249":{"kind":"REWORK","_number":1,"created":"2022-10-20 10:45:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/67/861967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/67/861967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/861967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/861967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/67/861967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/67/861967/1"}}},"commit":{"parents":[{"commit":"bd6f6255fd8761185a5c9a936d0a3712e4e047ae","subject":"Merge \"[tests] to pass a list to _is_requirement_in_global_reqs()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd6f6255fd8761185a5c9a936d0a3712e4e047ae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 10:45:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 10:45:41.000000000","tz":0},"subject":"update constraint for stevedore to new release 4.1.0","message":"update constraint for stevedore to new release 4.1.0\n\nmeta: version: 4.1.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iade8b6d4c5bbd8397747f0619b619894365b832e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Idd9f48fcb393e1eb7ca66803f9addcc2b1e19b89\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9017475d6ef717777beb1f0ca5b10361bfd3a249"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9017475d6ef717777beb1f0ca5b10361bfd3a249"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
