)]}'
{"id":"openstack%2Frequirements~861969","triplet_id":"openstack%2Frequirements~master~I749cd44cccbae01439dfbcce64c24361bd44ab62","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I749cd44cccbae01439dfbcce64c24361bd44ab62","subject":"update constraint for oslo.db to new release 12.2.0","status":"MERGED","created":"2022-10-20 10:52:43.000000000","updated":"2022-10-20 17:21:11.000000000","submitted":"2022-10-20 17:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861969-new-release","meta_rev_id":"e6424e85c013e4373de44473a458281344049bbc","_number":861969,"virtual_id_number":861969,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-20 17:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-10-20 15:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-10-20 11:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-10-20 15:39:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 11:15:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-20 11:19:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-20 12:19:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 15:39:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"038799d5a84e9b3af81d8f0fae3617ee4f1ed048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-20 10:52:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a31186ac2de6e36c36a2d9a9eacaf9e8a813d20a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-20 11:15:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d662f4825d9b404b361b36693434f03059d5eeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:19:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebf6c47f279c46ebba2f09048501613b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/77dd1729ea69419d9df255fa050e1e6d : FAILURE in 8m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c2f7993b91e52acb85e57e5248a59d6845147373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:19:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2c9dfafdac24887bc9b8ed5060e8e56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c6c321838894349b24edf76bb9d5a2b : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee60eccffa2436ca3a7d762cbeb54b7 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9794134ae7794c09bd6474e1137679e8 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1af6d37f36094b3bb6b1daea2690af6d : SUCCESS in 3m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0f2f75fce074f75b0d4c85155c4eb76 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96be3b18000c4ad58fa6f3bd59c3a1cb : SUCCESS in 3m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56305805e59844969de43b4034316f21 : SUCCESS in 8m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e2b81714a1e24ce69d4f53cb68e864e7 : SUCCESS in 11m 48s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/ec6bd4ad522c41db82b3474ebd10a641 : SUCCESS in 6m 54s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e351ae28631e4633bc9f95ba20a0250f : SUCCESS in 5m 16s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/e3efab27e5e945baa9499b0fe2bd74ad : SUCCESS in 6m 01s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/ded124ad03804792a0eec91acdd64e94 : SUCCESS in 10m 12s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/c1b2342fd3524350b86ff38250b9b2c4 : SUCCESS in 4m 12s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/d9db2fde998e4644a71e363379de874f : SUCCESS in 4m 54s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/9d6f3598ee464ea8b3334ae0e66a60a2 : SUCCESS in 7m 45s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/9cfd38af7b9a48edbd46c198cd6c44a4 : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/25bfaf89af0a4ff8941578fa25cf7f92 : SUCCESS in 8m 26s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/7512ae28af104ba983f31cc5f28dc0a4 : SUCCESS in 7m 22s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/948dc503007c477b9f581e5454dee084 : SUCCESS in 13m 16s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/0f945918c2314a3e9fc1b3db570af348 : SUCCESS in 3m 36s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/0b4e9978cc4142e582f7f003287d3712 : SUCCESS in 10m 26s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/c2e6f28d144d484bb974d947c6368361 : SUCCESS in 4m 00s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9d8eac527ee4495cac644b1d3b6932a4 : SUCCESS in 8m 40s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/5338479ca3274bf0b37e62cb52dfd04b : SUCCESS in 27m 00s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/8de9a081a7e74b2694c149f8098619be : SUCCESS in 12m 47s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/9bf02cda05a84c048a2069b21b564e93 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca206a27ed3348cc92f677116c53046d : SUCCESS in 3m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e211d0fff6ed4ed88d8473a14b8169ba : SUCCESS in 19m 26s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/19db9be9108b4b909d0d18b3bd084075 : SUCCESS in 5m 43s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/8c06c96029614d24a2aa3dfc7e34b703 : SUCCESS in 4m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04eb859e05084b2c88513d4b3a219873 : SUCCESS in 6m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/103cb8fa648b4ee4897dc60d9314930e : SUCCESS in 11m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a727302937fb45b7b5f4887c9aeee10e : SUCCESS in 56m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0aa5f26bd6094b36a285d98d812b6e5f : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":1},{"id":"64d85463aead64127c27b3b985bcef917cf996f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-20 15:39:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba619c3bcc5846b63caba811a24c1dc6014df6de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 15:40:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40bf4a3478b03b360295ce1ffb0afef7e2a034a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:16:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afd5e4a45948453aa874b443be2c35cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa7f1df34964d4b94bc1aca7d6e5ecf : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e30b2028ab1408995e76dba81af455b : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdca265d983e4c1e80261d10098f472f : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68f77406364247748112e61592d94f0e : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10fdc623f87548a09cc63014ed1ed88a : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dd13981e8a3340e99ed309b4b3904d81 : SUCCESS in 4m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/383eeb1f8ab64ec489cd923987731b76 : SUCCESS in 7m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7901378e86174cb8acb1558eb631c10f : SUCCESS in 13m 13s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/f1cbff24abf44d17990f0e9add491945 : SUCCESS in 10m 18s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/91fa57720bd44826a96d072c2d807603 : SUCCESS in 5m 40s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/4e7036a469a84a568f103e25dab697da : SUCCESS in 4m 58s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/140c8baa401148be9b151cd0038b8273 : SUCCESS in 8m 18s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/38ec179b63314993b9c3fcc4994b1efc : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7c4d762d44824c41949c88c9f4395adc : SUCCESS in 11m 42s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/d75e8200686f4ffba41d4314881da640 : SUCCESS in 9m 01s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/8a2bcfa791b242bc9c677470a9c6f094 : SUCCESS in 13m 19s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/e38affce2ab9460e81d6568673aa8875 : SUCCESS in 4m 13s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/9a11dc75099847928861893311ac5593 : SUCCESS in 9m 27s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/3a4bd7f9490149519f4240732607b1c3 : SUCCESS in 4m 35s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/6e91fba0744d4cb3bb407099ab87bc16 : SUCCESS in 26m 07s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/164beca6cbab4277940965353fdf6c14 : SUCCESS in 12m 19s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/c8faca82243c4eae950df92a5d032ccb : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf63e29ddc154135880a7a8066cf7238 : SUCCESS in 3m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fbedf34adb1242f081436e1b1d65c0b4 : SUCCESS in 23m 43s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/8325e5971a0e4a9f9638b2382613f7b9 : SUCCESS in 5m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/d3682f45bdf749508f359c19d2f92632 : SUCCESS in 6m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99645332e0ef43f3b2506f012a5aea24 : SUCCESS in 6m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/291127e617ed4c368456df87fc589636 : SUCCESS in 11m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65ac763d6225426da730ca66f01b843e : SUCCESS in 40m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20ac8e4d8c364c7aacdaf3215d848eca : SUCCESS in 1h 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b32dd12dc92a3aeae7476504dc502f8f96fc3869","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:16:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e6424e85c013e4373de44473a458281344049bbc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:21:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81136561ad8f445d98aa9735a4021aaf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150d7542899f49db817e45c48985de4d : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aee46ff7e8aeffdcc466e38f803c243457c87373","revisions":{"aee46ff7e8aeffdcc466e38f803c243457c87373":{"kind":"REWORK","_number":1,"created":"2022-10-20 10:52:43.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/861969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/861969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/861969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/861969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/861969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/861969/1"}}},"commit":{"parents":[{"commit":"bd6f6255fd8761185a5c9a936d0a3712e4e047ae","subject":"Merge \"[tests] to pass a list to _is_requirement_in_global_reqs()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd6f6255fd8761185a5c9a936d0a3712e4e047ae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 10:52:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 10:52:40.000000000","tz":0},"subject":"update constraint for oslo.db to new release 12.2.0","message":"update constraint for oslo.db to new release 12.2.0\n\nmeta: version: 12.2.0\nmeta: diff-start: -\nmeta: series: antelope\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iade8b6d4c5bbd8397747f0619b619894365b832e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I749cd44cccbae01439dfbcce64c24361bd44ab62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee46ff7e8aeffdcc466e38f803c243457c87373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aee46ff7e8aeffdcc466e38f803c243457c87373"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
