)]}'
{"id":"openstack%2Frequirements~861971","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I3dce496037469007fe87ae7a2cd965aee31b052c","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I3dce496037469007fe87ae7a2cd965aee31b052c","subject":"update constraint for stevedore to new release 3.3.3","status":"MERGED","created":"2022-10-20 10:53:49.000000000","updated":"2022-10-22 03:56:11.000000000","submitted":"2022-10-22 03:56:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861971-new-release","meta_rev_id":"b180cb77159eeda4f549f11dde86a0d114f882ac","_number":861971,"virtual_id_number":861971,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-22 03:56:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-10-22 01:45:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-10-20 11:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-10-22 01:45:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 11:16:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-20 11:19:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-20 12:28:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 15:39:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-10-22 01:45:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"237c8ce9d81bbbd3dcfe609809cc763d7282917c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-20 10:53:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c18983d8d278f7692fdbdee752e7a0679502653","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-20 11:16:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb678d566c6918319eed5a4f766491049b250830","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:19:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3b6fdadc04341fd953e9dd8a9ced33e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/af2ec96366c6447caab8a5e6840da1a4 : FAILURE in 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"b4a33ce1b856fb436b903d04892211af7599e1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:28:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30fd6dc42cbe4a80a31ed16b1cac55d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31e4cdc233f84f5381483c5459a34879 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/680fcc36909c46ceaac29a27cba6f699 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7cdb7eff35b4cb79382eb808cdd9258 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36f4f0de1b294eaba1fd0751e85f541a : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/97f9c1fa5a2c4fa79fd8fb09f2ae6b07 : SUCCESS in 3m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12062f6c7b3c45869829d9bb94a8be37 : SUCCESS in 3m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3081a5b7ea0649f7a22457effc4fc621 : SUCCESS in 10m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f12c79c1725241099fcccc64a0a6eb93 : SUCCESS in 8m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/f8fc7fae265d4b8f8c69c6ec0eda8deb : SUCCESS in 8m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/b05f29d69d184d6d9412f2f324f3b48e : SUCCESS in 5m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b182a2e9232c46efaf9d7365a254ef40 : FAILURE in 3m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0dbc6a082b62466a8b9cc1c8b21e8ac2 : SUCCESS in 7m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9dad791defaa4e6ab5cf37f212bd5e5b : SUCCESS in 7m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e669156bb84f403584301bbc65fb7df7 : SUCCESS in 13m 59s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/4773eda3afa84ca38a7bab735cf4f450 : SUCCESS in 4m 21s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c2158734f43d46c4a41058901acff91d : FAILURE in 2m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/445c29321bf249a3a939952ef55fcadd : SUCCESS in 11m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c5b74952ff6f4e538d51ec024b50b47d : FAILURE in 2m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/752b9521e6fb4e318c9bffbf4bac3540 : SUCCESS in 16m 40s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e6708c32348346ec9c667ec5df8b856c : SUCCESS in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/41c62ab2578e40c8a7b09eadb4be16f2 : SUCCESS in 5m 32s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ef3a4c5585714bdfb254f83b028fd353 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/627c5537f6484781be8b235c22713e6b : SUCCESS in 1h 01m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdf653276d074f45baacdb3b633fade1 : SUCCESS in 1h 32m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a968b046162a90ada6203f5b21d6378009029d82","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-20 15:39:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e727f18b59d9324256b186f4518a0707ba5297a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 15:57:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/995e375272db410793f2686d29be8b12\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2c5daf69f5ec4fb49d1046f8f4def6fc : FAILURE in 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3b3995e5b2bcafe32f87d63ab03de62258c5950b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:30:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70d0e6abeeb44e2097dfd4dd3e81fc3f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02de4940c10f46bbaa570377add24b96 : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20b253d7944749988a1d760780d704f7 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f0d9bf93fcf4585a93cc636eea611b2 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb777fc304b5405399c3703dd238023c : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac102389f2d143efb05d1c6df4bf0920 : SUCCESS in 3m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6fac9b256238472d8940180169405fb2 : SUCCESS in 3m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a5a90711e1b7438b8828fc135c3e6c26 : SUCCESS in 11m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f12ad29fcd5d48b0a34c8988e2a70466 : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/04f67c1d2cf744cc8d888592e7efa9dc : SUCCESS in 9m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/798c8f4c109146a294a67d7e7a336082 : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/b76b978572e4473689e2e1bd99265411 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2e853181a0d34e25a0589b55f2ad0f11 : SUCCESS in 8m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/61e4ad01296842fdae64855e9f996947 : SUCCESS in 7m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/4804ab4eadc540328d8fa9ac9dba2509 : FAILURE in 13m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/8447f5a800314fb5a200d011d57c087b : SUCCESS in 5m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/38c6750bb73d4eae9fe06cbb5d9aa45f : SUCCESS in 33m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/522c11fb9b4449d78abf247245ad4f85 : SUCCESS in 13m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72a0691b13c940589859200524f55c01 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/457501b9346e4863995d61332ec10777 : SUCCESS in 18m 32s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/96e08838c6664686a99c55ab6de3b019 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b4e826546edf4adbab2ccfc0bdfefbb9 : SUCCESS in 6m 29s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8ed7aab7975f4669a12afcbc374070b3 : SUCCESS in 12m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdda1a092b524606a81b43524b6a6ccd : SUCCESS in 1h 01m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a30a22e84ed74df8b25f3e5ac6c7f0c9 : SUCCESS in 1h 40m 37s","accounts_in_message":[],"_revision_number":1},{"id":"246ba81c676154df7b1c619652a5e8a0c9331a47","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-21 23:44:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1954650dfa97728e12cb2bf5ed8d5f708f09a663","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 23:58:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b23f7960e7fa408c9af7ceb9dbe95f81\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/31fe1d71c741433ab16727d5cb8863f9 : FAILURE in 7m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1a71a4781c097ec2122839aa662f85bbeab7a908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-22 01:25:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb9854e05255447ca48d0f97446066ff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e05911f85e409e8e7691e1b95c2afd : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0db814b8992b461c8aec93559014ae0a : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d642807ae0544edb0ed0eefe6577472 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd0ac6bc59714f9d9465413b0c46d085 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/044d0e8f98994866a104230caaa337be : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ef3b436580fc4bef8d03746d31c7f25a : SUCCESS in 4m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/84b6b27770ed4d3f9a333c91a7c61ea0 : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9fa6cad7b2a549f08a94a1e9eab07b5b : SUCCESS in 7m 51s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0a7b47c0759a4a6ba369d97cb1edb0ef : SUCCESS in 9m 10s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ecca4f4bfa214af2a91d1f4f8aa27561 : SUCCESS in 6m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3bbbc4c364764e4f9736aedc29b2c692 : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e0e78c5224f4ccf89d02366514911aa : SUCCESS in 6m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9fa210a5daa148f3a77439467a33d612 : SUCCESS in 7m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9a1e2a3a430d47a2a4c0cfb3c2cfd89d : SUCCESS in 12m 12s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/13da90998571406f82a635705fffb8b1 : SUCCESS in 4m 24s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/02c5802afb884abe96156157abb3036b : SUCCESS in 32m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3ba7c437833c4f9ba7f41dc79867318e : SUCCESS in 10m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8acba1ed8c0b4b0fb4ffa18466e53f54 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3e72618f2abb414eb214d3a6e50ab023 : SUCCESS in 20m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2ac9bca25e3f4c1b8a3e4f552e90f1d4 : SUCCESS in 4m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2b980ccdf284e899cd80665c0ebf484 : SUCCESS in 7m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f9f6dab32e9446e3a352da7c3f79df89 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79f114ebc1cf46648deb669b0634d02a : SUCCESS in 43m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2aec8cfd8604aa990cfaeee35925379 : SUCCESS in 1h 36m 02s","accounts_in_message":[],"_revision_number":1},{"id":"021c220999be98e611baa1ef67d1277942877904","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-22 01:45:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3cfe7e6422db8b184dda0057aa903bb4f24bacb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-22 01:45:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47138f69054774e4c47c38dea4290f388e0b0e6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-22 03:56:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a0f4a8924f4a0681cb91494ab366a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14096c6137af403cb8ae609569a6d72a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c73c70d0bf44e6a8be9c7a3885d019a : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a761abc73604073aa657807e9b2cb8b : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/59d2e301e74649faaa15f173d4f05499 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9089caa059ab432f89b48fd52b618fd6 : SUCCESS in 4m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0bde94c8304047889a064beff08ef07f : SUCCESS in 8m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a2d5cd5e987418393bd78ea47cd0045 : SUCCESS in 8m 55s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/67ea1faf6ddb47b9af10ff3d0193ae1c : SUCCESS in 8m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/7831a792077e4d3fbbe41d49671a2f6b : SUCCESS in 5m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3ae749337980430d8225bbbccdc4addb : SUCCESS in 8m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cde38b258825434381b10dd167b98d34 : SUCCESS in 8m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/91ccddb3344a45eb848592589d193b88 : SUCCESS in 7m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5ede0a1e5cad45fb934dd30e6a914904 : SUCCESS in 13m 23s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/dfc4cbb60c93470aaa9dcecbe338b902 : SUCCESS in 4m 19s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5d2f60ea824c40479eedf28d79ca974f : SUCCESS in 33m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cdc6008a3e6d41e29eac76e52a88dd8e : SUCCESS in 11m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c43f980d4946431a9b5108c19807a991 : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7551cf378a71413caf8597ae1165eab3 : SUCCESS in 19m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bc94d56fd5c54c3ebb4c2cc96e25c76b : SUCCESS in 4m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d3b8c0d71f94f5b8bda4ac3ff939d6c : SUCCESS in 6m 38s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fa3700fc5e564eb3b5e6b5a471191365 : SUCCESS in 10m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b296103d113c436582b235015b48ae2b : SUCCESS in 43m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/514811fad936431486670e40228db713 : SUCCESS in 2h 03m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b180cb77159eeda4f549f11dde86a0d114f882ac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-22 03:56:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9272e7ef21b5a38a8b95e63d5ac31045e039556","revisions":{"b9272e7ef21b5a38a8b95e63d5ac31045e039556":{"kind":"REWORK","_number":1,"created":"2022-10-20 10:53:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/861971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/861971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/861971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/861971/1"}}},"commit":{"parents":[{"commit":"b8b8783681801d1a9ba4814e77bfa6f786f33031","subject":"update constraint for stevedore to new release 3.3.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b8b8783681801d1a9ba4814e77bfa6f786f33031"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 10:53:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 10:53:48.000000000","tz":0},"subject":"update constraint for stevedore to new release 3.3.3","message":"update constraint for stevedore to new release 3.3.3\n\nmeta: version: 3.3.3\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Iaa6d18a49823a3131075db0f6723d548ba76ecaa\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I3dce496037469007fe87ae7a2cd965aee31b052c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9272e7ef21b5a38a8b95e63d5ac31045e039556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b9272e7ef21b5a38a8b95e63d5ac31045e039556"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
