)]}'
{"id":"openstack%2Frequirements~861973","triplet_id":"openstack%2Frequirements~stable%2Fzed~Idf8f6943cb3a46c2b6f5a2a53456153cb3e6b788","project":"openstack/requirements","branch":"stable/zed","topic":"new-release","hashtags":[],"change_id":"Idf8f6943cb3a46c2b6f5a2a53456153cb3e6b788","subject":"update constraint for stevedore to new release 4.0.2","status":"MERGED","created":"2022-10-20 10:58:44.000000000","updated":"2022-10-20 17:25:40.000000000","submitted":"2022-10-20 17:25:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861973-new-release","meta_rev_id":"54a18a9cedb6f7cbbba7114ed686c839145ee5f8","_number":861973,"virtual_id_number":861973,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-20 17:25:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-10-20 15:39:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-10-20 11:16:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-10-20 15:39:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 11:16:10.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-20 11:19:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-20 12:24:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 15:39:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"36525c992f606feb173b3d202eab19f088c64cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-20 10:58:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46233e0a855cf3d0017743d61fbe1690407b9d96","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-20 11:16:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f97ec4800ff1295671946d7a9c489b7a30b424f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:19:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/287a20753c0f4581bbe9cbdef6a1951a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3387001038f1457f939bdbfa08b8e88a : FAILURE in 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b939ca220fa059c83085a1e2e7a7153bfd745652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:24:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d1b1a0d0b214e168f5492dfe14d81ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b94db247a9841d59adbef9b653e023f : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d22982592b84608969917d9c6577ac9 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9233b67d7d44d9cbd4aa65dc8229b16 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0124f55dc4cb482da51a53edd227a0d5 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fafb0c46b98f4fa1be052822e90c04d2 : SUCCESS in 2m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f1a09c396cdd43d9b94f06a33a7c87d9 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9cfbd4b926f947f7bad097c6dc84afa6 : SUCCESS in 9m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f927e2b2a0b149279b405e9078b82c13 : SUCCESS in 13m 46s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e30d1c90122e43c7abd46e81de177e28 : SUCCESS in 7m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a67d46f00ddd476cb63f9f1283e00d19 : SUCCESS in 5m 19s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c4da8f300bc04871a3f64598251489f2 : SUCCESS in 5m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/81e9a1552bd643728a9586c51cd682ff : SUCCESS in 8m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8e107e63cf2c4d78bdd4965fc00b5213 : SUCCESS in 4m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e8d5cb19c78a4dbab45057738730d117 : SUCCESS in 4m 43s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e07f69181e4a4fd2acfc52d41a229bba : SUCCESS in 6m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1db3c6ac662f40c9bb6a64a4ac889454 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e09a2b4363a494d8b9dc377de6b1532 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/19a12651f0e248048e460555092f0164 : SUCCESS in 6m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2cb29f90c7b84540a1f169b5829247e3 : SUCCESS in 11m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ed5f9ae150e446ec961d8664b9a12951 : SUCCESS in 3m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/64bf98f289034d7f97a6eeb4f469f1c3 : SUCCESS in 9m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7840dc0abcef407083c482067d95a02c : SUCCESS in 4m 26s\n- cross-mistral-py38 https://zuul.opendev.org/t/openstack/build/33c35bf446bc4fa6b81fd2f81e22cfda : SUCCESS in 10m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/16ed239721ff4119ae25d8db93fd6b50 : SUCCESS in 26m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/20c7f0fd77ce4941bb5e40be53c83280 : SUCCESS in 13m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/51a7e5a20aa9471aad5f562c6f6a09b1 : SUCCESS in 4m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cbcb6716de1040c8812685e1298b008a : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b01968420c94f538ed3dd33bc7f9654 : SUCCESS in 18m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/14a61dd1b35d42d4b523e48482520f0a : SUCCESS in 4m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5a2e00075cb9457f8ae3585d60acc15b : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/85b08ada71524a528e106c9c9ad7f733 : SUCCESS in 6m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7ccf4af9b3d54011b6f88607a6c2b40a : SUCCESS in 10m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17128f944c44447c9a5806c3673e8e50 : SUCCESS in 59m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e93c0e4f677f412483a315f4c787252a : SUCCESS in 1h 21m 50s","accounts_in_message":[],"_revision_number":1},{"id":"df840b9f50ff06d3cf91b4fc71305663a2fea37f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-20 15:39:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1cb5d27f8350aed67e83b57d5c7224556110e1e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 15:40:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95a0315c43bf2487210573a2b8afb024a7ba3322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:25:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1ee9a8e09844345874a0c00084dadc6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2794ec9a493d4c68a0327422fe151fad : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb337c1f405546909b84ecba38ce37d5 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65d06d5be1af462e8106db98db2bef80 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b2849dcf36d44f6a73286ab6b208063 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d6d1eca61a854a8abf9369403fabeadc : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9675c6255e104443ae5467de4667fd17 : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0e7afcbe20414ae5ad46e32a14b9b702 : SUCCESS in 13m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/282035f88d004bf1bb5f295591096b76 : SUCCESS in 10m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e1f13863faee450f87a930a7481a83e4 : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e5d2d4f184cb4ae5b0f24495cdb2ba74 : SUCCESS in 6m 22s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/32f9f6f3507b4d0bb904d1d01495a544 : SUCCESS in 8m 32s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0fdabe95ce5a426e9cc120e22b4ff8cb : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83504c0c8c3c429ebfee420328a6c63c : SUCCESS in 8m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0113966337fe485cb1eea35278834e78 : SUCCESS in 7m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/64a5413f943c43eba8007c43cd315881 : SUCCESS in 12m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f410efb874ea4c02af8f170bd0b3b6b4 : SUCCESS in 5m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bca46530d42f4abbacbdc010fd88601e : SUCCESS in 7m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8e282e34f06642d78c1bc56bcad8cbc8 : SUCCESS in 4m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ba024711e9664c11be62b898f782766f : SUCCESS in 26m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2828703254b14ec78ac17b4e82ea14fe : SUCCESS in 15m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8583960901f24f78833841821e76858b : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/55fc573d54444e6289ed18d9e6a4224e : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f245e91793a14509a36cc82b5943ceb1 : SUCCESS in 22m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e207c9f8d70e45fba2a1a663aacece15 : SUCCESS in 5m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f056b10346db4db382f81fc8b182eaa4 : SUCCESS in 4m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c2409c02e794d6086dcea11b41dcad1 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/104f4f3f06ad4721a0e001cb9cc71a5e : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c415d3c0fb54ebbb453beeb99a7d5a2 : SUCCESS in 43m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/795e6dc1182a4263910d6eec74d1293a : SUCCESS in 1h 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"54a18a9cedb6f7cbbba7114ed686c839145ee5f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:25:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e3370304f7a7f4b7a05ab484f9573f3d78de661","revisions":{"1e3370304f7a7f4b7a05ab484f9573f3d78de661":{"kind":"REWORK","_number":1,"created":"2022-10-20 10:58:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/73/861973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/73/861973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/861973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/861973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/73/861973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/73/861973/1"}}},"commit":{"parents":[{"commit":"bacb0da499ef2a661423220fa57b1f6c2bae2c59","subject":"Merge \"update constraint for tap-as-a-service to new release 10.0.0\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bacb0da499ef2a661423220fa57b1f6c2bae2c59"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 10:58:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 10:58:42.000000000","tz":0},"subject":"update constraint for stevedore to new release 4.0.2","message":"update constraint for stevedore to new release 4.0.2\n\nmeta: version: 4.0.2\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Commit: Stephen Finucane \u003csfinucan@redhat.com\u003e\nmeta: release:Change-Id: Iaa6d18a49823a3131075db0f6723d548ba76ecaa\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Idf8f6943cb3a46c2b6f5a2a53456153cb3e6b788\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e3370304f7a7f4b7a05ab484f9573f3d78de661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1e3370304f7a7f4b7a05ab484f9573f3d78de661"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
