)]}'
{"id":"openstack%2Frequirements~862062","triplet_id":"openstack%2Frequirements~master~I79f3d7c97046af2c271034faf763efdc16d1f9c9","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-11-17 07:38:29.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-11-17 07:38:29.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I79f3d7c97046af2c271034faf763efdc16d1f9c9","subject":"update constraint for oslo.log to new release 5.0.1","status":"ABANDONED","created":"2022-10-20 11:11:32.000000000","updated":"2022-11-17 07:38:29.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"06274d2a3f3789162dd85f178f0b55a09bb78945","_number":862062,"virtual_id_number":862062,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-10-20 12:46:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":-1,"date":"2022-11-04 10:37:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-10-20 11:15:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2022-11-04 18:18:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-20 11:15:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-20 11:33:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-20 12:46:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-20 15:39:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-11-04 10:37:45.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-11-04 17:54:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-11-04 17:54:58.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-11-04 18:18:54.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"}],"messages":[{"id":"0e6acff796b850892b3f5f97e8f3fca34872812c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-10-20 11:11:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"46f2074fbc743171072a5f2b1459a382700fe5df","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-10-20 11:15:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0ffb62f314d834912fd04d7bbfc2fa03280931da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:33:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d97ca15a8c846d6b7ee76173d283a43\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/01f4b320377e4cb5945ff118e664d2cf : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"95a737a3b042d4a242c6e223190ec8360e869cea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 12:46:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2abfc91b09f4bd9972f94ca331ad28d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d07ba3f193ba4905977ca6f9517d47b8 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce86cd0411434638a04815d7af5db2fe : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbc9c173818e47e7911c3051214eaa00 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/419af4ac8afa467292764401fbcde753 : SUCCESS in 3m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e127b2044014047ba2f2cd9de5cbe56 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/426513f917d1486a8df9a65038d57087 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/33c55d4845f24aad8df0dcfa8f87480d : SUCCESS in 8m 31s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5dfab67e1b9b4870a634561352cece99 : SUCCESS in 9m 59s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/c5aae6ce2c114ccba87fea04d827abe3 : SUCCESS in 7m 18s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/e28409ab570846feb631b626f4240a7f : SUCCESS in 5m 15s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/720c45021ae54e42b02c6141ab34ff8b : SUCCESS in 5m 48s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2238ea116d8b457787886e417fc7e593 : SUCCESS in 10m 37s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/f7dc703bd62a436b97e0dca6e2dd32ff : SUCCESS in 3m 26s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/c153c396e6984b93969841ccfe7f83c3 : SUCCESS in 5m 21s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/37fb6c804ce447588098bb98068c23b5 : SUCCESS in 7m 22s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4920c30c7ed1451d8654576eb65c1549 : SUCCESS in 10m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19a6e7f963904740a507dd7743b845df : SUCCESS in 8m 24s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/bbe2ede22cc8495c95b0309ef1dee9bd : SUCCESS in 8m 00s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/474d1441225a4db79b7a1c6733d94575 : SUCCESS in 12m 10s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/119b3fd6aa404741997ace6ef42daa3b : SUCCESS in 4m 03s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/48475b50cf1e49919105e9e43a697100 : SUCCESS in 10m 05s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/dffe93e42a0a4a33bf6d6f6aa132e798 : SUCCESS in 4m 30s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/12059ac8214947b6b7f20eb191f82604 : SUCCESS in 8m 11s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/f3fe2250232945f2a809b641f80cb2a8 : SUCCESS in 27m 43s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/3ab619216f4240159d8c095d02f882a6 : FAILURE in 15m 46s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3b26cd2c66214995ad28c3c292942069 : SUCCESS in 5m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bfde1f68935b4d7aba3456321a2e1e93 : SUCCESS in 3m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffb360a918934493a0e0d0fa3ddf8e65 : SUCCESS in 20m 52s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/9d6a257caf4147018193f03083b46014 : SUCCESS in 5m 59s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/6dcc6c77ac97499e878c90e237a4f391 : SUCCESS in 4m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5c804c14d016471e89331b1fd2f8aea6 : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a4d5b2670eb84b2cad64d2369c82fc5b : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31588d4360df4637ac5c5e491f27d267 : SUCCESS in 1h 09m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c6ce2d0afea4dfabfd131770626ec43 : SUCCESS in 1h 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ea13c5189ccfd65d062ac4d9245ee69df8b3bdb4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-20 15:39:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ca2f12a53d798022d30a29dbcd603a41fe29710","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 15:57:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2899ca9806944c0b8c21314c96d799c5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f11928934bcc455cbc2a5b507964e4f9 : FAILURE in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"67f01216cdb187c52be6921d984e8ca08af40006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 17:31:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11ec1d4415a54d01882d66792458f5f2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeeb8edafb174e1eaca0401d3a215ce9 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/543c9346b5b1418ca9ecf88e99e285f7 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd1578076b9d400b86f3358ac5c1be38 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a448fb9bd26a4eb5a7d837e812a05c1e : SUCCESS in 4m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0f0e55e0d73d4e1da904224a11e24d6a : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44d7baeb8d204beda68776f00e966e6f : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/411859c91a2542d89c161dd91708b787 : SUCCESS in 8m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dbc4bbdc04294971a1e0cc19cdf29883 : SUCCESS in 11m 16s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/de8145e3cc894ab2a2f5057575d8f622 : SUCCESS in 8m 21s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/b0fec3bff4284c0b985b54b8f10124eb : SUCCESS in 7m 37s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/8bfd73d797ca40858ae75d14d4ace909 : SUCCESS in 5m 23s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/5be9a944c33d4b70bf07789876fbd660 : SUCCESS in 10m 19s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/42f736916ad44533a633287fee001fde : SUCCESS in 4m 52s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/ad4f9813694a4246abf6afeb39ffb7fc : SUCCESS in 7m 59s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/0a108e520333405e98071c4dd9e950ae : SUCCESS in 7m 37s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/a8b972ff644c4f1185ef649c4c022c8f : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22aecbfa21d84f98a5341f578b329575 : SUCCESS in 8m 48s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/a0e122510b024ac3af904dc290287ded : TIMED_OUT in 40m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/52e9d4cdddda45ffb675e223d1d2ada5 : SUCCESS in 12m 38s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/dd0a67bee5c74446a265d4911fab44a6 : SUCCESS in 3m 57s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/801be30cd97145a680257699d10df669 : SUCCESS in 10m 33s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/aabcc42dac984f8eac2fbfd0ec051a9b : SUCCESS in 7m 31s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/9b66b717de2f465b856ac02ffb505e06 : SUCCESS in 8m 47s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/2f18263197d14f43a9efcfd9ea850ddd : SUCCESS in 24m 25s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/53d6979911624b3e8953802929c963c6 : FAILURE in 14m 17s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/3346a62115904b4696a9fb719a857933 : SUCCESS in 6m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/641edb518a5e4cd28d1caf5c56bce689 : SUCCESS in 3m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/65320773135f46a899b87ff560ec843e : SUCCESS in 20m 27s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/ce09b0bf091d472f91c7096107c3fbea : SUCCESS in 5m 01s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/cc25dc0e8c634e6197879c468b1f4d30 : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cbcc8fb70f3943029a60b7debfaeb5fb : SUCCESS in 5m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3f7c133452864441a8ccd5e430db6647 : SUCCESS in 11m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a62308d576944160bc97d4cc94f32d84 : SUCCESS in 52m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef6e9ee123aa4502b7e887149b67f8bb : SUCCESS in 1h 45m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5548540fa469fed3794dccc342e3b9f06ac25ddc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-10-21 23:44:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"083f284c50f7538bc64332e4e91b3e2bdb4790d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 23:57:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f3523aed1884414a4eebcf0cb06d7da\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ef8a570e39a14261bcae6aa31888ad74 : FAILURE in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0bde069bbf6960b81269f9b7279d56082875e0c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-22 01:27:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1879f09ff2b5444881bd505e2d753cbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdd05708f48145858920f73aea9f77dd : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b3f31cba72454baa36ba41a398a000 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c6fc0e89daf488a91d145a8d1f7f320 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1c50bbde1314d3190bd74baa67f06d7 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e26197d25cd2407180744104010067f8 : SUCCESS in 3m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bbd06f85ff5346ada648938b5ecdfdf1 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/86e6e9dab1144c508a764e69cb704fec : SUCCESS in 7m 53s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1eea29ee6045459f9e8a26b2ff709d38 : SUCCESS in 11m 37s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/7b2715d449d94ea095468982a2cae575 : SUCCESS in 9m 02s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/6073064f5d2c408dbb35feeb35f5150f : SUCCESS in 4m 52s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/9ed3c3ca82da40faaa181ce4c10fb875 : SUCCESS in 6m 24s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/c6af3e2407494389b0bdcd8900c14ae1 : SUCCESS in 10m 52s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/d9e6bfdfbbcd48949af1a48042c7f941 : SUCCESS in 4m 24s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/bda2f3d821074198aaf23851ea048a0e : SUCCESS in 4m 19s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/1e07f7fbcedc4afaa9c90693ac9a1640 : SUCCESS in 8m 31s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/4012b6ec3bb448dc998dc09135907d12 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0ed073c6e9c5426786aaf7650efedf57 : SUCCESS in 6m 20s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/142908b0c3e24150987eb8d4f8e1b7a7 : SUCCESS in 10m 48s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/9d5275483beb425fa5858a375ee952ea : SUCCESS in 13m 15s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/71633434af804d27960ad496bcc88a49 : SUCCESS in 3m 42s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/10c9a53e9d6f426390dcaf530e3636dc : SUCCESS in 10m 43s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/235e2a11c2934434984d566b44724ed6 : SUCCESS in 4m 34s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/83eb1011b63a4b83ae9ff1a5a61015c7 : SUCCESS in 11m 09s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/e70e56d4f423455a8a82cdca768d42b5 : SUCCESS in 23m 57s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/2bddf4e4ea4643f6986e95924ca3a08c : FAILURE in 12m 52s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/0914a656bf0445278bffebf76dba749c : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5a3aae39d9d44be8e1a8be5f1141cf9 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/302405f4a0524ca9bb8e90348e53150f : SUCCESS in 26m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/b440b1cd68db41bc809725803c51b690 : SUCCESS in 5m 24s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/c9990d44535b45d8bdaa276e31c9a16c : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9de2f6056567471c90f66550086a45b1 : SUCCESS in 9m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d5496cb294c847c188b09dcbd67af3f6 : SUCCESS in 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/790faa3aafd94698b484736264a1cff2 : SUCCESS in 47m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd12cf4e8e2a4ae698e27773ca2b569a : SUCCESS in 1h 19m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e7a2a638a538afbcc7ebbb39dfcdcc3cc1053485","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-04 10:37:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25b3c752fbafce70818f7b7746d86eb5ba59c866","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-11-04 16:00:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6ca8896c3d6711574c3be44d45af4214ceaad394","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 18:18:54.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06274d2a3f3789162dd85f178f0b55a09bb78945","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-17 07:38:29.000000000","message":"Abandoned\n\nsuperceded by 5.0.2 https://review.opendev.org/c/openstack/requirements/+/864741","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b4d0dab3b4396a7c7e267d5d336b6cbef44fc94","revisions":{"1b4d0dab3b4396a7c7e267d5d336b6cbef44fc94":{"kind":"REWORK","_number":1,"created":"2022-10-20 11:11:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/862062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/862062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/862062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/862062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/862062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/862062/1"}}},"commit":{"parents":[{"commit":"bd6f6255fd8761185a5c9a936d0a3712e4e047ae","subject":"Merge \"[tests] to pass a list to _is_requirement_in_global_reqs()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd6f6255fd8761185a5c9a936d0a3712e4e047ae"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 11:11:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-10-20 11:11:30.000000000","tz":0},"subject":"update constraint for oslo.log to new release 5.0.1","message":"update constraint for oslo.log to new release 5.0.1\n\nmeta: version: 5.0.1\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ic25fcc50248019f3341d3f8bf793946e71198fc5\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I79f3d7c97046af2c271034faf763efdc16d1f9c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b4d0dab3b4396a7c7e267d5d336b6cbef44fc94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1b4d0dab3b4396a7c7e267d5d336b6cbef44fc94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
