)]}'
{"id":"openstack%2Frequirements~862648","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ie08ab9685e5b73934186af8ec1cd5c6af2d1abb1","project":"openstack/requirements","branch":"stable/yoga","topic":"yoga/dib-upper","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-11-06 01:44:52.000000000","reason":"Change was submitted"},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2022-11-06 01:44:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie08ab9685e5b73934186af8ec1cd5c6af2d1abb1","subject":"Raise upper diskimage-builder\u003d\u003d\u003d3.24.0","status":"MERGED","created":"2022-10-25 21:52:44.000000000","updated":"2022-11-06 01:44:52.000000000","submitted":"2022-11-06 01:44:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862648-yoga/dib-upper","meta_rev_id":"937dcd0cb6c4aa5faf323f2152f0099aa98929e4","_number":862648,"virtual_id_number":862648,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-06 01:44:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-11-06 00:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-11-04 18:12:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-11-06 00:01:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-25 22:13:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-25 23:06:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-04 18:12:54.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-11-06 00:01:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4bf2ca7f9c4c4d7ba0385026fdb78bb9838b05f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-10-25 21:52:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2b73d8645e9e80614cf992100141fe81f9f1fcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 22:13:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e72191be4c44a038440216e4392da50\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/30658442cdb6474d955b644a4ecdb05e : FAILURE in 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7ef0a9434f40b143cb2af6408ad6040ccbe1cc07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 23:06:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c184d15f03504358980c74ebd411d036\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c48cce932d640eb9d215be1e5972ff3 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c793ce9df5ec4221abf3cbe8e6e1b5a3 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeb08f69b0a942fa906cc77d34e1eab7 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c4fb5790336f422eacd8d576321cdbbf : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ae7faa5d34741659a17afbd47f90499 : SUCCESS in 3m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f226146446ee406498c044db6dea5bf3 : SUCCESS in 8m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21548cf94ef54cf891066e28c254c540 : SUCCESS in 7m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b647f0c69fcc46be8eb8a5c808e01019 : SUCCESS in 4m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/38887f430fce462a8117bbdb507950ef : SUCCESS in 9m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4da1d602d5e3447fbb8c440761abdd59 : SUCCESS in 5m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cd254365f2624acbbe3f6684ae4188c2 : SUCCESS in 5m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d9c357e51fbc445d9b150951f66171f9 : SUCCESS in 9m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9cec6a1aadf84761bdee267f3f42a34b : FAILURE in 7m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d462bb01e4cd4d25aae99de012160291 : SUCCESS in 9m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/fab9735921454d139c661c04bcf8eade : SUCCESS in 11m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c6dc724c7fb14eb1a64f2f2f15597066 : SUCCESS in 3m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c36176d947614acc92ebb9bf0611f858 : SUCCESS in 11m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/37b92eed8cc44f66ba8a1a0ba6f50cf8 : SUCCESS in 5m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e29983d3865442f58f46452d72fe4d0e : SUCCESS in 21m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9a4c1d0c33034af188da1e586a775e5f : SUCCESS in 11m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/eaf271a0f09e408d9b5c277e1d377109 : SUCCESS in 5m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b82936b683424a04b81fd1eb8716e3be : SUCCESS in 3m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/69183186b8024af38999137c10a7f3e0 : SUCCESS in 17m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/27e86cec56ae4814bf58c1958961dbd7 : SUCCESS in 7m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/327044318db6439fbeeb00d617a1bb3b : SUCCESS in 4m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e5585b17044f47e889128ccf1f5e81a9 : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b4aecc429ea248899b320d38a737f0d3 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc146d90e7ba4fd6a17e98b14065b3bd : SUCCESS in 47m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4173d3f5f16b494ea680d428ed7c3a05 : SUCCESS in 1h 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5b267cdb446c534d5ca150efb25fd92dee759dc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 17:52:10.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4e4f2e2dac08ea64e0becc3a670e57f39a72a9aa","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-04 18:12:54.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"07c876e010395bc9ed520d71d5de9a6da79d91fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 18:13:55.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad2bd5bcb4184b73be586dc8cea5eb30\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1b7d3de6ae244d4ba6949c6661b14356 : FAILURE in 18m 18s","accounts_in_message":[],"_revision_number":2},{"id":"97ece32f66f56da64753dba1b72006b83e85be40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-04 19:25:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f246ba6f70d44e3ba54ba509893bfb4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f896ab403f13479483d6479c5c77893d : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/991fcb73bdfc411b8175cfee11d5bf61 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea9997fba96a487ba0b13ea5777ea875 : SUCCESS in 4m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/479015b0547140a49eb2c9bccd187ba2 : SUCCESS in 3m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71ba0f194d8441cc92a7eab50fc1f6cc : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/010a83ea52544582b69de2b57b63492a : SUCCESS in 8m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fd68cee8ba7243e5964a85bbfa5242cc : SUCCESS in 8m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d326621d3ee8433f92830e2b8ae92654 : SUCCESS in 4m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8a74f67792574cf6bb4188df520e0e0e : SUCCESS in 10m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/92b518d7385d492eae64c9935278c672 : SUCCESS in 4m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c1dc075144c74d008450757cb4488c81 : SUCCESS in 5m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ac849449dd444c0d98d241858c90da70 : SUCCESS in 8m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5c1161c989b94db1a0d49a5bc90ef854 : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/52c2d1192f1140808fc553e1011fd5e8 : SUCCESS in 8m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/808c20690d12427fa7e268146ce5ac57 : SUCCESS in 14m 04s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/676b127c05ee4a0492545cbf4f844714 : SUCCESS in 3m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d9e4ee137b264abcb9a71cd2b3ce7549 : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/76aaca963f1547a1a912e89b555e29bb : SUCCESS in 4m 58s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ca1ed85eb434469c8138f9dc95aa580d : SUCCESS in 21m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/173da42a63f240fc9c27076f783ab214 : SUCCESS in 12m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/48696e848bf841e788861b3ab0eb9c0e : SUCCESS in 5m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/108e60526a734009aa950cd2d6c908b0 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc948d175e9f41a48d2344ffc2e517bd : SUCCESS in 20m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7fbccc02397e4a34966572ec20dd1421 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3cbff8a944924616a43eec48f8b758a8 : SUCCESS in 4m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/835f0dd046a54c9d9711a614b97939bf : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/54ac59c8a514441da64e3462864c0957 : TIMED_OUT in 40m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6744ea8be1a64be0ba82077c5a0e9b6c : SUCCESS in 48m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cfd2d84cd294c04bc3f5db6fef621b4 : SUCCESS in 1h 30m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9d6930f01d51c007eb91132a056a12b9a64dbdef","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-11-05 17:38:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"729acbe4efaec85e4a6050d9c531f3986876cdee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 17:51:23.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dba07fcd49f467cad81f415072f9c75\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/125ac915c1f34291928da5b8f61f28b2 : FAILURE in 8m 20s","accounts_in_message":[],"_revision_number":2},{"id":"711bf8f99d85d5d5caece2be729681b1d5d62b87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-05 19:03:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2f5fa9009534b5d9a8ca5bb1addfe90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edf2d9c78594913b5b5ef0370220785 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b52f1e5b55d48898ca56a4ea0ed3d03 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef463569a9a448c6b0113fbfe074e64d : SUCCESS in 3m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61e1957ff4a74876bfc9c3b5eee31ff6 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/538272abc3f5421c9686136ed0c2f549 : SUCCESS in 2m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5ae11158a4d94fceb68f6091510c4936 : SUCCESS in 8m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a5b78b0a1afd4ef994a2b300fec87a91 : SUCCESS in 8m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f93b61f4c3af46b096aa2f1422a62ae9 : SUCCESS in 4m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/967805f92d894862aed2ee50b814a864 : SUCCESS in 9m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/5c7942aee6d642f89d8285f847e749f2 : SUCCESS in 4m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2b9ab9f02b9740daa2902fc5fff96f3a : SUCCESS in 4m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1854acf56ad84869ae65c77ce4b117a4 : SUCCESS in 7m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27e79f12039a46eb885d52e18fa121db : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/938730888b6a4e2da0d014411ad42dac : SUCCESS in 7m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b92f80945e844adeb1f39b590d708978 : SUCCESS in 9m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9db19d7900f14ac5881386e1e1684909 : SUCCESS in 3m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b02dd8b593414d86baba1f43786d3905 : SUCCESS in 9m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1acf17796d464d6a833445c2a5535bfa : SUCCESS in 4m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5d022fe09d9e4238b75bfe51b0297d4b : SUCCESS in 17m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/be40e1df650c4d8385f587095cd60469 : SUCCESS in 11m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/082f041d87624c499e9cfc93c1b2bb7b : SUCCESS in 4m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0406a8508424e04bdba9ca4d1df0eb4 : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9eb72ddd51d143a4be943266e57000c7 : SUCCESS in 18m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e07ae43191fc4649ac8639b0fdc683ad : SUCCESS in 5m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ebbc5635e4394606b6a126101a21e4df : SUCCESS in 4m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/214327db96b548858c19684b9231f1e5 : SUCCESS in 5m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4cb5a53ceb7c46f18a6ccae978597484 : SUCCESS in 11m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c96eb9011081462989fb21408e2a0e4a : SUCCESS in 50m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1645d2ffcaba4e728b32ea41926e76f2 : SUCCESS in 1h 21m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7dd11bfd44692e1301ac672d7f234ab2766b7e0e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-11-06 00:01:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2735c6c5c4fd83edcfb1871c055e500e95233d2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-06 00:02:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c94f00e57cf22d54823da420334661301e4be984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-06 01:44:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0552dd09788a4a90bb6033b27f61c348\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1f4ca32f44428f814009f816621a03 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98f4ca98d3e24aa1846fadff529bc46f : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed4db8ad911d4b738e5e4321cb4ae1aa : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7d2161da7c1a4fb080a0d4f297898eed : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/736cf43e968a456e97c33a1de30b6b3c : SUCCESS in 2m 47s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4ccf4b0463b449f9bc37b92ff556d8bb : SUCCESS in 11m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68645ef7902b4fe5a4282cb2a29aa903 : SUCCESS in 7m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c34ba6055ef0456a989f46598c376787 : SUCCESS in 11m 01s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4afcbeb714b5493e811d6faf3e6c8757 : SUCCESS in 4m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a8c12c9252854f51bfc0f41e5b701c64 : SUCCESS in 5m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a368b3b8d2004c98b7902bc446170ba8 : SUCCESS in 9m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8024449a8cf0439c8b2b93cb3f9e3144 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/15e012226ae4489582dd06c381d46e1a : SUCCESS in 8m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/372634a9f1204e768cabe94b91f19807 : SUCCESS in 11m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9fa188593ce44e0da459ecd5e055ad19 : SUCCESS in 3m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0df19222b0f940b8a5507c2267331c55 : SUCCESS in 9m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f319a8937c84a75afda7923e5e85ec3 : SUCCESS in 3m 53s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/67aaf388047e41e6848b0d2cc97484d9 : SUCCESS in 27m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a13c2d13cbe54b5b8d3f17b21f10f691 : SUCCESS in 12m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5e0cead03013414bb3d5ec056d82b1b3 : SUCCESS in 4m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/158edddd6e6b4fe9bf99e8e8a9304e6e : SUCCESS in 2m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0b3410268c94ba5957f52da3d092bbb : SUCCESS in 21m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f935d90b5ccf410babee6c9072fb4722 : SUCCESS in 4m 45s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/01a6e7a8579a43df85e8ab80127829e1 : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f5991f973264c9cb52d93132706ed56 : SUCCESS in 6m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7c27436f6e004626ac3380c7c7209319 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcd2170fdaf44be08fd9ef2c5f0150e0 : SUCCESS in 36m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d50a130822f44aec95a14455956f868f : SUCCESS in 1h 42m 41s","accounts_in_message":[],"_revision_number":2},{"id":"937dcd0cb6c4aa5faf323f2152f0099aa98929e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-06 01:44:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"620b974f8d61d82e431475cef0f5dc36504cf28a","revisions":{"9365ac59772c5a0bdc3c908d864e910f3a806f8b":{"kind":"REWORK","_number":1,"created":"2022-10-25 21:52:44.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/48/862648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/862648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/862648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/862648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/862648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/862648/1"}}},"commit":{"parents":[{"commit":"9d58bffe63dfaa064d44ac59a1ea7dc37ddb0d2c","subject":"update constraint for stevedore to new release 3.5.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9d58bffe63dfaa064d44ac59a1ea7dc37ddb0d2c"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-25 21:49:38.000000000","tz":780},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-25 21:49:38.000000000","tz":780},"subject":"Raise upper diskimage-builder\u003d\u003d\u003d3.24.0","message":"Raise upper diskimage-builder\u003d\u003d\u003d3.24.0\n\ndiskimage-builder is branchless and has minimal pip dependencies, so\nit is desirable to run the latest even on stable branches.\n\nThis sets the same version currently on master.\n\nChange-Id: Ie08ab9685e5b73934186af8ec1cd5c6af2d1abb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9365ac59772c5a0bdc3c908d864e910f3a806f8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9365ac59772c5a0bdc3c908d864e910f3a806f8b"}]},"branch":"refs/heads/stable/yoga"},"620b974f8d61d82e431475cef0f5dc36504cf28a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-04 17:52:10.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/48/862648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/862648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/862648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/862648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/862648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/862648/2"}}},"commit":{"parents":[{"commit":"c02280a2018d1c8d02f989a8bdb3adc6ba11f009","subject":"Update Yoga branch to run npm jobs on focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c02280a2018d1c8d02f989a8bdb3adc6ba11f009"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-25 21:49:38.000000000","tz":780},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-11-04 17:52:10.000000000","tz":0},"subject":"Raise upper diskimage-builder\u003d\u003d\u003d3.24.0","message":"Raise upper diskimage-builder\u003d\u003d\u003d3.24.0\n\ndiskimage-builder is branchless and has minimal pip dependencies, so\nit is desirable to run the latest even on stable branches.\n\nThis sets the same version currently on master.\n\nChange-Id: Ie08ab9685e5b73934186af8ec1cd5c6af2d1abb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/620b974f8d61d82e431475cef0f5dc36504cf28a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/620b974f8d61d82e431475cef0f5dc36504cf28a"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
