)]}'
{"id":"openstack%2Frequirements~863083","triplet_id":"openstack%2Frequirements~master~If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-11-07 16:51:09.000000000","reason":"Change was abandoned"},"32927":{"account":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"last_update":"2022-11-07 16:25:18.000000000","reason":"\u003cGERRIT_ACCOUNT_32927\u003e replied on the change","reason_account":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"}}},"hashtags":[],"change_id":"If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2","subject":"[CVE-2022-42969] Reduce usage of package py","status":"ABANDONED","created":"2022-10-31 15:49:53.000000000","updated":"2022-11-07 16:51:09.000000000","total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"fba38e55df01f728ac885fd76ae7caa4cacdd531","_number":863083,"virtual_id_number":863083,"owner":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-31 16:01:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-10-31 17:29:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-01 04:27:24.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"CC"},{"updated":"2022-11-07 15:06:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1b0e89b4085a3c30a9bbcf750a98a2afb7a20982","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-10-31 15:49:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57b3e7365b9b3abfd6a57cad1b49f44d4de52329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 16:01:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00034882eb3640afbbd8cf250aae6446\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b8932e27c95a415abe4ec38f2840207e : FAILURE in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fc7379cfa1fa336b9ad086bb18502378310c532b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 17:29:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e6ae9dbbd7744c48dca05deb66aea80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb34fe4c526e49a0b09696d382d1b15f : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2602f66bde3d4a2b9bf58711779fde8a : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b28167f575843138fdc75503becc327 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a2458b08cb048e3be20f1ee9fc6d9c5 : SUCCESS in 3m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a4e83b139fc3424497672198b0059d04 : SUCCESS in 3m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/edd6a544b9ad4defb974b83020b0c744 : SUCCESS in 2m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e50a54d049894022b4a4ce49a688aa6d : SUCCESS in 8m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3e9d43beffb3485a89aaefa912f8dfbd : SUCCESS in 12m 45s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/4cefbc3dc35c49a4b93df3b6d2bc823b : SUCCESS in 8m 15s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/0ad7125a668b45a0bdd27b753269a3b5 : SUCCESS in 5m 01s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/5be4848551b2432d92f9816edda2755c : SUCCESS in 5m 10s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/2b731ef11b0848fda2554b34dec16f3a : SUCCESS in 10m 45s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/600dd6a567b04028a97f44bbdc3529c5 : SUCCESS in 4m 21s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/2a50130d486646008a8f57ed6d89eb8e : SUCCESS in 4m 52s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/ae7d43a76a744ed7a8472eec01fcaee3 : SUCCESS in 8m 13s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/8cb65edb8ecf4241a6fe614c2dd17358 : FAILURE in 3m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a2c814647a604c80997d652a24818967 : FAILURE in 5m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/04a538a3e04b40cb8130f098064338cd : FAILURE in 8m 29s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/f0e69f7935a84dee9769efe0c8e02378 : SUCCESS in 12m 31s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/643e67126c844748bf0ff1625a7f0541 : SUCCESS in 3m 09s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/564dd2fa472146578aa28af913c3e27a : SUCCESS in 8m 13s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/cf6af222dfd94ca894890951533df614 : SUCCESS in 4m 47s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/7d7b5048331e41d8adf825a74ecaf1dc : SUCCESS in 8m 49s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/19caf2f2b7f14dc98b56a34fece1a47f : SUCCESS in 24m 30s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/d71b078b631845598540aff4edefff90 : SUCCESS in 11m 21s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/91c15aa6343b495493142adbde2ff2cd : SUCCESS in 4m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59b71d76b76b469abc1c1c2d4f282137 : SUCCESS in 2m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea5b26c4cbc24ded8990280856df2db5 : SUCCESS in 21m 29s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e624b676dde046daa8f2ff0ca5c888ec : SUCCESS in 4m 14s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/1372c9aa43504cb3abdaa09d195c5fec : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9805815b0da743bca9717d8d175dd719 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ff2a3a1045d740aeaa8591dbc307b70a : SUCCESS in 9m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2aecb679e1b41b497c764f06c1b4809 : SUCCESS in 52m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/081cb7bd1a364706a65bac2144aa78c0 : SUCCESS in 1h 34m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1460d7ab8c5ea79a74fa89a95062646d97708eda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-10-31 19:42:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ab272b692a188c1f4c08bb6fda2473ea4c1d7999","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 19:59:06.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c935ef2dcb0242928286569c92451eea\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb9f89521af84279b49a88275dabe96f : FAILURE in 13m 11s","accounts_in_message":[],"_revision_number":2},{"id":"89570326a50e712b95183df49783c05d511e7795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 21:27:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb61d33d36514b9984b610a47c3e89b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4679c25e647c47b893dad6602232037b : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6a69f61c58416dbd110a2f76d63838 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dba2a974f44418eb2af15ef1441cdfe : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00e5a73167524345a4885242d9db6d09 : SUCCESS in 3m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7cbd7851fa834674805fe0588d8e3c84 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72e0125e84e243b0b04eef4304c01638 : SUCCESS in 4m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f3387e01a8c04ae5a7a2c8f23df7bd57 : SUCCESS in 6m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e5c60cfaa59e48d5be4ab4ecb177eb50 : SUCCESS in 14m 52s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/3f46546a343f41f48150fee525c7718a : SUCCESS in 8m 08s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/86d35537fb03434aaedbd9793933655a : SUCCESS in 4m 31s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/529ffa72eb58484490faad4dc74a398e : SUCCESS in 5m 28s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/6d15f85c6e8043038cf58627390fc7d1 : SUCCESS in 9m 41s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/a07be9f3640d450da0271be3b0c70c20 : SUCCESS in 5m 15s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/f88682f0eff34ad8b2cecc6d73ce9a42 : SUCCESS in 4m 55s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/c5048906e3c546f49678d5d18180f412 : SUCCESS in 8m 11s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/ed2843fbf35d4d7ab716e61235c22a00 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63e625574e904815af1be53d318c4b64 : FAILURE in 6m 23s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/29263b768bde4218801f6eae5a42e622 : FAILURE in 8m 02s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/650c2645c4ca402c995c391ea5ef66b9 : SUCCESS in 12m 00s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/c67767ef090149289765d34deb18688f : SUCCESS in 3m 31s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/434aa0230d0144a29fef95ed450cadcd : SUCCESS in 8m 58s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/6c0da98e9cf449d08393df869435c549 : SUCCESS in 4m 14s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/4ac8047bb8904637af6bd78b28f68cbc : SUCCESS in 9m 18s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/3ab773bb511d4abd82f5a7a390d2e446 : SUCCESS in 20m 36s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/c2719877860b4f79ad87b75c4347157c : SUCCESS in 12m 37s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/314591c62cd549c1839d303362d2bf2b : SUCCESS in 6m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ebcb1d16817948ed8be6884bf37dca9a : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ab312966d2704d878853a2d6e2178902 : SUCCESS in 20m 48s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/cb1d02524e84417abd9e38ebe1c6bff7 : SUCCESS in 4m 27s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/84165c1654514317aebe10f694f63a02 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/99648791322547d0bde7bf6cdde4a3b4 : SUCCESS in 7m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/21bee76bfb394cd9aae4da0ac69f10a0 : SUCCESS in 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/935042a67f0e4330a9694311bf2b3cf3 : SUCCESS in 49m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0bcacf4f01d402e959142f3a8bccdc7 : SUCCESS in 1h 43m 28s","accounts_in_message":[],"_revision_number":2},{"id":"386a23e0b24267c204b4767c6ba83524a3accb29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 13:20:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"940f2d15213df22ce50e1c07ee6a97ab8ebf0156","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 13:35:16.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/252a13f1b91047bfac95aab28aefcc82\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1c091ee0e7e74792ba0ecb2f7b9c51b3 : FAILURE in 9m 59s","accounts_in_message":[],"_revision_number":3},{"id":"96a4daa16800226c22d4bf48a4a1e88d4e63d382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 14:39:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1850b38af49943d380ef1ba8b193cd21\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d11597e82d4d2eabf3f58ba2b50387 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af6a21542c846c6ae2c95a19a0202b1 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8574b9c1a23f40c685d9ac870a1307dd : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2972a6172dda4502906c398c4f65e086 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6a9517cf47284139befad665d30c99d1 : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/041784a73b4746acb017a77122757992 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/81ab8353f385496cadb237b7ba827e04 : SUCCESS in 8m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2fba7cad45224c5797f1d511217dfb73 : SUCCESS in 11m 29s\n- cross-aodh-py310 https://zuul.opendev.org/t/openstack/build/fbcd9f4e7370444084b5e8e42e527042 : SUCCESS in 8m 56s\n- cross-barbican-py310 https://zuul.opendev.org/t/openstack/build/90315182f4634ddcb9d5796cf97a2275 : SUCCESS in 5m 22s\n- cross-ceilometer-py310 https://zuul.opendev.org/t/openstack/build/3f8f4beb47354eb98436e1971ac62c2a : SUCCESS in 6m 07s\n- cross-cinder-py310 https://zuul.opendev.org/t/openstack/build/de61486292b04b5486f9f9d6794a4f87 : SUCCESS in 10m 03s\n- cross-designate-py310 https://zuul.opendev.org/t/openstack/build/afb2a237ff79429684dc59527dd745b6 : SUCCESS in 4m 39s\n- cross-glance-py310 https://zuul.opendev.org/t/openstack/build/7ead2ad08ba1419b9e7107b30d7be60b : SUCCESS in 5m 51s\n- cross-heat-py310 https://zuul.opendev.org/t/openstack/build/a86b5baa254f4cf18e057326b14117bb : SUCCESS in 8m 25s\n- cross-horizon-py310 https://zuul.opendev.org/t/openstack/build/2475be38fd024b70be7f39ef481298a0 : FAILURE in 3m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/68189e0f4c0d47b0b48cc51a22d488ec : SUCCESS in 7m 33s\n- cross-ironic-py310 https://zuul.opendev.org/t/openstack/build/5b861458ec9b4ff9b2d23f3da6060e74 : SUCCESS in 8m 52s\n- cross-keystone-py310 https://zuul.opendev.org/t/openstack/build/22f905909dc74c158abdb40c362bef66 : SUCCESS in 11m 17s\n- cross-kuryr-py310 https://zuul.opendev.org/t/openstack/build/ea631a4ff7694a4196c9b734c083366f : SUCCESS in 3m 56s\n- cross-manila-py310 https://zuul.opendev.org/t/openstack/build/a91bc7412c42462895a60e4b90f7e631 : SUCCESS in 9m 34s\n- cross-masakari-py310 https://zuul.opendev.org/t/openstack/build/009ba63450cd460f9b782e989159be05 : SUCCESS in 4m 38s\n- cross-mistral-py310 https://zuul.opendev.org/t/openstack/build/0044cd478007465bb201f07711229a36 : SUCCESS in 10m 25s\n- cross-neutron-py310 https://zuul.opendev.org/t/openstack/build/98295290bcb14f22aa2332fd9600a82c : SUCCESS in 27m 59s\n- cross-nova-py310 https://zuul.opendev.org/t/openstack/build/ec374906b00841c4a8eab16c8f82accf : SUCCESS in 12m 25s\n- cross-placement-py310 https://zuul.opendev.org/t/openstack/build/672d439be70c401fb99c39fd3be1251a : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a68ae7ff495a4b9ab82fd8b740783301 : SUCCESS in 3m 37s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e670bd38bdf3419bbf33597456d57796 : SUCCESS in 20m 20s\n- cross-placement-functional-py310 https://zuul.opendev.org/t/openstack/build/e0e65e09eda54e5897cc52bff32bd6c2 : SUCCESS in 5m 28s\n- cross-octavia-py310 https://zuul.opendev.org/t/openstack/build/7721679a58514b3fa9a0b1201f1fb2fe : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e24b99b01d0943b294afdc048f60f489 : SUCCESS in 6m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4b3c9298ac1f4168ab149aa3d732a413 : SUCCESS in 11m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/306001a681354827873fac93cfd7c9b3 : SUCCESS in 48m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7ab5f500e6348eca88ef8b19ffa4a6a : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":3},{"id":"df0458c07f7e2f72f086df18db06a95f5ddc7009","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-07 15:06:54.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a36c2aa94df2de5ede9e421cc0d5e0b0737f2832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 15:54:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba6cf31b0208ae4e069cb4d0ec2ea9116c1ee46f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 15:57:35.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"8de5605764199874b19fc6f8b646b3b2203f0af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 15:58:38.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"752d727983ed1ea2ef9a7650306e9aba2d99506d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 16:01:16.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"f7011bb1b006dfe7dacabcfe415875422abcdf4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 16:02:10.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"72723c49ca7a201cee898c01127a0812a8dbbbc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:05:16.000000000","message":"Patch Set 8:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/863101 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"3f00f2ccd0d39ca07bd28f2005c216fc34f22ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:05:19.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/863101 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"88b95f8d0edcdebe133657713ad5845c9c5288e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 16:05:59.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"dd4ff2afb1d81df9e056dd99715f6e8c66ffb289","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:19:23.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c6a41888eb84b108720c4f26c21870b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0fcbecf5878843e0aaaf05df4778c471 : FAILURE in 6m 38s","accounts_in_message":[],"_revision_number":9},{"id":"08e3afc8d6d16e024ffcbcf561b513524f0ea4d4","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 16:25:18.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a3cc938a849f73f6e3c0885601fe98d239921904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 16:27:08.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"46992d2cdaa8fe5c09b6fff0a7ab864c2eed8ccb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 16:38:42.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24fce071d66d4c089f4defaf2c366beb\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1ee2068eb8474f08b2e23f9341c38a05 : FAILURE in 7m 24s","accounts_in_message":[],"_revision_number":10},{"id":"fba38e55df01f728ac885fd76ae7caa4cacdd531","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2022-11-07 16:51:09.000000000","message":"Abandoned\n\npatch reworked in 863893: Bump pytest version to the latest version to avoid py usage. | https://review.opendev.org/c/openstack/requirements/+/863893","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"7790b3478a660b811bde8c346c44b2f57cd32069","revisions":{"05ab3bb16390e3724920c095a4ccc907fd29d9d8":{"kind":"REWORK","_number":1,"created":"2022-10-31 15:49:53.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/1"}}},"commit":{"parents":[{"commit":"2bb65623c874685205e1598f9503a0e7a864e973","subject":"Merge \"update constraint for pbr to new release 5.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bb65623c874685205e1598f9503a0e7a864e973"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"subject":"[CVE-2022-42969] Do not use package py","message":"[CVE-2022-42969] Do not use package py\n\nSome weird CVE was found in package py.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\n\nStory: 2010389\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05ab3bb16390e3724920c095a4ccc907fd29d9d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/05ab3bb16390e3724920c095a4ccc907fd29d9d8"}]},"branch":"refs/heads/master"},"6d5a7df3edb59a0435cf795700d71faf2500280f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-31 19:42:59.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/2"}}},"commit":{"parents":[{"commit":"2bb65623c874685205e1598f9503a0e7a864e973","subject":"Merge \"update constraint for pbr to new release 5.11.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2bb65623c874685205e1598f9503a0e7a864e973"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-10-31 19:42:59.000000000","tz":0},"subject":"[CVE-2022-42969] Do not use package py","message":"[CVE-2022-42969] Do not use package py\n\nSome weird CVE was found in package py.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\n\nStory: 2010389\nDepends-On: I7549373c9ebba0a618708924d4fce6c464639dec\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d5a7df3edb59a0435cf795700d71faf2500280f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6d5a7df3edb59a0435cf795700d71faf2500280f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"49e1d4adc969691a2b848ce9846d79c00885bc1f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-07 13:20:10.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/3"}}},"commit":{"parents":[{"commit":"ae58e9fd05835170845ac1234029ee5ee96fec02","subject":"update constraint for virtualbmc to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae58e9fd05835170845ac1234029ee5ee96fec02"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-07 13:20:10.000000000","tz":0},"subject":"[CVE-2022-42969] Do not use package py","message":"[CVE-2022-42969] Do not use package py\n\nSome weird CVE was found in package py.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\n\nStory: 2010389\nDepends-On: I7549373c9ebba0a618708924d4fce6c464639dec\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49e1d4adc969691a2b848ce9846d79c00885bc1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/49e1d4adc969691a2b848ce9846d79c00885bc1f"}]},"branch":"refs/heads/master","description":"Rebase"},"47d7373f06db4043c9c9fc532cb323675c744df1":{"kind":"REWORK","_number":4,"created":"2022-11-07 15:54:46.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/4"}}},"commit":{"parents":[{"commit":"ae58e9fd05835170845ac1234029ee5ee96fec02","subject":"update constraint for virtualbmc to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae58e9fd05835170845ac1234029ee5ee96fec02"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-11-07 15:52:18.000000000","tz":180},"subject":"[CVE-2022-42969] Reduce usage of package py","message":"[CVE-2022-42969] Reduce usage of package py\n\nSome weird CVE was found in package py.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\nHorizon uses py and requires that\npy is set in requirements.\n\nStory: 2010389\nDepends-On: I7549373c9ebba0a618708924d4fce6c464639dec\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47d7373f06db4043c9c9fc532cb323675c744df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47d7373f06db4043c9c9fc532cb323675c744df1"}]},"branch":"refs/heads/master"},"de26a97a363dc7a8c958a8d8d552ba397620cc10":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-11-07 15:57:35.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/5"}}},"commit":{"parents":[{"commit":"ae58e9fd05835170845ac1234029ee5ee96fec02","subject":"update constraint for virtualbmc to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae58e9fd05835170845ac1234029ee5ee96fec02"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-07 15:57:35.000000000","tz":0},"subject":"[CVE-2022-42969] Reduce usage of package py","message":"[CVE-2022-42969] Reduce usage of package py\n\nSome weird CVE was found in package py.\nMaintainers support py in maintenance mode.\nSo it\u0027s better do not use it.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\nHorizon uses py and requires that\npy is set in requirements.\n\nStory: 2010389\nDepends-On: I7549373c9ebba0a618708924d4fce6c464639dec\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de26a97a363dc7a8c958a8d8d552ba397620cc10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/de26a97a363dc7a8c958a8d8d552ba397620cc10"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e4b6ec861d53209bbc001df9deff97e7e20a914b":{"kind":"REWORK","_number":6,"created":"2022-11-07 15:58:38.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/6"}}},"commit":{"parents":[{"commit":"ae58e9fd05835170845ac1234029ee5ee96fec02","subject":"update constraint for virtualbmc to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae58e9fd05835170845ac1234029ee5ee96fec02"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-07 15:58:38.000000000","tz":0},"subject":"[CVE-2022-42969] Reduce usage of package py","message":"[CVE-2022-42969] Reduce usage of package py\n\nSome weird CVE was found in package py.\nMaintainers support py in maintenance mode.\nSo it\u0027s better do not use it.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\nHorizon uses py and requires that\npy is set in requirements.\n\nStory: 2010389\nDepends-On: I7549373c9ebba0a618708924d4fce6c464639dec\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4b6ec861d53209bbc001df9deff97e7e20a914b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e4b6ec861d53209bbc001df9deff97e7e20a914b"}]},"branch":"refs/heads/master"},"a7740be1c4f85a3078dfb02970a4fd07ea65108a":{"kind":"REWORK","_number":7,"created":"2022-11-07 16:01:16.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/7"}}},"commit":{"parents":[{"commit":"ae58e9fd05835170845ac1234029ee5ee96fec02","subject":"update constraint for virtualbmc to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae58e9fd05835170845ac1234029ee5ee96fec02"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-07 16:01:16.000000000","tz":0},"subject":"[CVE-2022-42969] Reduce usage of package py","message":"[CVE-2022-42969] Reduce usage of package py\n\nSome weird CVE was found in package py.\nMaintainers support py in maintenance mode.\nSo it\u0027s better do not use it.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\nHorizon uses py and requires that\npy is set in requirements.\n\nStory: 2010389\nDepends-On: I7549373c9ebba0a618708924d4fce6c464639dec\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7740be1c4f85a3078dfb02970a4fd07ea65108a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7740be1c4f85a3078dfb02970a4fd07ea65108a"}]},"branch":"refs/heads/master"},"ee88d26ec00493212359c5de877846998abcc9f7":{"kind":"REWORK","_number":8,"created":"2022-11-07 16:02:10.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/8"}}},"commit":{"parents":[{"commit":"ae58e9fd05835170845ac1234029ee5ee96fec02","subject":"update constraint for virtualbmc to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae58e9fd05835170845ac1234029ee5ee96fec02"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-07 16:02:09.000000000","tz":0},"subject":"[CVE-2022-42969] Reduce usage of package py","message":"[CVE-2022-42969] Reduce usage of package py\n\nSome weird CVE was found in package py.\nMaintainers support py in maintenance mode.\nSo it\u0027s better do not use it.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\nHorizon uses py and requires that\npy is set in requirements.\n\nStory: 2010389\nDepends-On: I7549373c9ebba0a618708924d4fce6c464639dec\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee88d26ec00493212359c5de877846998abcc9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ee88d26ec00493212359c5de877846998abcc9f7"}]},"branch":"refs/heads/master"},"b7d29ccb32130c81de635c695ac341dba35d31c8":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2022-11-07 16:05:59.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/9"}}},"commit":{"parents":[{"commit":"ae58e9fd05835170845ac1234029ee5ee96fec02","subject":"update constraint for virtualbmc to new release 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ae58e9fd05835170845ac1234029ee5ee96fec02"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-07 16:05:59.000000000","tz":0},"subject":"[CVE-2022-42969] Reduce usage of package py","message":"[CVE-2022-42969] Reduce usage of package py\n\nSome weird CVE was found in package py.\nMaintainers support py in maintenance mode.\nSo it\u0027s better do not use it.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\nHorizon uses py and requires that\npy is set in requirements.\n\nStory: 2010389\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7d29ccb32130c81de635c695ac341dba35d31c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7d29ccb32130c81de635c695ac341dba35d31c8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7790b3478a660b811bde8c346c44b2f57cd32069":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-11-07 16:27:08.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/83/863083/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/83/863083/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/83/863083/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/83/863083/10"}}},"commit":{"parents":[{"commit":"90b48934f4f018cf6e04dcd42731731346422b03","subject":"Remove python-dev from bindep","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90b48934f4f018cf6e04dcd42731731346422b03"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2022-10-31 15:48:31.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2022-11-07 16:27:08.000000000","tz":0},"subject":"[CVE-2022-42969] Reduce usage of package py","message":"[CVE-2022-42969] Reduce usage of package py\n\nSome weird CVE was found in package py.\nMaintainers support py in maintenance mode.\nSo it\u0027s better do not use it.\nPytest does not use py since 7.2.0.\nSo we upgrade pytest.\nHorizon uses py and requires that\npy is set in requirements.\n\nStory: 2010389\nChange-Id: If16d88a54d7b86f4afb7ad24df00f5f12ba69cd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7790b3478a660b811bde8c346c44b2f57cd32069"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7790b3478a660b811bde8c346c44b2f57cd32069"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
